)]}'
{"id":"openstack%2Fkayobe~787938","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I3d85b341cf541cfb9b486021e66b316833fe9130","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I3d85b341cf541cfb9b486021e66b316833fe9130","subject":"Use released version of mrlesmithjr.mdadm","status":"MERGED","created":"2021-04-26 08:25:41.000000000","updated":"2021-04-26 18:01:34.000000000","submitted":"2021-04-26 18:00:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787938","meta_rev_id":"ee370c4a451fcc420289b0d73d03ea366344b6a9","_number":787938,"virtual_id_number":787938,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 17:59:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-04-26 11:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-04-26 12:23:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-04-26 12:23:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 10:23:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 11:22:57.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-04-26 12:23:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"3360f49e2f5f3d02f3adbaf5c5e69bf41b8b461f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 08:25:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"927dab4e1bface5605c45cc7e7d1917bb3e0ea94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 10:23:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3018093609b467ca236c50bad9930bd : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a87b8a7076a4e62b54c2444c1db0347 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a0ee106c95a487f94e3e26f3c10b303 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a1c1d7fa50d4625adc15eaf9f8e997e : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af33e9c4414846bdbafea340cefe0855 : SUCCESS in 7m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/149a1587bb1f4d0ebfa9907b861be708 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b619b89911ba4f39a5b63ca924b1ac3e : SUCCESS in 9m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77b454f1fbf246808665cbe5664f1508 : SUCCESS in 17m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6158604218114484bc9de78dc7b23367 : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e86e6e92650b452fa075f85912c7bd33 : SUCCESS in 16m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b5863b31cc9740219d065aa9ec18dce9 : SUCCESS in 1h 54m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f1edd39f36ab4f4f8d52b1879d9709f9 : SUCCESS in 27m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ecf7c7c945e84895ab6b8962bb6ddfaa : SUCCESS in 26m 36s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/340cf6feb99a46078279ade0089c08cd : SUCCESS in 11m 21s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"ac4d2870200cb3d266233b5d126f4780fa03a47a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-26 11:22:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c81e9345965470d653ff05a1edf6e366ada44606","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-26 12:23:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7e542fec7aaba0c76c3a0bbf14f93d681d113bbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:23:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e03c4a26d083fc749bb0320472676b8f22dbb976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 13:23:09.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"b496ec14d5a1d29cde5a2b482728de2e322f27fe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-26 14:14:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd8f4acd212a6fe454088f69efab9f7285bc6fe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:00:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d25c1ac897ce4b138269931f40d60270 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07e377cbfecc4e08a4c4d0e75a058399 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e668d17036441df977f4ba409e3d75f : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ff8ddfcde8b4b0fb3438a31718ff25a : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c4185d0b284e47bf040ce6eb6583cc : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78cb46df632f46cb899f138503761eba : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/95d12c7d078d41449eaff1bdc6b9f1c1 : SUCCESS in 7m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/66526876f20c4399a67a68606af76dea : SUCCESS in 18m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/345e6e302715467ba9b2fe0271327c00 : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/347b3e9eddc040d99055edd405b80f05 : SUCCESS in 12m 15s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/75c79067284f4969b71f3735e97458c3 : SUCCESS in 1h 44m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/595fc4d5f9dd41dcaa1275ab8c40aa21 : SUCCESS in 19m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b441bf5dea51430bb8a5a7defe71cb29 : SUCCESS in 25m 34s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fb9cb99b4b4a49b3b168810c28aaf13d : SUCCESS in 9m 37s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"422a51686fdd49b485a31e3f9320ce8cc49caf87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:04:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e0319a51e0d5b139d08e2dca0aa0f513a1b24ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:59:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e4e61ff941641ceae260fc586d6d6e9 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b6924b01c9e49fbb3036d112de974eb : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e75ab745c1c45eea61dff0d5f1ea94b : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d35f8fba9c84be5a3fbdfa0d6ab6f1b : SUCCESS in 5m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b3308a25094840af9dbafc078dfc4aa9 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68c1b420fa5c4f4ab2c954c68034946d : SUCCESS in 7m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/60075c9d02be44c4ad8d4b9d8a84a5bc : SUCCESS in 15m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6dcea5fb2ecc4726b6bb93d74ab71963 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8fa836beb7224043a1f4172ba5bb3a6f : SUCCESS in 12m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/010b145b2f4a4d8181bfc10aadc6cf21 : SUCCESS in 1h 53m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/42f460b9c2a648d7a55807e7cc8275e7 : SUCCESS in 31m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c8508fd584c74150998d183c822c0751 : SUCCESS in 24m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/42f114ea66ee4f1bb6307df0d5d760f3 : SUCCESS in 11m 06s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"31210fd12512966e8ca9b92785843192b22bc10e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 18:00:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ee370c4a451fcc420289b0d73d03ea366344b6a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 18:01:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d54c57b987eb4e269c6d129db0cebef3 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d633f938dddde58f5504663616f71dbb207aeb0a","revisions":{"d633f938dddde58f5504663616f71dbb207aeb0a":{"kind":"REWORK","_number":1,"created":"2021-04-26 08:25:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/38/787938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/787938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/787938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/787938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/787938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/787938/1"}}},"commit":{"parents":[{"commit":"5d41d2ea8ed7a4e55e8899d7c0f011fb654e4cc7","subject":"Merge \"Fix overcloud provision with fast track enabled\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d41d2ea8ed7a4e55e8899d7c0f011fb654e4cc7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-26 08:04:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-26 08:25:20.000000000","tz":120},"subject":"Use released version of mrlesmithjr.mdadm","message":"Use released version of mrlesmithjr.mdadm\n\nAn initial version of the mrlesmithjr.mdadm role was released 2 days\nago. There are only minimal changes compared to the commit we were\nusing. This should avoid the following issue seen in CI:\n\n[WARNING]: - mrlesmithjr.mdadm was NOT installed successfully: - the specified\nversion (5be3ee7d330aa17317897bd104dc87ff0df11915) of mrlesmithjr.mdadm was not\nfound in the list of available versions ([{\u0027id\u0027: 152374, \u0027url\u0027: \u0027\u0027, \u0027related\u0027:\n{}, \u0027summary_fields\u0027: {}, \u0027created\u0027: \u00272021-04-24T02:51:30.453954Z\u0027, \u0027modified\u0027:\n\u00272021-04-24T02:51:30.453980Z\u0027, \u0027name\u0027: \u0027v0.1.0\u0027, \u0027version\u0027: \u00270.1.0\u0027,\n\u0027commit_date\u0027: \u00272021-04-23T22:50:00-04:00\u0027, \u0027commit_sha\u0027:\n\u0027b8a9853b7eafa9afc62bbb7728f64fd5dc9f54de\u0027, \u0027download_url\u0027:\n\u0027https://github.com/mrlesmithjr/ansible-mdadm/archive/v0.1.0.tar.gz\u0027, \u0027active\u0027:\nNone}]).\nERROR! - you can use --ignore-errors to skip failed roles and finish processing the list.\n\nChange-Id: I3d85b341cf541cfb9b486021e66b316833fe9130\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/787937\n(cherry picked from commit 7411e624dea77a1b94d91bd37c9a388aa5011bb4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d633f938dddde58f5504663616f71dbb207aeb0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d633f938dddde58f5504663616f71dbb207aeb0a"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
