)]}'
{"id":"openstack%2Fkayobe~787960","triplet_id":"openstack%2Fkayobe~master~I5d105006fad6e7e80473b9d9fa693de644c35a6d","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I5d105006fad6e7e80473b9d9fa693de644c35a6d","subject":"Ubuntu: define implied VLAN parent interfaces in networkd","status":"MERGED","created":"2021-04-26 11:34:06.000000000","updated":"2021-04-29 12:14:10.000000000","submitted":"2021-04-29 12:12:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787960","meta_rev_id":"34adca925d4d11eba32f069a960ada339f2df296","_number":787960,"virtual_id_number":787960,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-29 12:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-04-28 16:37:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-04-29 06:29:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-04-29 06:29:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 12:19:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 16:28:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-04-28 16:37:27.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-04-29 06:29:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"19fb07a6ae05c4d0742614c42535d1e1c55f470d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-26 11:34:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"190f5c89587060ad2186a0536980391ff934f1fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-26 11:41:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5127ce1ab42e508f1e03cdc5d58c64323ab53da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:19:16.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1325873f1eb54f5a9daca4cb6498dea4 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64bb446fc7e486c84ee4af2a64b9bb6 : FAILURE in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4da6b90a524f4640b1b92fe8917eb14d : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d7eb9403f3f42bfbb5cd55adb3afd21 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66f6f691e74e47e7a936dcb091bb7dbf : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c3094a5e4a7414b93861b616c2958c5 : SUCCESS in 7m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce2556a191184ec199e8fffac8c6909c : FAILURE in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/02281320d54d42309bd48fda632a55c9 : FAILURE in 5m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4fd671c511e849f7acc58b29012c39b0 : FAILURE in 7m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/82722dc47e334571a16d7d19da6fbeb8 : FAILURE in 5m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b3e208cfeca4f3a900eb4caa3b24223 : FAILURE in 6m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/83f94e4fc4254800a643bf116b7d4de2 : FAILURE in 6m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ab5fb53293e8400fb0d8544df97d3b4e : FAILURE in 6m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fda7261f376a4a3d8b41c8d5061e1d64 : FAILURE in 6m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f75a75e929ca4b2b82a240a03c376057 : FAILURE in 5m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7f58d29a401c4d639ad5d8e674f07f82 : FAILURE in 7m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/860bcd1b36e54d53820b14c3b19b0bd5 : FAILURE in 5m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e1e77e97c8e143a58134ab98fd962711 : FAILURE in 5m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/62ca61a4e481448bb9744ea4ee832951 : FAILURE in 6m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28b83d2002e94c459d295895534cf895 : FAILURE in 6m 17s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"3577960e725350494121f25a8c97c6bccb71cff3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-27 06:52:46.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"5c706a8cd0d3d9f194ddd4019cc76da93e0c4962","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 08:27:24.000000000","message":"Patch Set 3: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdefae1be3684af9a97b6e435ef8bd2f : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e455dafe3f524063baefe3ce3f89b61d : FAILURE in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a59ebee7f7b64859ba83fb0e88790ef6 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df12bf2c21b14f65a78012c09b4de120 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43b8686d3bc54af8860d692601e4b7ed : SUCCESS in 10m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c6b7542fdc4202b368c7e0d070dced : SUCCESS in 8m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2289517555d542b0996d6e744dc55a19 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec595522e8a447efb0b9b8f9190319e6 : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f83a42623f54c73bc56e90cde7776cc : SUCCESS in 18m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a3ef7787e4f24cdba036b9bc91e95163 : SUCCESS in 57m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30377846fbe7475eae01e5b58605ffa5 : SUCCESS in 43m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9c8efa7aeddb42bbbf7b67c3673a159e : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/082b96a06d924d7d9ebb495cd3691e04 : SUCCESS in 16m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc7723dae0884c75804052e66a3010a5 : SUCCESS in 11m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1e08d3dc359b4477b0cb458d7401018f : SUCCESS in 1h 32m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/32a884034e294b9ebc33d3dbf98719b2 : SUCCESS in 27m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4b1b3fb38904cd19ad30ef14c458858 : SUCCESS in 23m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/67322eeae49d44368d5761963aeadd11 : SUCCESS in 21m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d166f99ea83646aa9d8b9baa8af31373 : SUCCESS in 11m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e56cf6088e7e4abfae05634e13f5c2f8 : SUCCESS in 8m 54s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"d557fe60880051d327313d8df140a878f976ac4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 13:52:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4097debbcc55cd74676455b618834106ef0f21cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 16:07:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1aa520359db74d9ca8f5806426d9adb8 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6033560399cc4c7981ce9c1951808863 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd23ca4d87354d009e01f087ce005cba : SUCCESS in 8m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c95a6a1218214aa6b10023f0a801e77a : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14cefe8deeef4f8ea2f3c80c768a2e11 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30e0a01a13bf4c36bd898c9032fa71dd : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2404f10566024c32879eb85678cd63bc : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b8ce2f93f9b4cd5aa12ef12b831e621 : POST_FAILURE in 9m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ba499c3d4cd419c9e22dcc2bb417507 : SUCCESS in 12m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/58a85f0809c04858877a9a16f4b0128a : SUCCESS in 19m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/61141b6913b340339a393d4a71a1fd00 : SUCCESS in 1h 07m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/496a24488aa744c79289bb9981826a20 : SUCCESS in 43m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cf5b934949744c7fa040c439082107a8 : SUCCESS in 1h 13m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bee57096672e422d85a5b3d96bec1e9c : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/425edfd687e74b18a7a028866d5a0fad : SUCCESS in 11m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ca225cc989b64b47aedd9ba88f894269 : SUCCESS in 2h 09m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0931632900e8459da0e065bdcc0ca128 : SUCCESS in 30m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de991646ea0842dab6dffaed02eca797 : SUCCESS in 23m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f22b1045658f4865aba62c0f49c81efe : SUCCESS in 27m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5bca3fc2a154473786dfae48d31e8099 : SUCCESS in 13m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c9bb497304747ee9c7d8a1b19651712 : SUCCESS in 9m 29s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"05bd6ae5f9dbe114f63bd10700640050f9dec764","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-28 16:28:26.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d62b846b9e77847d9bb938b1447b19a5fddf98a7","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-28 16:37:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"049a2f44f3ca38ff2b8d02a5dec23da5e82ce259","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:32:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0154c62dbdb4ea097f1c0996aa691a4 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6dc473724e14e2c8167cac137d788c7 : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58fd46ab014042e69a213b19f2a216bc : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dca27aa1cb248c498741925c47dd969 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93ea720e575f4c08bc6d793fd3ffe977 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6540c5d1b1314b918b1f583c131af42d : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcf88a66de694d168becee7a809caf98 : SUCCESS in 5m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/134bd1124a72407bb91b792dc3b48c3d : SUCCESS in 7m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee5daeb3425242e9943c5565a1092e91 : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36edd62bd9a64ec4a05c53f789f2c367 : SUCCESS in 17m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4053e5576f9b4bc09d966018f6cc57f0 : SUCCESS in 1h 08m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f7b88e85f654ddc87b05e1ed6417618 : SUCCESS in 48m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/651aec5d96054153901da9d23f9796ec : SUCCESS in 1h 03m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4dc57373729c4508a4523cc5eb52cd50 : SUCCESS in 16m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76275b91a948494e9c5902d838ee8831 : SUCCESS in 10m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7becd52f818b4530b5795c70b00ca099 : SUCCESS in 1h 45m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/73c00c1cf7c1468f8914dfe14fa050c1 : SUCCESS in 28m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8275f6060e7d47f6acbd758e517fb358 : SUCCESS in 25m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e41b33c09a048bc8f3e69fe177ce1a5 : SUCCESS in 26m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/de34dad79b41486cb2ef9d4a233aeaff : SUCCESS in 13m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11c1527d945f4e1ba0d39fd664569a6a : SUCCESS in 10m 06s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"d0080f8243f45697ed9476c79d9ce6586508aa72","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-29 06:29:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"55e9afa05db9616767d375f42ba218b0d8baff3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 06:30:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"5228c9411446781b948767e1dca8afa7f9883889","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 08:23:07.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9b180c9440450998d1f4b22bcc4799 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/800d45c77bd044b8a94404731db24c60 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed398a20e0fa4434a7d96b232eb34cb3 : POST_FAILURE in 20m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3df5f74ccc4252b6b993154ffc986c : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1513c58d7ba44309992eda5498fc92f : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d20acfac096941c38e9c1fd2427989b0 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b6fcb4ec7524ca887cccdbe15e274db : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7872958b13d84fbd8d03809f8d22d496 : SUCCESS in 19m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8d680b248fb44597b16f70be8786329e : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ade6f35b9d44750a3d2554a3319c3a6 : FAILURE in 35m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/22f1350b50fb4c5581f5ad91825d0429 : SUCCESS in 51m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d6b942e3b08646eaa0af5e16c02453da : SUCCESS in 17m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2766f25f7dc046fdb92b2cfbcedb87ad : SUCCESS in 11m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3c3a3165407144abac111883d030d481 : SUCCESS in 1h 50m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b2196bb39d51449f863884e1a4b096bf : SUCCESS in 29m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/398fd044c02e4ffa994244d1e29135a8 : SUCCESS in 22m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/afd46f063121471dbad3a464174752f3 : SUCCESS in 23m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/62c26920206244959b21e405e2f66d4b : SUCCESS in 12m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ef5ad8aefea40c6a6525250d8c1fc17 : SUCCESS in 10m 04s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"909305bd3e66962d3e455b2381bbbed671433563","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-29 08:40:10.000000000","message":"Patch Set 4:\n\nrecheck\n\nvbmc failed in tenks","accounts_in_message":[],"_revision_number":4},{"id":"af652d7c41e67d2d26b47ddc8efbf7cd4f7b948d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 10:23:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dc17e65575a4d9ea0ee528aec6618d1 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d714f31955b6460d9905481e6337b5c1 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85156293fe6b4d5880759feca55a8767 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41940f87c48341fb9e5dbd6801a11428 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5a36e5606e54de8a12d9e8965e00ea1 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede3fea8fbaa470d83d30d6200222eab : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a78f8301f9c46b08888644942e89d47 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/586f9f144a4f4e9aaa3d853c0b39df9d : SUCCESS in 5m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2bd0f742ec74bc1b50319cea5e3c0f4 : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0750195e5d6a4a279095f977ac644e98 : SUCCESS in 19m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/28434b8b3bdc42b79a9f3bf7e3ffab01 : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2ce7de208fc4794830a5c1806017cf6 : SUCCESS in 48m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/db8ae839d3844efaaa0d06c26e46eeff : SUCCESS in 1h 03m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b1853e2ae0064a9086d293e83fb37b71 : SUCCESS in 18m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32df46b5a8ff476fb6b281135926dc82 : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a12e16d08c2b47c4bba60f9dad35b0e1 : SUCCESS in 1h 38m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/16b4063e3f0840cb90841106daa88599 : SUCCESS in 27m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ceef6dd143843cfa7b14bd092cafe04 : SUCCESS in 23m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/860576402817497ea7665aea897de665 : SUCCESS in 24m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4d0d29b08c86412785de442960d624c7 : SUCCESS in 12m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b04d16dce284995a987ddb4c4282b6a : SUCCESS in 10m 42s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"a8e7d24b0b5949d4c9ec383ef37c2b9bffdf34ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 10:23:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1789f2f2c00fda1edf5ac87ddcd9de661689e351","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 12:12:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab6753266a6f4525a9e2c1741a343f5f : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dbc16b35bf041cb9c37b54bbbba0e55 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b4afee6d6974d6fa558deacbddf1923 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca03995eed52449dbf07f41903a92bd1 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae9ba71f0e3b435ebc8e2efe9d0aa518 : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/910e8e3741fc480f8e0feaf2fd2d5ee0 : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/50e6f36c6f1943608b420b6c513e82fc : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c921a0c6094f44d99641c9de2f309009 : SUCCESS in 18m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fea09f229d99498cb84c265aa3120ca5 : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03fd69f900c149978e6cb9eeb083476a : SUCCESS in 43m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/eeede4f5bf774bec9c8d8f319b40bce4 : SUCCESS in 1h 03m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/460daae32ddc4005a910f17c2754ce87 : SUCCESS in 16m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5608d71b95274dd48528f3d8f9c91457 : SUCCESS in 13m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6b571d13875c4e00bc1f5bd78e9fed9e : SUCCESS in 1h 47m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7317c60d13364d919d2d4eb0866513a7 : SUCCESS in 26m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d75df08d01ba4ddea1382e312a60447f : SUCCESS in 23m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1dc80f5f9e354342bd1fb5f8255d2379 : SUCCESS in 25m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/760e8cadf1234ce5908ca9dc9cb45e72 : SUCCESS in 13m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e6270ccd2f347cb8660e1954bf8c792 : SUCCESS in 9m 38s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"43b0d50eb1c02c0cda1b7d5eb75efc95c431f19b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 12:12:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"34adca925d4d11eba32f069a960ada339f2df296","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 12:14:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85c8c5572ba94e4fa60cd8fee60a2f85 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cf7e8f6a19c4d2aa84613210e1c5511 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"1ae12dc6f8e5375ca7ee7023af68d8004b2b0410","revisions":{"59d8b5be4c6315b7db13601ee32e2b8663c851f1":{"kind":"REWORK","_number":1,"created":"2021-04-26 11:34:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/787960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/787960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/787960/1"}}},"commit":{"parents":[{"commit":"8eb7e053db140ee4a42e35ccf44f328be8f0afce","subject":"Merge \"Ubuntu: support policy-based routing in systemd-networkd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8eb7e053db140ee4a42e35ccf44f328be8f0afce"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-26 10:43:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-26 11:33:58.000000000","tz":60},"subject":"Ubuntu: define implied VLAN parent interfaces in networkd","message":"Ubuntu: define implied VLAN parent interfaces in networkd\n\nOn CentOS, when using the MichaelRigart.interfaces role, defining a VLAN\ninterface implies the existence of its parent interface. This is handled\nvia the CentOS network interface scripts. With systemd-networkd, the\nparent interface is not implicit, and must be configured. The parent\ninterface network configuration file needs to contain this:\n\n    [Network]\n    VLAN \u003d eth0.42\n\nThis change adds support for these implied parent interfaces. It also\nremoves the Broadcast \u003d true option, which was added in an attempt to\nfix an issue but is not supported in the [Network] section, and\ngenerates a warning:\n\n    Unknown key name \u0027Broadcast\u0027 in section \u0027Network\u0027, ignoring.\n\nStory: 2004960\nTask: 42368\nTask: 42367\n\nChange-Id: I5d105006fad6e7e80473b9d9fa693de644c35a6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59d8b5be4c6315b7db13601ee32e2b8663c851f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/59d8b5be4c6315b7db13601ee32e2b8663c851f1"}]},"branch":"refs/heads/master"},"34f168dacd2c7d8f258bc03dd9131cf86c34057f":{"kind":"REWORK","_number":2,"created":"2021-04-26 11:41:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/787960/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/787960/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/787960/2"}}},"commit":{"parents":[{"commit":"8eb7e053db140ee4a42e35ccf44f328be8f0afce","subject":"Merge \"Ubuntu: support policy-based routing in systemd-networkd\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8eb7e053db140ee4a42e35ccf44f328be8f0afce"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-26 10:43:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-26 11:41:27.000000000","tz":60},"subject":"Ubuntu: define implied VLAN parent interfaces in networkd","message":"Ubuntu: define implied VLAN parent interfaces in networkd\n\nOn CentOS, when using the MichaelRigart.interfaces role, defining a VLAN\ninterface implies the existence of its parent interface. This is handled\nvia the CentOS network interface scripts. With systemd-networkd, the\nparent interface is not implicit, and must be configured. The parent\ninterface network configuration file needs to contain this:\n\n    [Network]\n    VLAN \u003d eth0.42\n\nThis change adds support for these implied parent interfaces. It also\nremoves the Broadcast \u003d true option, which was added in an attempt to\nfix an issue but is not supported in the [Network] section, and\ngenerates a warning:\n\n    Unknown key name \u0027Broadcast\u0027 in section \u0027Network\u0027, ignoring.\n\nStory: 2004960\nTask: 42368\nTask: 42367\n\nChange-Id: I5d105006fad6e7e80473b9d9fa693de644c35a6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34f168dacd2c7d8f258bc03dd9131cf86c34057f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34f168dacd2c7d8f258bc03dd9131cf86c34057f"}]},"branch":"refs/heads/master"},"ff54127c19ea65a77a1bc2c7c55c5c673ad4f5ed":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-27 06:52:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/60/787960/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/787960/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/787960/3"}}},"commit":{"parents":[{"commit":"7411e624dea77a1b94d91bd37c9a388aa5011bb4","subject":"Use released version of mrlesmithjr.mdadm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7411e624dea77a1b94d91bd37c9a388aa5011bb4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-26 10:43:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-27 06:52:46.000000000","tz":0},"subject":"Ubuntu: define implied VLAN parent interfaces in networkd","message":"Ubuntu: define implied VLAN parent interfaces in networkd\n\nOn CentOS, when using the MichaelRigart.interfaces role, defining a VLAN\ninterface implies the existence of its parent interface. This is handled\nvia the CentOS network interface scripts. With systemd-networkd, the\nparent interface is not implicit, and must be configured. The parent\ninterface network configuration file needs to contain this:\n\n    [Network]\n    VLAN \u003d eth0.42\n\nThis change adds support for these implied parent interfaces. It also\nremoves the Broadcast \u003d true option, which was added in an attempt to\nfix an issue but is not supported in the [Network] section, and\ngenerates a warning:\n\n    Unknown key name \u0027Broadcast\u0027 in section \u0027Network\u0027, ignoring.\n\nStory: 2004960\nTask: 42368\nTask: 42367\n\nChange-Id: I5d105006fad6e7e80473b9d9fa693de644c35a6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff54127c19ea65a77a1bc2c7c55c5c673ad4f5ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff54127c19ea65a77a1bc2c7c55c5c673ad4f5ed"}]},"branch":"refs/heads/master","description":"Rebase"},"1ae12dc6f8e5375ca7ee7023af68d8004b2b0410":{"kind":"REWORK","_number":4,"created":"2021-04-28 13:52:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/60/787960/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/60/787960/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/60/787960/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/60/787960/4"}}},"commit":{"parents":[{"commit":"7411e624dea77a1b94d91bd37c9a388aa5011bb4","subject":"Use released version of mrlesmithjr.mdadm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7411e624dea77a1b94d91bd37c9a388aa5011bb4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-26 10:43:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-28 13:50:16.000000000","tz":60},"subject":"Ubuntu: define implied VLAN parent interfaces in networkd","message":"Ubuntu: define implied VLAN parent interfaces in networkd\n\nOn CentOS, when using the MichaelRigart.interfaces role, defining a VLAN\ninterface implies the existence of its parent interface. This is handled\nvia the CentOS network interface scripts. With systemd-networkd, the\nparent interface is not implicit, and must be configured. The parent\ninterface network configuration file needs to contain this:\n\n    [Network]\n    VLAN \u003d eth0.42\n\nThis change adds support for these implied parent interfaces. It also\nremoves the Broadcast \u003d true option, which was added in an attempt to\nfix an issue but is not supported in the [Network] section, and\ngenerates a warning:\n\n    Unknown key name \u0027Broadcast\u0027 in section \u0027Network\u0027, ignoring.\n\nStory: 2004960\nTask: 42368\nTask: 42367\n\nChange-Id: I5d105006fad6e7e80473b9d9fa693de644c35a6d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ae12dc6f8e5375ca7ee7023af68d8004b2b0410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ae12dc6f8e5375ca7ee7023af68d8004b2b0410"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
