)]}'
{"id":"openstack%2Fkayobe~788083","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I804fafabcfc7c7385dbc7508d0673fabdb750ecb","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I804fafabcfc7c7385dbc7508d0673fabdb750ecb","subject":"Document Update Specify Branch When Cloning","status":"ABANDONED","created":"2021-04-26 21:03:36.000000000","updated":"2021-04-30 19:23:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c675a21cf88e5ac5c550b769d85387f1dd649506","_number":788083,"virtual_id_number":788083,"owner":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-26 21:13:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":-1,"date":"2021-04-30 08:05:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 21:13:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 22:27:23.000000000","updated_by":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-04-30 08:05:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"cb54f935bee209e91056e2f84b1614fae9cebf26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-04-26 21:03:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b55e92c9f6c19e02b3840960cb15d5c2f3343b61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 21:13:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88fc14a950254435b41c12dbc58620ca : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b479dad379d1424296986db6cb1a4ce9 : SUCCESS in 8m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"5f0996322371918e879ee92e1991aca1ecd20334","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-04-26 22:27:23.000000000","message":"Patch Set 1:\n\nHi Pierre,\nCould you please look into this.\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"f50d4936d8b64e1165585e121b94fbe264128e49","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-30 08:05:07.000000000","message":"Patch Set 1: Code-Review-1\n\nSee comment in Victoria patch.","accounts_in_message":[],"_revision_number":1},{"id":"c675a21cf88e5ac5c550b769d85387f1dd649506","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-04-30 19:23:52.000000000","message":"Abandoned\n\nMore suitable patch @ https://review.opendev.org/c/openstack/kayobe/+/788702","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41a19135d498d3edabd4b60eb4cdefcfc0785e77","revisions":{"41a19135d498d3edabd4b60eb4cdefcfc0785e77":{"kind":"REWORK","_number":1,"created":"2021-04-26 21:03:36.000000000","uploader":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"ref":"refs/changes/83/788083/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/788083/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/788083/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/788083/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/788083/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/788083/1"}}},"commit":{"parents":[{"commit":"d633f938dddde58f5504663616f71dbb207aeb0a","subject":"Use released version of mrlesmithjr.mdadm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d633f938dddde58f5504663616f71dbb207aeb0a"}]}],"author":{"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","date":"2021-04-26 20:48:19.000000000","tz":330},"committer":{"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","date":"2021-04-26 20:48:19.000000000","tz":330},"subject":"Document Update Specify Branch When Cloning","message":"Document Update Specify Branch When Cloning\n\nThe current installation document does not instruct to specify\nthe branch when cloning the source code to install from source.\n\nThis patch adds the stable branch name to the git clone command.\n\nStory: 2008851\nTask: 42366\nChange-Id: I804fafabcfc7c7385dbc7508d0673fabdb750ecb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41a19135d498d3edabd4b60eb4cdefcfc0785e77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41a19135d498d3edabd4b60eb4cdefcfc0785e77"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
