)]}'
{"id":"openstack%2Fkayobe~788084","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ie2a3a6f6458195a707e6650958ac7463c0ad3afd","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"Ie2a3a6f6458195a707e6650958ac7463c0ad3afd","subject":"Document Update Specify Branch When Cloning","status":"ABANDONED","created":"2021-04-26 21:04:01.000000000","updated":"2021-04-30 19:26:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e281f1d9ee5e7f2a85db57c1eab58a54f1c5c130","_number":788084,"virtual_id_number":788084,"owner":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-26 21:18:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":-1,"date":"2021-04-30 08:01:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 21:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 22:27:34.000000000","updated_by":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-04-30 08:01:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"0de16d3c50f6e2b23ea4d546a2d191774923b467","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-04-26 21:04:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f1d7350a93eb4fd84a679db3e6639d6c4f421f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-04-26 21:08:37.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e639f4a8319ed9725cbf1b6a624b548eb1891728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 21:18:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b812714a019424aafb50a3e76cf1ae4 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb49fb1bda94165b7169a309fa4ea4c : SUCCESS in 7m 55s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"5c8a568dbcb48ae8164b24d049747bbcf28be63d","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-04-26 22:27:34.000000000","message":"Patch Set 2:\n\nHi Pierre,\nCould you please look into this.\nThanks","accounts_in_message":[],"_revision_number":2},{"id":"ac8c44a3529956dd5c04e9257e156b2541cb5399","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-30 08:01:18.000000000","message":"Patch Set 2: Code-Review-1\n\nThank you for your change, Buddhika. Seeing the different code in each branch made me realise we should make this more automatic, but I didn\u0027t want to ask you to work out how to do that as it is a bit more involved than a simple docs update. There are also a few other places where we need to use the correct branch.\n\nI have proposed this patch, which we can backport to all branches. https://review.opendev.org/c/openstack/kayobe/+/788702 Please review it.","accounts_in_message":[],"_revision_number":2},{"id":"7180ab5363c07bf48fc027e6d115164bcacc92b4","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-04-30 19:22:34.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e Thank you for your change, Buddhika. Seeing the different code in each branch made me realise we should make this more automatic, but I didn\u0027t want to ask you to work out how to do that as it is a bit more involved than a simple docs update. There are also a few other places where we need to use the correct branch.\n\u003e \n\u003e I have proposed this patch, which we can backport to all branches. https://review.opendev.org/c/openstack/kayobe/+/788702 Please review it.\n\nDear Mark, I really did not have any idea that the document section supports variables to be used. Rather than separate patches for each version, using variables should be easier and cleaner","accounts_in_message":[],"_revision_number":2},{"id":"e281f1d9ee5e7f2a85db57c1eab58a54f1c5c130","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-04-30 19:26:11.000000000","message":"Abandoned\n\nMore suitable patch @ https://review.opendev.org/c/openstack/kayobe/+/788702","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5120d883ea40511f6db5c3d3ab425cf7dfdec0cc","revisions":{"1363518b3787efef23390f8c6adaa965b8a6a25b":{"kind":"REWORK","_number":1,"created":"2021-04-26 21:04:01.000000000","uploader":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"ref":"refs/changes/84/788084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/788084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/788084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/788084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/788084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/788084/1"}}},"commit":{"parents":[{"commit":"7130772296ca2174b2eb49a0f33cba1c6a5d87bb","subject":"Merge \"Fix overcloud provision with fast track enabled\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7130772296ca2174b2eb49a0f33cba1c6a5d87bb"}]}],"author":{"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","date":"2021-04-26 20:48:39.000000000","tz":330},"committer":{"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","date":"2021-04-26 20:48:39.000000000","tz":330},"subject":"Document Update Specify Branch When Cloning","message":"Document Update Specify Branch When Cloning\n\nThe current installation document does not instruct to specify\nthe branch when cloning the source code to install from source.\n\nThis patch adds the stable branch name to the git clone command.\n\nStory: 2008851\nTask: 42366\nChange-Id: Ie2a3a6f6458195a707e6650958ac7463c0ad3afd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1363518b3787efef23390f8c6adaa965b8a6a25b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1363518b3787efef23390f8c6adaa965b8a6a25b"}]},"branch":"refs/heads/stable/victoria"},"5120d883ea40511f6db5c3d3ab425cf7dfdec0cc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-26 21:08:37.000000000","uploader":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"ref":"refs/changes/84/788084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/788084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/788084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/788084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/788084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/788084/2"}}},"commit":{"parents":[{"commit":"bf10fce87a47c3ee732c6fd8c8a4d2951ba805ee","subject":"Use released version of mrlesmithjr.mdadm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf10fce87a47c3ee732c6fd8c8a4d2951ba805ee"}]}],"author":{"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","date":"2021-04-26 20:48:39.000000000","tz":330},"committer":{"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","date":"2021-04-26 21:08:37.000000000","tz":0},"subject":"Document Update Specify Branch When Cloning","message":"Document Update Specify Branch When Cloning\n\nThe current installation document does not instruct to specify\nthe branch when cloning the source code to install from source.\n\nThis patch adds the stable branch name to the git clone command.\n\nStory: 2008851\nTask: 42366\nChange-Id: Ie2a3a6f6458195a707e6650958ac7463c0ad3afd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5120d883ea40511f6db5c3d3ab425cf7dfdec0cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5120d883ea40511f6db5c3d3ab425cf7dfdec0cc"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
