)]}'
{"id":"openstack%2Fkayobe~788276","triplet_id":"openstack%2Fkayobe~master~I371e8a583c439264a88dbc4bdae14c472955a063","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-11-29 12:33:36.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-29 10:03:34.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-11-29 12:33:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I371e8a583c439264a88dbc4bdae14c472955a063","subject":"Honor custom ssh args in kayobe-ansible-user.yml","status":"MERGED","created":"2021-04-27 11:57:37.000000000","updated":"2023-11-29 12:34:44.000000000","submitted":"2023-11-29 12:33:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"788276","meta_rev_id":"6a1d4dcfddddfeb7c84bc69ca74ecee5758822f7","_number":788276,"virtual_id_number":788276,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-29 12:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2023-11-29 10:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-31 10:30:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-11-29 10:03:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-11-29 10:07:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-08 14:37:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 13:51:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 08:28:21.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-05-10 07:59:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-07-27 07:46:30.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-08-27 00:11:02.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-11-29 10:07:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c350416f4e730ddd56e1863e3c319d062830a476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-27 11:57:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db0fa3128e8b71e4b52de440dcba57812c3dd4e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 13:51:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fc85d32417b48688f32972e1b1acebd : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/221613eee4ce431883b80d95d0c54068 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c89695ccdb03430db05ec0c7c0f71c06 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/605ac68907aa4959aa07a1733446108a : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7ab8489c4d545319fccb4bd2d1d30b0 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39d2b6cc3c674c089f717d97e6fdcdd1 : SUCCESS in 8m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28a1b956f4464664a725cf35361205c9 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d56bd34f34d459da94a0cf6ad4f47b3 : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5471128d06974c92a88625c6ab5432a3 : SUCCESS in 18m 05s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1906c9c74b444d5f8d02027e89e7480d : SUCCESS in 1h 01m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/950c47a8b0294dc7b035ec692f0c33dd : SUCCESS in 47m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/eba21e2aa2c04735a37401a106ed3de6 : FAILURE in 9m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bff69ae777c64f40866c745a2ed3ab8f : SUCCESS in 13m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f27d201822e4feaa2ea57ee391be979 : SUCCESS in 10m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f5bd5dbd93c54290932b1707a303d438 : SUCCESS in 1h 48m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6da7929df6fa46f7976b4994f17945a3 : SUCCESS in 30m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b68dfce9f087409cb7c04da45f2cdc12 : SUCCESS in 22m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e8d622b153344d759606d72ba5ea5740 : SUCCESS in 26m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c510a74cba5d496e934c17e51e4b5034 : SUCCESS in 11m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a435666adae45f0960efb176911dfe4 : SUCCESS in 10m 23s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"47fbf364fa09cf3b1424ec61939301555a6f11d5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 08:28:21.000000000","message":"Patch Set 1: Code-Review-1\n\nThe TLS job failed.\n\nTASK [Check whether the host is accessible via SSH] ****************************\ntask path: /home/zuul/kayobe-venv/share/kayobe/ansible/kayobe-ansible-user.yml:15\n\u003c192.168.33.3\u003e ESTABLISH SSH CONNECTION FOR USER: zuul\n\u003c192.168.33.3\u003e SSH: EXEC ssh -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"zuul\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/a3f63c1ce5 192.168.33.3 \u0027/bin/sh -c \u0027\"\u0027\"\u0027echo ~zuul \u0026\u0026 sleep 0\u0027\"\u0027\"\u0027\u0027\n\u003c192.168.33.3\u003e (0, b\u0027/home/zuul\\n\u0027, b\u0027\u0027)\n\u003c192.168.33.3\u003e ESTABLISH SSH CONNECTION FOR USER: zuul\n\u003c192.168.33.3\u003e SSH: EXEC ssh -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"zuul\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/a3f63c1ce5 192.168.33.3 \u0027/bin/sh -c \u0027\"\u0027\"\u0027( umask 77 \u0026\u0026 mkdir -p \"` echo /home/zuul/.ansible/tmp `\"\u0026\u0026 mkdir \"` echo /home/zuul/.ansible/tmp/ansible-tmp-1619525818.9171584-31265-149696984233076 `\" \u0026\u0026 echo ansible-tmp-1619525818.9171584-31265-149696984233076\u003d\"` echo /home/zuul/.ansible/tmp/ansible-tmp-1619525818.9171584-31265-149696984233076 `\" ) \u0026\u0026 sleep 0\u0027\"\u0027\"\u0027\u0027\n\u003c192.168.33.3\u003e (0, b\u0027ansible-tmp-1619525818.9171584-31265-149696984233076\u003d/home/zuul/.ansible/tmp/ansible-tmp-1619525818.9171584-31265-149696984233076\\n\u0027, b\u0027\u0027)\nUsing module file /home/zuul/kayobe-venv/lib/python3.6/site-packages/ansible/modules/command.py\n\u003c192.168.33.3\u003e PUT /home/zuul/.ansible/tmp/ansible-local-311713_je78la/tmple_6k22m TO /home/zuul/.ansible/tmp/ansible-tmp-1619525818.9171584-31265-149696984233076/AnsiballZ_command.py\n\u003c192.168.33.3\u003e SSH: EXEC sftp -b - -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"zuul\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/a3f63c1ce5 \u0027[192.168.33.3]\u0027\n\u003c192.168.33.3\u003e (0, b\u0027sftp\u003e put /home/zuul/.ansible/tmp/ansible-local-311713_je78la/tmple_6k22m /home/zuul/.ansible/tmp/ansible-tmp-1619525818.9171584-31265-149696984233076/AnsiballZ_command.py\\n\u0027, b\u0027\u0027)\n\u003c192.168.33.3\u003e ESTABLISH SSH CONNECTION FOR USER: zuul\n\u003c192.168.33.3\u003e SSH: EXEC ssh -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"zuul\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/a3f63c1ce5 192.168.33.3 \u0027/bin/sh -c \u0027\"\u0027\"\u0027chmod u+x /home/zuul/.ansible/tmp/ansible-tmp-1619525818.9171584-31265-149696984233076/ /home/zuul/.ansible/tmp/ansible-tmp-1619525818.9171584-31265-149696984233076/AnsiballZ_command.py \u0026\u0026 sleep 0\u0027\"\u0027\"\u0027\u0027\n\u003c192.168.33.3\u003e (0, b\u0027\u0027, b\u0027\u0027)\n\u003c192.168.33.3\u003e ESTABLISH SSH CONNECTION FOR USER: zuul\n\u003c192.168.33.3\u003e SSH: EXEC ssh -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"zuul\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/a3f63c1ce5 -tt 192.168.33.3 \u0027/bin/sh -c \u0027\"\u0027\"\u0027/opt/kayobe/venvs/kayobe/bin/python /home/zuul/.ansible/tmp/ansible-tmp-1619525818.9171584-31265-149696984233076/AnsiballZ_command.py \u0026\u0026 sleep 0\u0027\"\u0027\"\u0027\u0027\n\u003c192.168.33.3\u003e (127, b\u0027/bin/sh: /opt/kayobe/venvs/kayobe/bin/python: No such file or directory\\r\\n\u0027, b\u0027Shared connection to 192.168.33.3 closed.\\r\\n\u0027)\n\u003c192.168.33.3\u003e Failed to connect to the host via ssh: Shared connection to 192.168.33.3 closed.\n\u003c192.168.33.3\u003e ESTABLISH SSH CONNECTION FOR USER: zuul\n\u003c192.168.33.3\u003e SSH: EXEC ssh -C -o ControlMaster\u003dauto -o ControlPersist\u003d60s -o KbdInteractiveAuthentication\u003dno -o PreferredAuthentications\u003dgssapi-with-mic,gssapi-keyex,hostbased,publickey -o PasswordAuthentication\u003dno -o \u0027User\u003d\"zuul\"\u0027 -o ConnectTimeout\u003d10 -o ControlPath\u003d/home/zuul/.ansible/cp/a3f63c1ce5 192.168.33.3 \u0027/bin/sh -c \u0027\"\u0027\"\u0027rm -f -r /home/zuul/.ansible/tmp/ansible-tmp-1619525818.9171584-31265-149696984233076/ \u003e /dev/null 2\u003e\u00261 \u0026\u0026 sleep 0\u0027\"\u0027\"\u0027\u0027\n\u003c192.168.33.3\u003e (0, b\u0027\u0027, b\u0027\u0027)\nfatal: [controller0]: FAILED! \u003d\u003e {\n    \"changed\": false,\n    \"module_stderr\": \"Shared connection to 192.168.33.3 closed.\\r\\n\",\n    \"module_stdout\": \"/bin/sh: /opt/kayobe/venvs/kayobe/bin/python: No such file or directory\\r\\n\",\n    \"msg\": \"The module failed to execute correctly, you probably need to set the interpreter.\\nSee stdout/stderr for the exact error\",\n    \"rc\": 127\n}\n\nUnsure why the others passed.","accounts_in_message":[],"_revision_number":1},{"id":"cf1b086dc4038a44884fe9fac0b82204ba670185","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 08:30:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"337d33f7db60922a83ac39fff2a7c4d251e6d7be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-28 16:20:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"38e115eaa2a0c441b184167b4e454b430c009ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:27:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40e6d5dfed8c4fa48fe5d94c520ac428 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b849210ec642c59e66d518f83b70c7 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d104166443a14a4da716b8a60709107b : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b45e9f7308d4e7a96cde33f0c08ac55 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3836a53e71a548e093977db910533a86 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbdff5de8b974c91a40071b3b8914d87 : SUCCESS in 10m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/425b4b5f5e74480eaa409e7935870f2e : SUCCESS in 8m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3596011fd1f94a439d66c9d90fea7c9c : SUCCESS in 15m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a94d7c00237543a4981cd19a9f91d1ec : SUCCESS in 20m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/120aee75c5d3495b8d8ea26581bf6480 : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd1f632f4c8145f49d3f0496c5f1b66c : SUCCESS in 48m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e0d23c6561d94ca78c390245b1c1f4ed : SUCCESS in 56m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/30dd08842dea4862a3542ea57046dcb8 : SUCCESS in 20m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8710cb4180fc4968825bade7ef1db58f : SUCCESS in 24m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a2335c59c0a2477f9bbb755885b4d60c : SUCCESS in 1h 59m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dcb850d6a0a9437a91013eeaff09ff21 : SUCCESS in 28m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d24b38403474dadb968db685946393f : SUCCESS in 33m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e7d430c40cb94f749ba0890c12849754 : SUCCESS in 24m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0451ba74cb5d408ba972ffae1799f0ac : SUCCESS in 11m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5dfdca5a9d58403690fc52629b0d3f9c : SUCCESS in 12m 44s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"b282e802c78526c9ba94f2b5986a1fdf361fbfee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-29 08:51:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"af3e70a65afd48780d320f5ae58b022de6eae231","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-10 07:59:25.000000000","message":"Patch Set 2: Code-Review-1\n\nWe could add a release note for this.","accounts_in_message":[],"_revision_number":2},{"id":"57f8c02f212bb99064335ffc7886890ebddd85f4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-10 07:59:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"96df9dd69f5111a3922ad624123bb711aa5e39b3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-27 07:46:30.000000000","message":"Patch Set 2:\n\nIf there\u0027s a change in behaviour, then probably it doesn\u0027t fit the TrivialFix category - reno needed - as Pierre mentioned.","accounts_in_message":[],"_revision_number":2},{"id":"68563f1f6761f5c60d6d0fa318236cf898dffd05","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-12-14 10:19:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a958a2b51df9644900e123e45042636c5402f05","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-27 00:11:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da41e66dca5c7ca0968bc54be492fd3be60cc30b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-08-31 09:58:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7f28e4e838afef323303f42b23b43a03f0af89b5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-31 10:30:52.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"00fcf76b5704f42ee525897f35081d90ab84c9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 12:09:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef12761047054b089f979aecd43b637f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f5cbe31cd6c4164a5e00554c3c9267c : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94a13464932c4f19967ba14029a99ea5 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47a51155ba9f4da895bd2c9b01396a64 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d9192a866954f58bb97385213a4bcc8 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa8e8b389f1e48f58be9a733a82addda : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8f2144e5003436081b2b9ad7057282e : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a3fcfce8c424058aa694fc63440270c : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d6d957e2e8445d69126ea01cdaafcf5 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b33d99248594ee58081336731f6d91e : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c896844d248f4ed88f217461940d3eaa : SUCCESS in 14m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/05a1eb96b4714e748e4f184e1bbea524 : SUCCESS in 12m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/43407571bb3e47c7acd205c015be92fb : SUCCESS in 57m 23s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c6cbc7baa2d841f4af4b1e58c9227ba4 : SUCCESS in 56m 23s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78be3b13c68c4d58b6efadd431b20dba : SUCCESS in 53m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8286dcbe4bbf4348b9388bf79154c66e : SUCCESS in 1h 05m 34s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e44a05b882c4469688903317f75d48ef : SUCCESS in 18m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7d36895b087248ceac16a8ede8ca73af : SUCCESS in 17m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d25d325a4af84e9cb66b5f84cc4e0962 : SUCCESS in 16m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/686b430aa6744356a2337defc22d5f46 : SUCCESS in 2h 05m 35s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08b171dca050434db75204f34dfa0044 : SUCCESS in 2h 02m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2f89194bff474fcabef1e70a55303e9e : SUCCESS in 13m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee3aa3cb1bb8422ca046397f30424fd9 : SUCCESS in 14m 23s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a56890d4cfcc4bf6a0c0fb05e9244a93 : SUCCESS in 37m 55s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2d61c922bd834455829026aafe0c8279 : SUCCESS in 30m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8130a39d1cc74052aec4f2a1a17ea20c : SUCCESS in 33m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/07bda7651f024469a08613dabaeaf2d9 : SUCCESS in 13m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/847c4b4f5afb4af995de0d01d56028db : SUCCESS in 11m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/71f7631a768140d8ad276ea432323f30 : SUCCESS in 35m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2404a531c5ef404d988c439fdae4fb8e : SUCCESS in 28m 16s","accounts_in_message":[],"_revision_number":3},{"id":"94fa1dea542babafd053715b1d2e081e08076321","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-08 14:37:02.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"61f2bb27a03d06c1fa81778710bf75d703aaed44","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-29 10:03:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8a8034f26606e2e831148c8b5abf8096c423df8d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-11-29 10:07:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"76a9592a0ee7a78802fc27f6333766b82163f831","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 10:08:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"148c06ccb577c9e08079cace3a856e4853112b02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 12:33:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16890cec7d924e96ac8b8487584e8bca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ac9bc325dd4adb89f4f953efdf7a13 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a464b065b22747cc8ce6b1d6acf7ab8d : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fc754adfa06c4f678b108342b2157114 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98a61883a13a4aa0ad78076ae5d564c5 : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/40d30b5fe52d4f78ae7a6adfa98a1c9a : SUCCESS in 10m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/05fd8e5107db4c9d9a34e6b68802741f : SUCCESS in 45m 57s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6519f5b6474c4999926da89d3118a94b : SUCCESS in 51m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/81e2d0fa73ae45d3a8aabd26d7c08a9e : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ceb24342d06444f9b27e5e0bdab21800 : SUCCESS in 17m 59s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0905d77d2ecb41d2bd4d436edcd9ea9c : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4b59d6b843884eb7ac374c7b960f5945 : SUCCESS in 2h 20m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c946153f7e9e4a25b1de3712d3c5dad4 : SUCCESS in 2h 07m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a615c36d9c9142f39857352332c76c26 : SUCCESS in 16m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e3f5ac471e54bcfbb2540f1a8e2defb : SUCCESS in 14m 24s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e593d2bcabef475594c7c0303214db7d : SUCCESS in 26m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ac2d2e0f8474c7b8b58b4070807b4a8 : SUCCESS in 27m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9d20835785174abca21e4959559eadd2 : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1c9c6894f274423be1a9b0a35101952 : SUCCESS in 11m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4aefd6f2507b436cb598e1087d13e17a : SUCCESS in 15m 05s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf831e1f157c4fe4b377138e1d79d02c : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":3},{"id":"12ef907ab55e21f7cee238e790c708be36ee21e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 12:33:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"6a1d4dcfddddfeb7c84bc69ca74ecee5758822f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 12:34:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4dc05a28cae4d999b594300efe29414\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c369568e082a4954a79168baa32d3e33 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/572fc2ea93a7485ca4c79f980cc60103 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ffbd1a7833b69b9dfdd56e5e4ff71e9bad925c8b","revisions":{"0130122c3ed89a27b69831e53a52c2feb89ae3e6":{"kind":"REWORK","_number":1,"created":"2021-04-27 11:57:37.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/76/788276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/788276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/788276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/788276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/788276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/788276/1"}}},"commit":{"parents":[{"commit":"7411e624dea77a1b94d91bd37c9a388aa5011bb4","subject":"Use released version of mrlesmithjr.mdadm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7411e624dea77a1b94d91bd37c9a388aa5011bb4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-04-27 10:32:49.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-04-27 11:55:29.000000000","tz":60},"subject":"Honor custom ssh args in kayobe-ansible-user.yml","message":"Honor custom ssh args in kayobe-ansible-user.yml\n\nUsing the command module directly will honor ssh args. The slight change in\nbehaviour is that it will print an unreachable message and show an unreachable\ntask in the summary.\n\nTrivialFix\nChange-Id: I371e8a583c439264a88dbc4bdae14c472955a063\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0130122c3ed89a27b69831e53a52c2feb89ae3e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0130122c3ed89a27b69831e53a52c2feb89ae3e6"}]},"branch":"refs/heads/master"},"08fdec14051bab9f68a72a737bc76588b90e9914":{"kind":"REWORK","_number":2,"created":"2021-04-28 16:20:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/76/788276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/788276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/788276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/788276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/788276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/788276/2"}}},"commit":{"parents":[{"commit":"7411e624dea77a1b94d91bd37c9a388aa5011bb4","subject":"Use released version of mrlesmithjr.mdadm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7411e624dea77a1b94d91bd37c9a388aa5011bb4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-04-27 10:32:49.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-04-28 16:20:00.000000000","tz":0},"subject":"Honor custom ssh args in kayobe-ansible-user.yml","message":"Honor custom ssh args in kayobe-ansible-user.yml\n\nUsing the command module directly will honor ssh args. The slight change in\nbehaviour is that it will print an unreachable message and show an unreachable\ntask in the summary.\n\nTrivialFix\nChange-Id: I371e8a583c439264a88dbc4bdae14c472955a063\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08fdec14051bab9f68a72a737bc76588b90e9914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08fdec14051bab9f68a72a737bc76588b90e9914"}]},"branch":"refs/heads/master"},"ffbd1a7833b69b9dfdd56e5e4ff71e9bad925c8b":{"kind":"REWORK","_number":3,"created":"2023-08-31 09:58:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/76/788276/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/788276/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/788276/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/788276/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/788276/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/788276/3"}}},"commit":{"parents":[{"commit":"3d5d72d53f51ff9e3b3f9a67b240ae56a72d18d7","subject":"Merge \"Speed up calls to Bifrost\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d5d72d53f51ff9e3b3f9a67b240ae56a72d18d7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-04-27 10:32:49.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-08-31 09:57:40.000000000","tz":60},"subject":"Honor custom ssh args in kayobe-ansible-user.yml","message":"Honor custom ssh args in kayobe-ansible-user.yml\n\nUsing the raw module will honor ssh args. The slight change behaviour is\nthat it will print an unreachable message and show an unreachable task\nin the summary.\n\nChange-Id: I371e8a583c439264a88dbc4bdae14c472955a063\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffbd1a7833b69b9dfdd56e5e4ff71e9bad925c8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffbd1a7833b69b9dfdd56e5e4ff71e9bad925c8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
