)]}'
{"id":"openstack%2Fkayobe~788457","triplet_id":"openstack%2Fkayobe~master~I205b40c142adcbe0deb2ea6296c246533fd49def","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I205b40c142adcbe0deb2ea6296c246533fd49def","subject":"Ubuntu: install qemu-img on seed-hypervisor","status":"MERGED","created":"2021-04-28 09:09:05.000000000","updated":"2021-05-05 11:21:45.000000000","submitted":"2021-05-05 11:20:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788457","meta_rev_id":"dba61bbb8a1e863361cbad05c66a5198df904c06","_number":788457,"virtual_id_number":788457,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 11:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-05-05 09:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-28 14:41:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-05-05 09:28:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 11:05:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 14:41:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-05-05 09:28:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"06fa7def0ba6e6f29b6e67f5c88c3cb4df3e4f9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 09:09:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6a057d91d110ae8b2971f6dfeb5fce3df99261b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 11:05:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db8cc79d50d941b1b0d0a3586bc246ce : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e62e435829374e5baec03aa39659ca08 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed964ba6967044eb92d2d18cd0747e66 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad16ca9e42c54867a5cf57370a41c570 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d57d145ae02d4c8997fe5ddfc269497f : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebbe0dd106ee49fd8b688ba8d4ab7819 : SUCCESS in 8m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5836cd6a22e94268ad21126a5c46a581 : SUCCESS in 6m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/820690183a084ab9a8c279c184cf2da2 : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/889576aaf63c49dd83621999629a19b8 : SUCCESS in 18m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6f9258611df5490184b151cb0a6e1a64 : SUCCESS in 55m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e238b7c9fa14ff2b1e62790a34a7d68 : SUCCESS in 42m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3711921781b2419096a7d3b473be36a5 : SUCCESS in 56m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5e1c832608104db8a6124a7bdac3f04a : FAILURE in 7m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/863198639e4d462e9285ee1b480398fb : SUCCESS in 11m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7de5c1f3dddd41a7ba71dc3523792e69 : SUCCESS in 1h 50m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3a72319b2ec74ec28b46217c063c7c2d : SUCCESS in 28m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a359653270c407fa29bd2a76ef5671f : SUCCESS in 26m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c8e34a25c9147d69320e9f3492e320f : SUCCESS in 23m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f65e9caca2d54c3f91f11fdb45ea65a6 : SUCCESS in 12m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7fc74b769e334a6697b2be88cffaabc5 : SUCCESS in 10m 10s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"1916f81a98afbadf65913a1b503794225f91ef18","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 12:34:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2edaf8abac3e859e5709c68acb1f89b37f07db6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 14:28:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/daf2b9721356416883916a5a05b67311 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdca1e6da27c45099c1f75adf05e9e18 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c32c7dd4ee74ec685a0337d07f45b7a : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1987df5fbf484e15a7e8a532c7ae6f28 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a5d761208a34ad9aad6b9a6bbe566fd : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a373db7aee3448599735b3b99ad54b0 : SUCCESS in 8m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fc9cc0afeb064b198233101961292a00 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4821f45907b24bc4b9b5fd1501370174 : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/459c92bf034840dc8947d01d6b1c14ec : SUCCESS in 21m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4a86fa6cf809477e8e1e597b128bbce5 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0897ce906114d5e8f053a187fe19764 : SUCCESS in 46m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/07c1e4a487ad40d2ab1e25e0258ed3d8 : SUCCESS in 1h 04m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9fb3e341b93246a0b0f5b88aa3de36b1 : SUCCESS in 17m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46a4e69b068c45e3846b9daebc7d4583 : SUCCESS in 14m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bf8c26dae2d946538dd88deaed8cd094 : SUCCESS in 1h 51m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c0ba805a4899452dba18bce40205604a : SUCCESS in 29m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea6ed671cfbe441e98aff58d299c4f19 : SUCCESS in 24m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c9b0024525104c99ae8ba51fe604b7f0 : SUCCESS in 26m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/58bd90e6a8ca49369d450c116608d48d : SUCCESS in 12m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b9914a8fc164fc1824986ee58f392a4 : SUCCESS in 10m 32s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"cf735c1c32bdb7b6c392f4e36373843b230be4b0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-28 14:41:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"35f4079ee2853e7898c1ae71d6209ccb3acc584e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-05 09:28:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1b2a0d65a1f265aa5ba331770f51ba64d07b97d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 09:28:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1277dd2db36b4289a4e588abfc2655159a773bac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 11:19:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bbfdb96af254b8bb2655f48434c09fe : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab8f89a354b2452c88ffa5f88ecfa0d6 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8295799a0d674f28a58b77868bd53803 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2738dad93d454a9150b5a49266e0a1 : SUCCESS in 8m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7147178ec97f4e96bf3fc0dc05b47438 : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd341c639d884fe49cd58d67158f4e32 : SUCCESS in 12m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c473b3ded9bc4b90b0c3bd4171e26a72 : SUCCESS in 19m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9cf66936972a442eb3b538275720d20a : SUCCESS in 57m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2a4cfb50bed4e9aaa53cdec5cfa54d4 : SUCCESS in 41m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fcf6be10b37b48ea8f95552ba3232af3 : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5f0f16ac373a4c9899938711672fd4d2 : SUCCESS in 17m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94d2353afad04e74b490d3b16dceeba0 : SUCCESS in 12m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8600f051394d452b89786cad49522948 : SUCCESS in 1h 41m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c3b91a92d0c14696bcf914c6acb23aea : SUCCESS in 26m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26b49922a1a4481b920843fad3038cb7 : SUCCESS in 47m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/970c4b9a2b3d46ebb4f29778ac11d7af : SUCCESS in 24m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e79a79ddef8c4383ab62434945080e3b : SUCCESS in 13m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53352030e8bb4588bdc63eac8904e30c : SUCCESS in 10m 41s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"393070bac3a009c2ac8b10b1aac72feb52775e74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 11:20:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dba61bbb8a1e863361cbad05c66a5198df904c06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 11:21:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04514fd0a5b64da396d82ea9d28c5178 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2cca07c7f6d2d8d5167382074170f70c97d9526","revisions":{"f2cca07c7f6d2d8d5167382074170f70c97d9526":{"kind":"REWORK","_number":1,"created":"2021-04-28 09:09:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/57/788457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/788457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/788457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/788457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/788457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/788457/1"}}},"commit":{"parents":[{"commit":"174e3994df5f9208282255ffd41be8be760b4f2d","subject":"Merge \"Use OpenStack Wallaby release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/174e3994df5f9208282255ffd41be8be760b4f2d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-28 08:55:29.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-28 08:55:29.000000000","tz":60},"subject":"Ubuntu: install qemu-img on seed-hypervisor","message":"Ubuntu: install qemu-img on seed-hypervisor\n\nOn systems without qemu-img installed, seed VM provisioning will fail\nwhen creating qcow2 disk images. This is the case with some Ubuntu\nimages. This change updates the stackhpc.libvirt-vm role to v1.14.1,\nwhich installs qemu-img.\n\nChange-Id: I205b40c142adcbe0deb2ea6296c246533fd49def\nStory: 2004960\nTask: 42330\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2cca07c7f6d2d8d5167382074170f70c97d9526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2cca07c7f6d2d8d5167382074170f70c97d9526"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
