)]}'
{"id":"openstack%2Fkayobe~788478","triplet_id":"openstack%2Fkayobe~master~I76d658b93b9195c760c3a91fe5aee96bad8d6b42","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I76d658b93b9195c760c3a91fe5aee96bad8d6b42","subject":"Ubuntu: CI: rely on os_distribution, remove interfaces pause","status":"MERGED","created":"2021-04-28 11:54:06.000000000","updated":"2021-06-09 11:26:14.000000000","submitted":"2021-06-09 11:23:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788478","meta_rev_id":"f481802f648ec153dba9109d0ab9a02c7423682e","_number":788478,"virtual_id_number":788478,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 11:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-09 09:26:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-03 14:58:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-09 09:26:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 13:47:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 16:56:19.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2021-05-03 14:58:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-09 09:26:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"77f7c0b359b293345a282936bb27d8e7d053bd47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 11:54:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff479f2e31c7738e508931c33cad70d0fa8f43f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 13:47:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/530edc07b13e407588b75c0958b30946 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac97744453e84f97bf8f3e15b4f5eb07 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e97b0dfefae429c9d84661f04463d2e : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7000ad7652bc42f69c979e270cb7638d : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5a6ec603a2b408c91eabc1668f428a2 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/039878b3b98a4510ade09b18a779001d : SUCCESS in 8m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/df00d8a479fe4c979071bc5fff9eaa94 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2b5cea4c30994c679beac9497e959e2e : SUCCESS in 10m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f5d735c5fe36435a84d35c8d9cc939cb : SUCCESS in 21m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/00926c4902d24dea8ec7755b4889c4f9 : SUCCESS in 1h 01m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca42a253047b4dab935958f2613c7b59 : SUCCESS in 45m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b30b52f2b4cb4468b8249fc3eef8f543 : SUCCESS in 1h 01m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fb1dfdb8b3f14aaabb36f3ce0b687002 : SUCCESS in 16m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91462bef0b674819bafcd373c82e7c34 : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/61fddeb1232746bba9d77c4d0452ead2 : SUCCESS in 1h 49m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e689beeef44047458be6c612198a8b8f : SUCCESS in 27m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbc4cea007cc4a32bf3143434d84d985 : FAILURE in 17m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/daf8891c5bc64c88a876b6c9a80e098b : SUCCESS in 24m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f364f930ea2b49aaba0aa9d3bae8f04b : SUCCESS in 11m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba4a66117d014faaa6492b24875650b1 : SUCCESS in 9m 55s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"184c684bf66054fda242dfa9b9939309fe59c8b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-28 14:55:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4a54b44b2dc354dcd1f1d1c461354e877ad568e0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-28 16:56:19.000000000","message":"Patch Set 2:\n\nNoticed that the previous patch built a bionic image\n\n https://zuul.opendev.org/t/openstack/build/fbc4cea007cc4a32bf3143434d84d985/log/primary/kayobe/ipa.stdout#177\n\nDo we want want to build a focal based one? Note sure if that will solve the pip install error.","accounts_in_message":[],"_revision_number":2},{"id":"c44000633bafda006158080d7f2068ee2d9e39d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 17:11:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/969ec1c4de424c288096fc87140dec42 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/048111ff0782453689f8d69b662c1e8c : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/519d64a314d54b43a1ce20546fca83e2 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/075d7856feab4b27858b8a12a4fecf51 : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10ee1f82db3b47ff935f0c5693b8102f : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a9b9fed026148c5bc1087853c5da89b : SUCCESS in 10m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee3cc61dd3b643edb501b8f5f92dccc3 : SUCCESS in 7m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c4486b08b2b9400692a1e38597e3ada6 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dc05be3ad54946cca6861b03b5a650bb : SUCCESS in 20m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/56ff7454c531443d96b5fdb3974c86fd : SUCCESS in 1h 03m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/332a472457e54fd5bb6846f9921756a8 : SUCCESS in 44m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8d8126270a6344a9856b74566869a8c9 : SUCCESS in 1h 07m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2c2b7af31575456db9a778f63dc700b2 : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb1047b491224bc2b837551b184f4358 : SUCCESS in 16m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/17150b120e6540d4aa8118e380d95cd0 : SUCCESS in 2h 06m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/347ceb4d50dd4bcc9f824b1b68f8ce6f : SUCCESS in 34m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca8ef06ae7f24d7c9fcd973772311b7b : SUCCESS in 29m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5fce2b69b31548bd9d637eeea67ab56e : SUCCESS in 31m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c0c7ee5aa3cd4a6d8f8a890d092040b1 : SUCCESS in 11m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6473c8892777471f92f5c2c58203d6bb : SUCCESS in 12m 37s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"62f0d60f8321e8a36332113e8044cd3e17cb64ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-29 10:17:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4c39226942e0f771edff82732d5fec89d518c90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 12:03:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/187cb1504c374e6dba1bfef46b9fe0a1 : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2949272b194246269dfba515bfa0bf5f : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a4e07456c4a43f7921f9f31ab024a99 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/868908876f594b4d95d24435a6c8f76a : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c0774828cc146e281e7ec87cebb1aa1 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f42a72f769943898180e3825e24841e : SUCCESS in 8m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fccd54b271cf4a318e9970a31b120664 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/387a3fa6c6f8480a8d70ee9fc97c9248 : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a1b09873efa4c969df9643ac46e69a3 : SUCCESS in 19m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fd755da750984358a661178e76912e3b : SUCCESS in 1h 12m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6bb7152633940b8b98621109eeb3241 : SUCCESS in 46m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/03f183ad36a9487ea7ba4ca1a8813428 : SUCCESS in 1h 14m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f327709039304156801224dcb63c6a56 : SUCCESS in 17m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab2e6518316f4c3ca93e0a128031e497 : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d0442c457bd8494fa2ed7bd9b0e3551b : SUCCESS in 1h 41m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/866a22dbbe314995b25e4fa8a2531875 : SUCCESS in 31m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a4a24314a884286bba50f169252df73 : SUCCESS in 28m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f4a68054bb1e4e59afd9d93229de9a49 : SUCCESS in 28m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/09b7b9d13e314d51a4024392e8ff3270 : SUCCESS in 11m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10e05bf94511470497683dd5c1a19fc1 : SUCCESS in 10m 17s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"3b9896d615a8da736251f9f73d022b7f58b56b5d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-03 14:58:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8fcbac1ebeb9c7fc86d84d3c7f5135643f7f9f69","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-09 09:26:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6f9465119a0eba102ce7c2ea322c1539bede11f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 09:27:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4dfcee4328757868cdc9c83562a8d61185c25969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:23:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65691782d9c5430eb85a1ed0b22d02df : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b6ceb8ee0df41429ed60aa40f2e6019 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c826d6ba00d402e81503e56fd8d3b79 : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c681d41faa09471f9e488271fef31d2e : SUCCESS in 9m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de7b60e8ad384801a84e1a875cd655c5 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b661552078f24f20a2189d0c684504cd : SUCCESS in 10m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1641515624534d1393d6f2651f0ae970 : SUCCESS in 21m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/24920d75572b499fa509d4bcd594eded : SUCCESS in 1h 07m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9d0a7d525684e5f8c0cd814b9d5b120 : SUCCESS in 49m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fe3e2c353abe4c86b9065be2673c93ea : SUCCESS in 1h 09m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b7d78c84c0714c09b0e47e4d07b01ef7 : SUCCESS in 18m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58ecfa31c3da4b5dbff856dd137ec4a5 : SUCCESS in 13m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/feacd41851454cfda05c9bcbdd7c9355 : SUCCESS in 1h 40m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/51310863a8c44fb8ba695189a6011581 : SUCCESS in 31m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fee96e05a81e475cbab7296a81e80d1f : SUCCESS in 26m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2cc61607ca6c4165a08f02962829de8a : SUCCESS in 27m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/df8d2cb817644c53ab1ae5ed4c8b7151 : SUCCESS in 13m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebbaf0e418bb449c8ad8c429a9abca2d : SUCCESS in 10m 19s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"6e6c2eb3b660bfca62a29673444146de4b2841d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:23:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f481802f648ec153dba9109d0ab9a02c7423682e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:26:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33593c9989734267bada107e3b60a0bd : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"111375fb1ab335a02109dde34558d3f46aeaa124","revisions":{"02084633fca52deefcc8636ebb2825eb20ee0a79":{"kind":"REWORK","_number":1,"created":"2021-04-28 11:54:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/788478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/788478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/788478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/788478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/788478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/788478/1"}}},"commit":{"parents":[{"commit":"38f363a7ea891f54cd3c3a253086595d9c2fa895","subject":"Ubuntu: simplify OS distro selection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38f363a7ea891f54cd3c3a253086595d9c2fa895"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-28 11:51:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-28 11:51:09.000000000","tz":60},"subject":"Ubuntu: CI: rely on os_distribution, remove interfaces pause","message":"Ubuntu: CI: rely on os_distribution, remove interfaces pause\n\nWe can now rely on the os_distribution variable to set sensible defaults\nfor Ubuntu. It is set automatically in kayobe-config-dev.\n\nThe interfaces pause workaround is no longer required since switching to\nsystemd-networkd.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/788475\nChange-Id: I76d658b93b9195c760c3a91fe5aee96bad8d6b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02084633fca52deefcc8636ebb2825eb20ee0a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/02084633fca52deefcc8636ebb2825eb20ee0a79"}]},"branch":"refs/heads/master"},"e1049e209e9ec51ec583dee601c0415203a9f416":{"kind":"REWORK","_number":2,"created":"2021-04-28 14:55:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/788478/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/788478/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/788478/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/788478/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/788478/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/788478/2"}}},"commit":{"parents":[{"commit":"38f363a7ea891f54cd3c3a253086595d9c2fa895","subject":"Ubuntu: simplify OS distro selection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38f363a7ea891f54cd3c3a253086595d9c2fa895"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-28 11:51:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-28 14:55:31.000000000","tz":0},"subject":"Ubuntu: CI: rely on os_distribution, remove interfaces pause","message":"Ubuntu: CI: rely on os_distribution, remove interfaces pause\n\nWe can now rely on the os_distribution variable to set sensible defaults\nfor Ubuntu. It is set automatically in kayobe-config-dev.\n\nThe interfaces pause workaround is no longer required since switching to\nsystemd-networkd.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/788475\nChange-Id: I76d658b93b9195c760c3a91fe5aee96bad8d6b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1049e209e9ec51ec583dee601c0415203a9f416"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1049e209e9ec51ec583dee601c0415203a9f416"}]},"branch":"refs/heads/master"},"111375fb1ab335a02109dde34558d3f46aeaa124":{"kind":"REWORK","_number":3,"created":"2021-04-29 10:17:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/788478/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/788478/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/788478/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/788478/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/788478/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/788478/3"}}},"commit":{"parents":[{"commit":"d6aa4ea16e917996d6009956b4edcd1ed4275c72","subject":"Ubuntu: simplify OS distro selection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d6aa4ea16e917996d6009956b4edcd1ed4275c72"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-28 11:51:09.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-29 10:17:24.000000000","tz":60},"subject":"Ubuntu: CI: rely on os_distribution, remove interfaces pause","message":"Ubuntu: CI: rely on os_distribution, remove interfaces pause\n\nWe can now rely on the os_distribution variable to set sensible defaults\nfor Ubuntu. It is set automatically in kayobe-config-dev.\n\nThe interfaces pause workaround is no longer required since switching to\nsystemd-networkd.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/788475\nChange-Id: I76d658b93b9195c760c3a91fe5aee96bad8d6b42\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/111375fb1ab335a02109dde34558d3f46aeaa124"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/111375fb1ab335a02109dde34558d3f46aeaa124"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
