)]}'
{"id":"openstack%2Fkayobe~788683","triplet_id":"openstack%2Fkayobe~master~I567e83cac4d5e80bd097b30ddefa0cc188118ad2","project":"openstack/kayobe","branch":"master","topic":"throttle","hashtags":[],"change_id":"I567e83cac4d5e80bd097b30ddefa0cc188118ad2","subject":"Replace custom serial loop with throttle","status":"MERGED","created":"2021-04-29 07:25:03.000000000","updated":"2021-05-20 10:29:40.000000000","submitted":"2021-05-20 10:28:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788683-throttle","meta_rev_id":"ad6e75e32f9dd8c7ddbca1275ecb6a8a88480ba9","_number":788683,"virtual_id_number":788683,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-20 10:28:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-05-20 08:58:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-30 08:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-05-20 08:58:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 08:42:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-04-29 09:02:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 08:58:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"4cd1e40c80118a25b51e5f05f33844b54f5d7cc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-29 07:25:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"080b91c4415ddc4c517f2456d5597b44bf5e96f7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-29 08:42:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"80c79b14d86eadc308506be6fe4c9b9960fda761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 09:02:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/863b1826b5324891b66f24f720f0a32a : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9890112144e42a1af4cc21a91cc715f : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a43b0b9f2f284f7480eb0017880edddd : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dda96a7e1b50456b8c947056c8a6dc5d : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ee7743dbb044b498925741a9e32484b : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8b5186a06534daf99192bd3c134e670 : SUCCESS in 7m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f6327dfa836f4f3599f7203f1e4af0b2 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e3bce8f1a0a4858804cebba07038ff0 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a074ce2955942d4952f9bbb91b2c1cb : SUCCESS in 19m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/14abda79c86549b0b70584e3834f0f85 : FAILURE in 6m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0cfd36490594ed9b683b4c7977f71d3 : FAILURE in 7m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d6265ae6e2c041c09f23b54c3b8fb8d5 : FAILURE in 7m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f82bf97e7008460686f8e6566cee250e : FAILURE in 8m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57056e6c9ce048e59b87d1c1256b41f0 : FAILURE in 7m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cecdffe8d2c24752954f20d95a550186 : SUCCESS in 1h 35m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/607e5bc1d3984632b6bbaa3fb1d1614d : FAILURE in 7m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ffdf10d9f2f42cfbfeb664046dd34c5 : FAILURE in 7m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/64b32a5a178d46eab3757b38d50661cf : SUCCESS in 24m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e7c6d9382ef14ebba61992548b608048 : FAILURE in 5m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd59146be5344591aef576f53e1ca2b9 : FAILURE in 7m 25s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"2d30e6c84102580062f075c4fdfd639af21c1854","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-29 09:38:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f94578dff38bd192c70b18233bb47b6be89b506b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 11:29:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85dda0d5bd9c42d4bf5a1a54edd7d9c2 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed48cd4d85941c5b208ff54d41db87a : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3727aeaec85c4275b2a5c18f16748c91 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01bf49a8f71d4e9c8e0401d0c7225d88 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/252e1082d13b42ee9a74d0ab16544218 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d5685facf448bca1e30eb3e7c92ae5 : SUCCESS in 8m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/09609c23d59b44a387fc72691239994d : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1411456605d141d0bc71185c94f00097 : SUCCESS in 10m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/635cebabf2f64d6f867ba07fb594b003 : SUCCESS in 20m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/16ee18c2125a4772a73f7701f369287c : SUCCESS in 59m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fb76b1c0ffb4caeb7053605f7471af1 : SUCCESS in 44m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/058c2a18474c4d7192f11891eef5d8ed : SUCCESS in 1h 03m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c01a4dde3d9049ac96c5893dfb4aa653 : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05776ad0b2ca4ad5a3104ceb30a907e1 : SUCCESS in 12m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4a3a704b0e1741f8b854e0e3b84bc3ea : SUCCESS in 1h 48m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b98476465f31460dab63347480f84dbf : SUCCESS in 26m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0461caaff7e84836ba22230d62271549 : SUCCESS in 23m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a31e28ffee6b4ac7ba1976acbc38c432 : SUCCESS in 21m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6138b8e8c7754ed596c4e5a187eb62e5 : SUCCESS in 12m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0f530e35f8c49f9910ee7e6b566e0af : SUCCESS in 10m 13s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"6ca8611b2457417d7ce4f346381f7f703666fbae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-30 08:05:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4d23c529279ba2972ddd15c0f9120e0976c8929c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-20 08:58:09.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7912a71fb12676bc75d9c706d69ada13ad64f22e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 08:58:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6198f4de6fa0e9d652d1dfa8e1ff2e6347f97294","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 10:28:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc304f2d6ae3459f82c12541fed6c38b : SUCCESS in 11m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3168f5d86aa64d179081a65ceee91db1 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0c922bc8b5f487fb3cdc7dfe933a52d : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e15db45333044bb990055179395ad3a : SUCCESS in 15m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/98cce8f94dc0450dbe202b305c16ccd9 : SUCCESS in 7m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0bc0fa1afe41459385f0949917330386 : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9313f7c64f7a47589dccb4dd0be5e558 : SUCCESS in 19m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1d5f43f646d4418b99636cbcfb054979 : SUCCESS in 1h 29m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fba287409e9047f09cba6bc4f2887a4d : SUCCESS in 37m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e118d9768b784d63ace95ad558ff7a35 : SUCCESS in 1h 10m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/089e6e1810a944248bf03abba28de610 : SUCCESS in 16m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ef0ae249e824619824313ebfd9f1b90 : SUCCESS in 11m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/72a23c9b83fc4f7694791dc3d2f47a3a : SUCCESS in 1h 26m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0677ab4a60874d11913344ca91591f67 : SUCCESS in 26m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85c2219b28e74abf9ec9293d28bf3215 : SUCCESS in 30m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/520ffec2ba6f4870b41cdd1383177840 : SUCCESS in 25m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ca4afb55afb04f389fc0da7ef27422af : SUCCESS in 13m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a836ed9ba2624722a1afaa3c3bfb3dc5 : SUCCESS in 10m 52s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":2},{"id":"572a62f1162f3e853d9cd04b31fb7e6ef7312358","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 10:28:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ad6e75e32f9dd8c7ddbca1275ecb6a8a88480ba9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 10:29:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0933cd41f1694c2fbca5e15fd718fbd3 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9033d28435ecfbacc941045ab470e4de3869e4e6","revisions":{"8cc7a01d3c8f76fff8e9922b007b4306b73ab8ba":{"kind":"REWORK","_number":1,"created":"2021-04-29 07:25:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/788683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/788683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/788683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/788683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/788683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/788683/1"}}},"commit":{"parents":[{"commit":"48fdca1dfcb58d537477decd6741619c9f767820","subject":"Merge \"[release] Tidy release notes for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48fdca1dfcb58d537477decd6741619c9f767820"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-29 07:24:23.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-29 07:24:23.000000000","tz":120},"subject":"Replace custom serial loop with throttle","message":"Replace custom serial loop with throttle\n\nWe can use throttle now that we require Ansible 2.9.\n\nChange-Id: I567e83cac4d5e80bd097b30ddefa0cc188118ad2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8cc7a01d3c8f76fff8e9922b007b4306b73ab8ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8cc7a01d3c8f76fff8e9922b007b4306b73ab8ba"}]},"branch":"refs/heads/master"},"9033d28435ecfbacc941045ab470e4de3869e4e6":{"kind":"REWORK","_number":2,"created":"2021-04-29 09:38:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/788683/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/788683/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/788683/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/788683/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/788683/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/788683/2"}}},"commit":{"parents":[{"commit":"48fdca1dfcb58d537477decd6741619c9f767820","subject":"Merge \"[release] Tidy release notes for Wallaby\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48fdca1dfcb58d537477decd6741619c9f767820"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-29 09:38:29.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-29 09:38:29.000000000","tz":120},"subject":"Replace custom serial loop with throttle","message":"Replace custom serial loop with throttle\n\nWe can use throttle now that we require Ansible 2.9.\n\nChange-Id: I567e83cac4d5e80bd097b30ddefa0cc188118ad2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9033d28435ecfbacc941045ab470e4de3869e4e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9033d28435ecfbacc941045ab470e4de3869e4e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
