)]}'
{"id":"openstack%2Fkayobe~788944","triplet_id":"openstack%2Fkayobe~master~I62ef7746aca4c6889526692cd0897e422937e70e","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I62ef7746aca4c6889526692cd0897e422937e70e","subject":"Ubuntu: avoid shadowing of networkd config files","status":"MERGED","created":"2021-04-30 12:54:10.000000000","updated":"2021-05-05 14:27:54.000000000","submitted":"2021-05-05 14:26:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788944","meta_rev_id":"bfe4d37fcb09b50c2a1cb4f549cd1354c198d274","_number":788944,"virtual_id_number":788944,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 14:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-05-05 09:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-03 09:30:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-05-05 09:25:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-03 09:31:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 14:25:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-03 09:30:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-05-05 09:25:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"6a23c3276dd398ab2e32cfd10b5dbbe3cf70c70e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-30 12:54:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e51a0b0e6ccf87bf50104fec2bd959b51da6826f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 14:25:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db53b65851ea427daf92a5a4d1c31834 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f06ba2b329e3485f86d15e9e74f6422b : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bee5c4ce8cf644c29cbc869b8c559cab : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f807b0c4c884c0ea654df44b4380b2e : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9514e64096b843f88f2d4ff13a30fc63 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/316df4215c524296bc5322a5ea8d13d4 : SUCCESS in 9m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84d0911fb4af4d93ad74527f74ca3988 : SUCCESS in 8m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/977ecb7af9154902ad1fa588582d7f7e : SUCCESS in 10m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/11a77cbc16ad4c24957a61664068bf28 : SUCCESS in 16m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/77309d79824f4d428cc64dafdbe67bb4 : SUCCESS in 59m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f17232f2cc62424eaa5b806139efe943 : SUCCESS in 36m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c85979a89ee84ca68762363bff013d8a : SUCCESS in 57m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d475bd0df7254ee893f3c9c71bcf0261 : SUCCESS in 14m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ced16863d76149079b94067bbe90c58e : SUCCESS in 13m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/420eee9ea81e456ea164e1a2cd54d8cf : SUCCESS in 1h 28m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/21a61faba84f4859ada1e9568442009a : SUCCESS in 41m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fdd9f0dd0314a91a9733d25f821449f : SUCCESS in 25m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/da533c60e3ec45d891a43ecca21ccd66 : SUCCESS in 20m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/43b1e5e01fda4700941beb6d0a0b4e43 : SUCCESS in 18m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/643a8805f65c4e36bb2f3285e38bbedc : SUCCESS in 9m 29s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e4c28809d98ad25742c4347e2baa977ef70211a8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-03 09:30:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9315c40c72ae10b863765ab859b51a8bbab61e57","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-03 09:31:02.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"75cfb0cd8563eda8cf4833c112b51617e0a50e35","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-05 09:25:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2379fefadd1e30bcfe8975db31b4a15373e9357f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 09:26:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a3019ba94870cd0e07a23fe9d27a693e6b9b0f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 10:52:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62823c2558dc472c8c1f68c0873f10ee : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36402c2604d8430e8ac6e1b36d5bf8df : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c58575524ce42ce8223d70780987cfa : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c98f1eb26267429d8f18d7c62eb96a9d : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d0cefaad2de54dbcabd5fc87c5e50c6f : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5f38c78cf48e4b7c97b1ded802f9fe0f : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1860c32c8e414df1b8c88d654b492ca8 : FAILURE in 8m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e9ab1f9eff43471e860a4fb13b4ff507 : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/567155550a2c43c5bba6430b2c663a97 : SUCCESS in 47m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e8ba74cdd09d4819a76201dd50f07657 : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/baefedb1871c4056bc458e55383e0230 : SUCCESS in 15m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4204f162bbe44042a71fc80bbb4d6348 : SUCCESS in 13m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bf5ad85246bf4ea9a7d2ee5767a27691 : SUCCESS in 1h 25m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3c83a8641d514745a11dd56d561313d1 : SUCCESS in 26m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4aaa5ed192f14c59bfcbcb3ca3c6b96b : SUCCESS in 23m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d1ab7aa0a9b240458e17ee843e940a87 : SUCCESS in 24m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0ebc009cac4a4b17846fc34b424d512a : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5db8ac82ceb40fc80183a80542f3606 : SUCCESS in 11m 32s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"c60dae3c9ebc72dc8a39a5f86f733cb877c0d5ef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-05 10:55:36.000000000","message":"Patch Set 1:\n\nrecheck\n\ntimeout from galaxy","accounts_in_message":[],"_revision_number":1},{"id":"62b3486f918196372f2797fa6957d531577a0212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:24:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfe4fd2c0bf541698f4ae7dff0be7d9b : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/778279a41b0848cbaf3f9f03f5555b54 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdcdc1f1437e401986d45265b19309a7 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbc33af370c74761a73494cc5a9847b2 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30a8ea5e86a3460ea9df453b6ad0344b : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb7263131dc644688f7ada70b1fea210 : SUCCESS in 9m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4850d7fb01414f1f898985ac44c08dc9 : SUCCESS in 7m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/441af21756c94b1daa1c56afae2d77f0 : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/116ec8c52d8545b199496f7a8a3232d7 : SUCCESS in 18m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a2f96aed772141939c0118ad431882e4 : SUCCESS in 1h 12m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7905ce7a818243e598a7581c9da08df6 : SUCCESS in 48m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4b08d04375a041abbff9da2bf1894dd6 : SUCCESS in 57m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5a5b495fb154484ba99770a4512660dc : SUCCESS in 16m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d8b9badb5a247dd9f0aa209c06595fa : SUCCESS in 13m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/728afe2bf2c44f2eb2bd22eb9fa57062 : SUCCESS in 1h 27m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4ace575114e143ad872ed6634512c0a4 : SUCCESS in 35m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64da1101fa3246f1ad5ffd6656061de7 : SUCCESS in 20m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f5ce79daf3614bfbb323b58ae03ee7b1 : SUCCESS in 25m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cb50b7bd5b2b4eb7be8e6aed2a977f9e : SUCCESS in 13m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d871571cd09047658643e20753dd29b5 : SUCCESS in 8m 25s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"0ec7c6f0edc03973a09b05cad29ccd717dad1b5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 12:25:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"477d3093dd72a22a49f6288c4348952110d079c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 14:26:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a669c5c8c2f7453d972bcf417eb30c69 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f94ff54579c8451da6f49c1caeae489d : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0f4c0f6830f4ccbacedd07d40542919 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/099ae543a0fe49068019831395c6ec23 : SUCCESS in 9m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a951021f0abf458fa15c71eccd621b8b : SUCCESS in 6m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8553152c38604df98af7d573c71e42ba : SUCCESS in 12m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f5da645d71b46df8c36be775e7bc75d : SUCCESS in 23m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0994130e15a44a75bb38fd4323a6087d : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75ecdb1230934baaa3d5f64633e555c2 : SUCCESS in 52m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7d3213788e334d64a79f4b0164dbfcaf : SUCCESS in 1h 08m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fb94d308363941cfa059c70446c0bf4b : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92b1fb9317da42348b07c1dc03a1a915 : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b5117ce4e72c424c899985907ce59cae : SUCCESS in 1h 55m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e493116da7394b168504e7624f3d65b1 : SUCCESS in 32m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44de83cd13f347059e0d45ef236c4236 : SUCCESS in 23m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/68d35f51ebc646a889fec35afbe79821 : SUCCESS in 26m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d1fd2e86f82a40c19edbf15ddbf701f2 : SUCCESS in 11m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/373f216d03c749f59a854202e0736caf : SUCCESS in 10m 28s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"5eebfe2214758c2a829a004dc168ea7a12a736f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 14:26:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bfe4d37fcb09b50c2a1cb4f549cd1354c198d274","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 14:27:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95371d90465846b98d04c34414a13afd : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c355935c9c94fc9376474f18780dbb4d81e1c09a","revisions":{"c355935c9c94fc9376474f18780dbb4d81e1c09a":{"kind":"REWORK","_number":1,"created":"2021-04-30 12:54:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/44/788944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/788944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/788944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/788944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/788944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/788944/1"}}},"commit":{"parents":[{"commit":"42a13037370e4ffa30fb74e06f6ff8f2c0ad3146","subject":"Merge \"Ubuntu: simplify OS distro selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42a13037370e4ffa30fb74e06f6ff8f2c0ad3146"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-30 12:50:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-30 12:52:04.000000000","tz":60},"subject":"Ubuntu: avoid shadowing of networkd config files","message":"Ubuntu: avoid shadowing of networkd config files\n\nKayobe removes any systemd-networkd configuration generated by netplan,\nsince it can shadow our own configuration. This is performed via a\nhandler, but in practice it may run after the handler that restarts\nsystemd-networkd. This can leave the system using the netplan\nconfiguration, which might not be correct. For example, if the interface\nshould be a bridge port.\n\nThis change fixes the issue by flushing handlers before importing the\nsystemd-networkd role.\n\nChange-Id: I62ef7746aca4c6889526692cd0897e422937e70e\nStory: 2004960\nTask: 42415\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c355935c9c94fc9376474f18780dbb4d81e1c09a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c355935c9c94fc9376474f18780dbb4d81e1c09a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
