)]}'
{"id":"openstack%2Fkayobe~789768","triplet_id":"openstack%2Fkayobe~master~Id4c3841d506d6b6ea082d0b9a294897761105dbd","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Id4c3841d506d6b6ea082d0b9a294897761105dbd","subject":"[release] Set previous_release to wallaby","status":"MERGED","created":"2021-05-05 09:35:52.000000000","updated":"2021-06-09 10:25:47.000000000","submitted":"2021-06-09 10:23:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789768","meta_rev_id":"bae428f6b2094c60d3c7d3387b9ec0339e488527","_number":789768,"virtual_id_number":789768,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 10:23:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-09 08:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-08 20:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-09 08:37:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-06-08 18:36:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 09:43:37.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-05-05 10:48:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-08 20:26:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-09 08:13:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"7e4d87f7783801c6c3868c83d4eea3b6c8206745","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-05 09:35:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0e86ea0b9282a1db3e7513fc31a9d0f4d95d6e1d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-05 09:43:37.000000000","message":"Patch Set 1:\n\nShouldn\u0027t we change this after we revert to use master, which is something we might not do right away if I understood right the Kolla PTG summary?","accounts_in_message":[],"_revision_number":1},{"id":"6e0a79f210fdedc5ae177d3fb20cbfe8769d9e3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 10:48:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fa6c64486d94d8ebbbe6757c9cb7611 : SUCCESS in 7m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20f473bf7cb8427dab04b1dcb70b06d8 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcf63ec0fd1c41f1b3a85210e13231ac : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b10807f37fd4493ea6073677cc3f9c0e : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2bd8e7ecfcf441f7ac2b5cf9a94159a6 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65d5a5b84594d028f1a6ff7f5dfbadf : SUCCESS in 12m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c21ba933a0594e329bf49a794fa88622 : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d067518c839045948f5003970f95bf0c : SUCCESS in 46m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cae740269da543d5946d9a66327cf51f : SUCCESS in 1h 06m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b607693a55d34d29aeee52be9ce58a37 : SUCCESS in 17m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3ad1ad715f84b5a9ee508b3b6c8db85 : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d662fd380346451c90f03086d6da7757 : FAILURE in 1h 05m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7c0fdb26fb0c45df969711a3cf34013f : SUCCESS in 30m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44c920ce763f4e018ca7f42097e2f95d : SUCCESS in 31m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3300e5c8bff746d5bd0ca974402bfa25 : SUCCESS in 23m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/aec5cc4274b846399ddd4c198166c435 : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b860ff8b3174df3859159af7758111b : SUCCESS in 11m 27s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"2dd7bb28515446e8fda4c6d5ad5f7fd0debad1db","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-18 08:50:01.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Shouldn\u0027t we change this after we revert to use master, which is something we might not do right away if I understood right the Kolla PTG summary?\n\nWe\u0027ll discuss this on Wednesday. Proposed new process here: https://etherpad.opendev.org/p/kolla-release-process-draft","accounts_in_message":[],"_revision_number":1},{"id":"cd58c6e123203fea01faa8dfe052aa4accb1e3fe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 18:36:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"21a9041105a4cdbbec020d2ba7995e860660202e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 18:36:51.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"0970398d396acdd6be939e86119e9a5d8ddf13ee","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-08 20:26:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"843fec3c036881163cf00d43eaf114699103725b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 20:39:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ceea67da7c449dbbfac5b78a4217603 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc637a27871d4463b27b2502b706d6c5 : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d588362b031447e5af69c7b64fd6feb3 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8779974bb1c44e0483d99053e8a7d0ce : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63e111bb7a5549509c60a01aa1eec593 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc68ca917ff464b878d710fc8fcc214 : SUCCESS in 7m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9847c221ad174ef8b6d8288599656470 : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba078d63fc184ffeb4d079b4a8c72d7c : SUCCESS in 45m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cceba1d9e4f84c3883a1ba23ca301a45 : SUCCESS in 1h 00m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/490e30446ee540a990e8481fff04a992 : SUCCESS in 20m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9f016aab4714a9bacc667fc0ad21bce : SUCCESS in 13m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3bda57d9716b44e7a13f8e30a129e476 : SUCCESS in 1h 57m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3b6e382098b640e1831ae70446b75275 : SUCCESS in 29m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8a3eeb3895d48489aa845eb49905cab : SUCCESS in 24m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0954092baa59459aab5e8f36544bbe4e : SUCCESS in 28m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/23c4ff3445c84ee8a72e005d40ddbaee : SUCCESS in 13m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b0f464b2e5b46179b057af4372c02c4 : SUCCESS in 11m 09s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e00071834c230e0fc27b022d60e771f04b6b7047","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-09 08:37:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acb91cbb742d7847c371831d3367d79f630ac15e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 08:37:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6cd6983ff94e02942cbc5571726fe8165b84823","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:23:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c8711bc95ed439483b186efb487bd8f : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab70ed3ba6514f0596dcabc1123aff26 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d3ef706564a4dcb8ddb39277bf51432 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aa901c0b59b405e922e0c0ae3e8914c : SUCCESS in 9m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/75ce69c12a4240b38dc81973cd2f6235 : SUCCESS in 1h 10m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2acb83c5f5d042af9cbf85b1141d3682 : SUCCESS in 44m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/eb3fd709237240c981be4c566a0a5371 : SUCCESS in 1h 08m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3de64b490b1040c1bec53e2ea1b07c0e : SUCCESS in 19m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15b35f1ae120457bb893724dbe41ce58 : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/02f051654ce94d7dbdfad2ea6f41c6d7 : SUCCESS in 1h 43m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d07a4e2d370349d48aeeb6d618ed2bf0 : SUCCESS in 33m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d2aa80e1d6b43bab3fb4430fef6e9ba : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/98f599f16ebc4035b0f36231b687d476 : SUCCESS in 24m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/af18b68856b248f598643dd58eacdc27 : SUCCESS in 12m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eda31305a1d349959d9be8d5635b6ab1 : SUCCESS in 11m 19s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c9a3506feb3df21be8cfb46c0c62f9facda8cf81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:23:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bae428f6b2094c60d3c7d3387b9ec0339e488527","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 10:25:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2eb743b1e0f84f50905029a23ae3e5b4 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e825118c526a88ca3478f70c5a4db0d25112ca8","revisions":{"8e825118c526a88ca3478f70c5a4db0d25112ca8":{"kind":"REWORK","_number":1,"created":"2021-05-05 09:35:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/68/789768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/68/789768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/789768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/789768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/789768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/68/789768/1"}}},"commit":{"parents":[{"commit":"8315542318068c94ba9e13dfe3ba3092819921ce","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8315542318068c94ba9e13dfe3ba3092819921ce"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-05 09:35:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-05 09:35:21.000000000","tz":60},"subject":"[release] Set previous_release to wallaby","message":"[release] Set previous_release to wallaby\n\nThis controls which release we are upgrading from.\n\nChange-Id: Id4c3841d506d6b6ea082d0b9a294897761105dbd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e825118c526a88ca3478f70c5a4db0d25112ca8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e825118c526a88ca3478f70c5a4db0d25112ca8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
