)]}'
{"id":"openstack%2Fkayobe~790072","triplet_id":"openstack%2Fkayobe~master~Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443","project":"openstack/kayobe","branch":"master","topic":"centos-repos","hashtags":[],"change_id":"Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443","subject":"CentOS Stream: update DNF repo file names","status":"MERGED","created":"2021-05-06 14:08:09.000000000","updated":"2021-05-13 10:51:01.000000000","submitted":"2021-05-13 10:49:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790072-centos-repos","meta_rev_id":"a7f54d322ff6fe010aeb5ae684d262120ba0bdd6","_number":790072,"virtual_id_number":790072,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-13 10:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-05-13 09:10:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-05-12 07:20:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-11 20:51:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2021-05-13 09:10:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-05-13 10:51:01.000000000","post_submit":true,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 14:38:08.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2021-05-06 15:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-07 07:40:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-05-13 09:10:22.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"58f349af2737b0c8356002031815e8cacad5a7fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-06 14:08:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfa968d8251b62509f2b676afd3c59622306a282","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-06 14:09:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9bba04b131a953a608b707d0fde52e34df38b670","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-06 14:38:08.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a2992e60438dfc183309a6dfac1c3b761fd5c7d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:50:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5e19f670a2847b3937da8c9e2aa1729 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0470c097b6d34caea998cdf4ee07fe1b : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e4e65f806354feba229c2f1e4261f39 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e7a985e86bf4d9b8e86ce78bd53009a : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a60f736673414f7e97aa776216d7a966 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc57adc22a1044b6897b8b71993c960f : SUCCESS in 9m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d44e50822ae423abbe6345dbba4b4a7 : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ffa736c71b89467e8627ada48d2b39b9 : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/62de5695eeef4eff80016bb2bb1a442f : SUCCESS in 18m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8c7f0baa1f7346edb5568d1f4d7e4810 : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faec1cda35b6475b9f22109bbf7d1d59 : SUCCESS in 42m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/43879f0f3b134662a2d51334729a86b4 : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4cf63052a73a4cb6aa357c7888a3271d : SUCCESS in 19m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e50518d9b9e44d4e9869ba68c3f4b885 : SUCCESS in 11m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bd9aa20e59e34e9b8a7e7e934ebe4573 : SUCCESS in 1h 38m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e5a36c5251ca4d0c9713dcd70d9c71db : SUCCESS in 31m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/deb643a5ca4f4b8ca00431908ef526b2 : SUCCESS in 24m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4311fa798a154736ae40577cad7e81af : SUCCESS in 23m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/649746a57dd34a53a1b3515f80b44455 : SUCCESS in 13m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56566532051e401e85be508799f2f498 : SUCCESS in 10m 52s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"024b323cc9ded5cbf0ac95707bd9cba264fc3259","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-07 07:40:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"52444ec297197c494beee7f1dae3f3222f4baf07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 20:50:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0942def429a66d385095f02ff4bd927bdf853b17","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 20:51:10.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"35ec4c9378da7fd2ae61cce452df562f7734892c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 22:39:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cdd402a9f5846ef89102527cda96db5 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8b7b87956bd42469b7b39ce7935d237 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/255b6ce2b610410b9e5c739061ba6123 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cf43f41e276454b95306db73e0ffd80 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97b9149769af4c439cee213134b6ac78 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12cdd0d4755c41f49a9665141e6848b4 : SUCCESS in 9m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4be37fcce4934d0e83166ad7dc71d585 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e25ff98558040ad939ca2f6cf4fd14f : SUCCESS in 10m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5b7ea17b14da4f04a89f782f2a0cd88e : SUCCESS in 18m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9d0fabcf1195432082111f2d7aa4b8be : SUCCESS in 1h 02m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dd96d1ebd7b423e88bad736e649f9f6 : SUCCESS in 48m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7d7b687dfa604c728c4ce068f961cf57 : SUCCESS in 1h 04m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/479939a3c37441efa3e4d9e06ddfc2ed : SUCCESS in 17m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e97fc5897c0a4ea2b57811c34c97a324 : SUCCESS in 13m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c9f25e55d36f48f6b260e1a05c7e37b5 : SUCCESS in 1h 47m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1f1f3af2ad6941d1a1c7970fabc9e74a : SUCCESS in 41m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0273c6f9f057403d8535b1fd27bf089b : SUCCESS in 41m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/24ff5ecd009b4dab87a63f72d0941e21 : SUCCESS in 26m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d13148b1cd524966aafeec3a2073718d : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f63ee6a79d4b49c0abebdcfcd046ad76 : SUCCESS in 10m 37s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"7359fae7f7cb760fc77d528deca6c27d3673d046","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-12 07:20:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"3c4f94c247742ffb602c6a04354b7e03d06daab2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-13 09:10:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6fe7d1865b37430d8d90b5a66cd125f030b4f501","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 09:10:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6745ecc669751bf74f06dfae50f31796d008f1e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 10:49:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e689636197644f1af2d4719cb16e9ab : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84d86d92228d4254903df9dc25cf8726 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/278d353511524e64921d514cc5db2d82 : SUCCESS in 13m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/677292d7b9184a10a785b84610043ffe : SUCCESS in 23m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41ba36e084444e00be32e4f26d8e8388 : SUCCESS in 14m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73c0c2a35d734747b51160e69e0fddd4 : SUCCESS in 19m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9931e3a5a3804ed6b38629b665a73a77 : SUCCESS in 18m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8fcdd868d32f40839ac0b4d3b8d61da2 : SUCCESS in 55m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/202e43d69e9e4d93b18b84755cd7bb9b : SUCCESS in 36m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1217bbfc63f44c91b793b70eeea82477 : SUCCESS in 56m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e02ecb8a7cb248b1a363b2ab7b2e0d1e : SUCCESS in 18m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/373632094dff4df7970e98e59f3683a8 : SUCCESS in 22m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ae4169459b6549758c656e1b31c778e2 : SUCCESS in 1h 37m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/166dd3b5ad36420e8d13b2e5a2e25f53 : SUCCESS in 27m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68f084701507464c9ed2f4be0dcc859e : SUCCESS in 26m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ca9e453087c0426eba3a8d27fe831fda : SUCCESS in 25m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1e7536fed8984ae0b2f67d6d49218be2 : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/910537518ef9411e82170aab4246275e : SUCCESS in 11m 07s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"6eb5bb9c7a7172880d5b1bfcfa0c2252bc562504","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 10:49:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"4d6b64e2b1253a9b52a17f4e6f291e06c5aaac41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 10:51:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c258335cebd1413697981f2c99d2c204 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ece41b7a677341d5023eb3f5e4db7f5326cfc86c","revisions":{"9b0dbbc78da0d36a76d19ae8ee20a8b2b9e46e6f":{"kind":"REWORK","_number":1,"created":"2021-05-06 14:08:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/790072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/790072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/790072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/790072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/790072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/790072/1"}}},"commit":{"parents":[{"commit":"cbf9082294f4468e64f0c36ea87c5e6e50de31d8","subject":"Merge \"Add os_release variable, build CentOS stream images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbf9082294f4468e64f0c36ea87c5e6e50de31d8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-06 14:02:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-06 14:02:21.000000000","tz":60},"subject":"CentOS Stream: update DNF repo file names","message":"CentOS Stream: update DNF repo file names\n\nCentOS repositories were renamed following the separation of CentOS\nLinux and Stream. This change updates their names when using a local\nmirror. It also removes the old files.\n\nChange-Id: Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443\nStory: 2008795\nTask: 42435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b0dbbc78da0d36a76d19ae8ee20a8b2b9e46e6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b0dbbc78da0d36a76d19ae8ee20a8b2b9e46e6f"}]},"branch":"refs/heads/master"},"b8042336ec88512276efb7970e1416de191a00a6":{"kind":"REWORK","_number":2,"created":"2021-05-06 14:09:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/790072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/790072/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/790072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/790072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/790072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/790072/2"}}},"commit":{"parents":[{"commit":"cbf9082294f4468e64f0c36ea87c5e6e50de31d8","subject":"Merge \"Add os_release variable, build CentOS stream images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbf9082294f4468e64f0c36ea87c5e6e50de31d8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-06 14:02:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-06 14:09:02.000000000","tz":60},"subject":"CentOS Stream: update DNF repo file names","message":"CentOS Stream: update DNF repo file names\n\nCentOS repositories were renamed following the separation of CentOS\nLinux and Stream. This change updates their names when using a local\nmirror. It also removes the old files.\n\nChange-Id: Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443\nStory: 2008795\nTask: 42435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8042336ec88512276efb7970e1416de191a00a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8042336ec88512276efb7970e1416de191a00a6"}]},"branch":"refs/heads/master"},"ece41b7a677341d5023eb3f5e4db7f5326cfc86c":{"kind":"REWORK","_number":3,"created":"2021-05-11 20:50:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/72/790072/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/790072/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/790072/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/790072/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/790072/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/790072/3"}}},"commit":{"parents":[{"commit":"4daf293e7824a9444906fd3510c2def0a736801a","subject":"Merge \"CI: Remove unused Docker images after overcloud upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daf293e7824a9444906fd3510c2def0a736801a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-06 14:02:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 20:50:31.000000000","tz":120},"subject":"CentOS Stream: update DNF repo file names","message":"CentOS Stream: update DNF repo file names\n\nCentOS repositories were renamed following the separation of CentOS\nLinux and Stream. This change updates their names when using a local\nmirror. It also removes the old files.\n\nChange-Id: Ifcc3781d771e150b5771f2bf20cd8bcbc5faa443\nStory: 2008795\nTask: 42435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ece41b7a677341d5023eb3f5e4db7f5326cfc86c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ece41b7a677341d5023eb3f5e4db7f5326cfc86c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
