)]}'
{"id":"openstack%2Fkayobe~790247","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Iaf0dfb34750ba8b8d74a043587a2035b511de6ad","project":"openstack/kayobe","branch":"stable/ussuri","topic":"dnf-centos-repos","hashtags":[],"change_id":"Iaf0dfb34750ba8b8d74a043587a2035b511de6ad","subject":"Sync the names and content of the repo-files for CentOS8","status":"MERGED","created":"2021-05-07 14:01:24.000000000","updated":"2021-05-11 11:32:40.000000000","submitted":"2021-05-11 11:31:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790247-dnf-centos-repos","meta_rev_id":"331e0b4cd36b4aaa8e5c5d88ca8b138ceafddaa1","_number":790247,"virtual_id_number":790247,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-11 11:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"all":[{"value":1,"date":"2021-05-08 19:07:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2021-05-07 14:05:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":2,"date":"2021-05-10 09:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-07 15:32:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2021-05-07 14:06:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},{"value":1,"date":"2021-05-07 18:59:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":1,"date":"2021-05-10 09:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-07 14:03:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-05-07 14:03:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-05-07 14:03:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2021-05-07 14:03:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-05-07 14:03:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2021-05-07 14:06:29.000000000","updated_by":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"reviewer":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"state":"REVIEWER"},{"updated":"2021-05-07 15:53:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7c968a8e8dc1c36cb3c22f3da70537cd606a2463","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-07 14:01:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61de845d8aca2c701508e54b369dcf64b9cfe191","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-05-07 14:05:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5f6cab1dd56ec6abf2dd075e4014b0e24d6bc537","author":{"_account_id":18511,"name":"Brian Wickersham","email":"bkw86@bellsouth.net","username":"bw6938"},"date":"2021-05-07 14:06:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9d3179bab71e793d7cbfb5ac6ce31e8149fbcfdf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-07 15:32:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f2362be7751ce13e829a0c9c2cd2e7b288b3d5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 15:53:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49d24f6a347a414da4951f68d4efcb13 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1078fd2c43405abe2d28021a4c043d : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86ccdab83dd44da5a657f0ea0c69f6b0 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b694a5332af442e28f738a2c54d48ce1 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66bde24b0d254861a3acb8a69ddcfeeb : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f54acea4cd254be3aa29cd3683f3f0ea : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a34c783dda8f49bb99b0b7485994fcf4 : SUCCESS in 8m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f92a9031dd3448d9d3b8c923e67ac6d : SUCCESS in 18m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/00837e8a8b494051886d1abd57d40893 : FAILURE in 53m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/72df868c10554db996b2d00b96fea46d : SUCCESS in 14m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ce64b43248fd4f45b8d1503f0d1fda13 : SUCCESS in 1h 49m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/15d1bf7b783b451481a723b9565baa73 : SUCCESS in 25m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2956f67e880a4402b5cd5e61d4606fde : SUCCESS in 25m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/21f3371e21394d3e8fcb317cedf9ba41 : SUCCESS in 10m 17s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"6b2759a3529a4466e9a84e90ad156cb8af2421f1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-07 16:03:16.000000000","message":"Patch Set 1:\n\nrecheck\n\nCan\u0027t connect to MySQL server on \u0027192.168.33.2\u0027","accounts_in_message":[],"_revision_number":1},{"id":"9a1348e3c64c1defed2a145bb172e1fc27e081bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 17:25:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f93a568d22e41aa9e52ee0b471005cb : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ee8cac53374541ad0feeb8fd915f99 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f01b6d3deb5b4b53ac10945f30791a12 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/67a86a7241a742568bbc7a44cb077b21 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/301c24532a6147dcb6b70d8ccbdbadbc : SUCCESS in 7m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4ce7d67b2a849b88a3f4eb331dc150a : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8b088a6928f46df8c276c830d7d5fa6 : SUCCESS in 8m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b85a8d044054dc38fd9dd6f2de18e06 : SUCCESS in 16m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b7bd6d6beebc46259b98242e63fabfdb : SUCCESS in 52m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/efbe79ee0e5545bba2687c0553ebe743 : SUCCESS in 14m 55s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/997213c5a6b843aaa23522a135efb114 : SUCCESS in 1h 21m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/954be0d2031f472bb41c124194fa6248 : SUCCESS in 23m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4126ab8da2ee4362b1d7231a2cd96d25 : SUCCESS in 19m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0c480f3d45db4727ba0518faf9f8f96c : SUCCESS in 10m 14s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"e0d837108af06254169777835e88b519c92e78d7","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-05-07 18:59:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"84774d4d96c658e5de71648f29b2fb6ca7680d71","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-05-08 19:07:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fb37297dacf3ace6045cc8ca0d88e62a653b9698","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-10 09:26:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f360fb4bbaedefa0a847cdbcc8d6ba869d9b4bab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 09:26:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d230892ce9a9cc72e731b27d4d90a525fda5cf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 11:23:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/836cc157b6db4b1598e2529860094234 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf244f42f64747a7a4aa15365817160f : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1dc825a807f64026a3698d92460386e7 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e5b0672434145d08c490637488422da : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f34378d1fe9e438f844565919e096f9e : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/933e1a5058da448e8976f1a65551b813 : SUCCESS in 9m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8134bc47a9c648d6ad61c5a4d993f253 : SUCCESS in 17m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d539b9d02c994cd4b3f08b2a17534900 : SUCCESS in 1h 10m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bd782d586fac4c90a05b2094c4dbc5d8 : SUCCESS in 14m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/72ad3f6c088e4cf784feaddb9e42af96 : FAILURE in 1h 52m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f38deac736a0410f8f89466e8f9193b3 : SUCCESS in 26m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e8c2b85d3ff9422e9fac446173b902f2 : SUCCESS in 26m 17s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7ffc9d11680144d0836d184f568b2495 : SUCCESS in 10m 27s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"792ab89e28322fa7d2d627fedec50f86bf67dc3b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-10 12:24:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"41f00fd752b315294215db57da6da25547fd99fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 14:09:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5598505e32b3415a939e190642485364 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/988ead293bf84e099efd46058f2ea9ab : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76ec112e0485436db857469cdcb53846 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a71dbb95b8e4153b1e43460ce0b6a92 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2eead8863dd4c00b0486ef4cda628a5 : SUCCESS in 5m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4097314f680d444bbd02052a184f5022 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/191c0058356a40959c6ca1f6b3d2c17a : SUCCESS in 9m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/785d6d7af94f4f0f8dd240fc0f7b363f : SUCCESS in 16m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/95a0f582c2974c1c94b8a716187ff4c9 : SUCCESS in 52m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fbc80fb8c6ca4bc7a3e6abf4f91b6a73 : SUCCESS in 14m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/64f65c40a4a44fc89629d77134ffdc6f : FAILURE in 1h 44m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d30285244f9943fe936b10c8bc3693a2 : SUCCESS in 29m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/90bdad667b694bd6b62a35d87aae78e1 : SUCCESS in 26m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a36bc186c2144805bc3ad21069b27326 : SUCCESS in 10m 01s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"25a5f7cd04e95a51cc5a2edfc298ef3fd9b0c629","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-10 14:24:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a1c82889fd975f7be4a82c25ee3fc7c52f4e46fd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-10 15:09:06.000000000","message":"Patch Set 1:\n\nNo need to recheck, ussuri and older are broken due to sphinx release","accounts_in_message":[],"_revision_number":1},{"id":"39b6c495c2409ca39fd84d9a814b18b44ec4e448","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-10 15:09:39.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e No need to recheck, ussuri and older are broken due to sphinx release\n\nSorry, posted on the wrong patch","accounts_in_message":[],"_revision_number":1},{"id":"6a2c054fcf43719f5d23d0013ad0e4ba28153072","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 16:02:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ff6dcb07c96467ab0d417b15d33dc89 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/080527a014e54a5f96ef23c212584bad : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cc0f0f8fa8c470eb700479ef149a51d : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/23230c1e683342d4bb31257d871bb8b0 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e99b9a016bb4a35bef651a199ddd02d : SUCCESS in 8m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b7080c30da847acad425ae02c7f3fcd : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eba465a7b2c743f1805852efc027982f : SUCCESS in 8m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dcdd45ed1d38492eba2817e267f2db20 : SUCCESS in 17m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bb95d043293d4005934fd507d0b67018 : SUCCESS in 59m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/637fd8f513e9484991240b28598b1cf4 : SUCCESS in 14m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3adb7afe7f0e4aeb8d865ef6e7d0eb34 : SUCCESS in 1h 36m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/401cf39a5223408c99331d7e52ba127a : SUCCESS in 25m 45s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cc2ab89e05fd4c3a8d61e474e8a65863 : SUCCESS in 24m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/534d46d32f2b4cb69428b0fe5103bd3a : SUCCESS in 9m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"ed42000eb2f06c9fb83eab781db3fed362dc3ab6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 16:03:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8f2249b80842cb4548e6b282a03b57eaabd27af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 19:49:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bb6e77b14a141958b57513bb8c51cbb : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc313ba65f2645caabe1885b0530edef : SUCCESS in 6m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dfd59b2966414d8ca7de337ccc58d6ab : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b22a4ea19914c57ba1ee6ee21ca9d89 : SUCCESS in 8m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fd1b814021a34b35968ac68be621c4b1 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f9249b1ade1f4cdc95fe446df1b7c849 : SUCCESS in 8m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b5707d14acc4335be5bf0c610573fa4 : SUCCESS in 19m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8e83b0c0bde847aea93e83514bdc1514 : SUCCESS in 58m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/584abc23d32d4cfb917826b2d391cac7 : SUCCESS in 14m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c7b0ea8e7a1f4edeaa5afdf8e4f0741f : TIMED_OUT in 2h 31m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d9e73a5199af44a79dcfc7989c8982d4 : SUCCESS in 27m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/55aa4c7ff04d431b8df4b75ea09208f6 : SUCCESS in 1h 04m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2d9f7e38316243d99a4bdc887b529e57 : SUCCESS in 10m 34s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"09d7385e1b66de2c88b681c67ad5347b52602cb1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-10 20:02:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4160ee90f7d92ecce50f8968c113b567124516eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 21:59:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3af85ef56ef645e495228bca9b105e89 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa6e9b0697f241eea1ae0b07a40d3740 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ed46008536b4a4da6c6d018995654c8 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4d9c6a9c8d0e471a99db42c6667f8e43 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dcd83264d38482184e6b0b7f5f2b2f3 : SUCCESS in 6m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0b2fbf0e9124462d8f1e4254381d46d1 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ab90d35134a04ee89e694849903dac10 : SUCCESS in 7m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/90011dfa2eb64dcf97995a10bc868707 : SUCCESS in 16m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f70d6e58a154440681cef5b9a27d4b88 : SUCCESS in 57m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e90de2e1dcd24a7da86bc6431ffbec2d : SUCCESS in 14m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0cef3a3e39a2439b9a36e9ca8b446cf8 : SUCCESS in 1h 52m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1befeda7b0a04075b99bc52ae4780fc5 : FAILURE in 25m 55s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/51f7509adb4940a296038387677e7c22 : SUCCESS in 1h 19m 05s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c766c9b6b2d14046b0dec17c565c1c66 : SUCCESS in 9m 23s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"6a993fb7290fead8e084b73d82442228d6d5e62a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-10 22:00:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"601bd647ce45decb409ce892944254b44feea61a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 00:03:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e642ddc45dd49389c7622b128eb6bab : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0209a42593c947dcb8cf3743aa467be9 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/093566e1d9c74ab880bcb78dfc7be1b7 : SUCCESS in 4m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0273fb8321714caa9d833bfa66da4485 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bbbc4e80c3f4ee3b07f7c7f89d92374 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb338fef5488451ab6588e42dc86ba64 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f6f36ceada04382b9b0f52bd06c2f2b : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f736779197704d88a5212c2affda2fce : SUCCESS in 18m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d8b6c5ebee1b46e9a802bb60d236c543 : SUCCESS in 1h 07m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6a0fc743128d4d13a502fc22f3c08fa6 : SUCCESS in 13m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c24c0a37abf14121b0e14f65baa76a4a : FAILURE in 1h 57m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/48e3c8b2ab53436fbf254d76fcc360be : SUCCESS in 27m 19s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/73ebee819c6449b1bf3f2af3915858a0 : SUCCESS in 26m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/93f823c6ea594108a4e43ec248e24bc7 : SUCCESS in 9m 17s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"610faf31d1ed3e13e6c7561bee2d2f1387dbc2cc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-11 06:21:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e41b06d104c5414e585142f003d1687e93a10f95","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 07:49:41.000000000","message":"Patch Set 1:\n\nThe CI failures in the overcloud-upgrade job are due to the disk being full towards the end.","accounts_in_message":[],"_revision_number":1},{"id":"da77cad1d9ca9da06cbe83f8d4966f69304ed814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 08:01:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/125b2966853a409dab79c1f0d07f0cd2 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac3e54a2d07b4a9a98ea3f90de893fad : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a0bc16e4ec440218df9890e09c59622 : SUCCESS in 9m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb7572af0e31409e913eab184aeb4b19 : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bf8b72cbdf74765a4e062a02b2bab43 : SUCCESS in 12m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cffe716a0eee42689564263261f3fe65 : SUCCESS in 7m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7ddc4922c7d41159f1c3f2b2396cfd2 : SUCCESS in 13m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ccd0df1422414582be8c309182bde4cd : SUCCESS in 22m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cfa40d9756b4481fa9e8a4a14f876d23 : SUCCESS in 1h 04m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0f0cfed2de384545982c332900ddc1b8 : SUCCESS in 17m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/63abc4f98df441a39f767dee5dd9ef3f : FAILURE in 1h 35m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/23b6a60c04c94cbfbf055d445ae4b8e6 : SUCCESS in 28m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/64625994b19a4774bec2ca9e7ef5523d : SUCCESS in 27m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ca4a7a2450f84d84a38842057ebed5dc : SUCCESS in 12m 29s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"5bc6b1fc4071fb1248ee53da99ade69d229f3fd8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-11 08:04:13.000000000","message":"Patch Set 1:\n\n\u003e The CI failures in the overcloud-upgrade job are due to the disk being full towards the end.\nI see, but I can\u0027t free up space on CI) only can recheck and pray for success)","accounts_in_message":[],"_revision_number":1},{"id":"cba42b05f25623d6d49e2582e029eb5964ac2690","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-11 08:04:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f8fee669e1f735c58418f18e6ec6c5075236435","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 09:19:15.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e The CI failures in the overcloud-upgrade job are due to the disk being full towards the end.\n\u003e I see, but I can\u0027t free up space on CI) only can recheck and pray for success)\n\nI have submitted this change to help: https://review.opendev.org/c/openstack/kayobe/+/790587/\nIf it works, it will need to be backported to stable branches","accounts_in_message":[],"_revision_number":1},{"id":"a546180256d90c443cc259dd46ed997558743099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:58:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6234881803d349db8f3964ed6c6f4d0d : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c14cef25cec44ea7b637d0b740d94d7c : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85b434d24bdc45a38981deff3da5ff6f : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/03f06ac489fd4810b4d00054241f32e0 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0705a1b05841478ab2e5c749d24d3c : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bfec6688d5214d4e8e6f99308fdb8b7c : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0911ad58a6748aca80d6035a781601b : SUCCESS in 8m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e1f15d24045441f9636bffa19c5c37e : SUCCESS in 18m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/57698464d38b465fb5d6473d0c548c51 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/20af87945f984e5680fe8c2c3af20c4c : SUCCESS in 14m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3a164bb21e38490c99b3defad7d9cbed : SUCCESS in 1h 50m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dc57e246a7b8492ba68ce8b9b9e84b01 : SUCCESS in 29m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/645c1abbc0a0475fab56b43e500872c7 : SUCCESS in 24m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b634606d8a7e48d5a8249d767b97ea81 : SUCCESS in 9m 49s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"6c9c847fb69b02c32ad5143b61976ad48e795edf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:59:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9d5b530a3eb2ccd8ebd7ff63e9131f28a159323","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 11:31:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e9f1ba3ce9f4b2cba64425371fa7988 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a035ee5b3074044889c02ba8257c35e : SUCCESS in 6m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/479fccec5c5842beb57d27186c6e973a : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1725548c2cc543f096606a996fad4326 : SUCCESS in 7m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c33f5cc2f084698888ceef3da2dd499 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/38f0c6f4efac41c8861526a7269007d7 : SUCCESS in 8m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e9ff50bfaed042fe9cdda4b84923919c : SUCCESS in 17m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/42aa5eae9ea84782a07afb06a4b32159 : SUCCESS in 52m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b73ba10743f24686a08b4e2d99b27f3c : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e0be93483263472e9912d7df4cb74bee : SUCCESS in 1h 28m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/44dd13c2285143bca2a9d2bdd5348e87 : SUCCESS in 26m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3d350041815f423cb621f03dc8abf040 : SUCCESS in 25m 57s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1a850801dd0545db88a78a980f3d1d22 : SUCCESS in 10m 36s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"096c1c6ccd6b2b35649b95c07b43186a3d6a1aec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 11:31:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"331e0b4cd36b4aaa8e5c5d88ca8b138ceafddaa1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 11:32:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9a63e67c4af48578d2685ab4bd3cb0e : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fdecebeff8cba152f5aa82008bd58d02875004d","revisions":{"9fdecebeff8cba152f5aa82008bd58d02875004d":{"kind":"REWORK","_number":1,"created":"2021-05-07 14:01:24.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/47/790247/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/790247/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/790247/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/790247/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/790247/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/790247/1"}}},"commit":{"parents":[{"commit":"4eafeab93a4a0e33b2a538aca83e22632a6e4f66","subject":"Switch to reboot module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4eafeab93a4a0e33b2a538aca83e22632a6e4f66"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-05 21:46:43.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-07 14:01:06.000000000","tz":180},"subject":"Sync the names and content of the repo-files for CentOS8","message":"Sync the names and content of the repo-files for CentOS8\n\nSince 8.3.2011 release of CentOS the repo-files renamed and the content\nis slightly changed. This change sync the templates used in role with\nupstream changes.\n\nChange-Id: Iaf0dfb34750ba8b8d74a043587a2035b511de6ad\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit cdcb90809a2429c6e57d80a2f93fc9b6b218590c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fdecebeff8cba152f5aa82008bd58d02875004d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fdecebeff8cba152f5aa82008bd58d02875004d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
