)]}'
{"id":"openstack%2Fkayobe~790400","triplet_id":"openstack%2Fkayobe~master~I14289df5e248b5b47fb8e60c6eea1ac845f50fd3","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I14289df5e248b5b47fb8e60c6eea1ac845f50fd3","subject":"Fix overcloud post configure Python interpreter issues","status":"MERGED","created":"2021-05-10 10:43:04.000000000","updated":"2021-06-09 11:23:27.000000000","submitted":"2021-06-09 11:23:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790400","meta_rev_id":"2ee2f84b4bb786394dda86b6010ad51b414ec3f5","_number":790400,"virtual_id_number":790400,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 11:23:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-09 09:24:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-17 10:04:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-09 09:24:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-10 12:30:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 20:46:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-09 09:24:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"63a9251cf6583b24d6b540230c3c715a0af901f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-10 10:43:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"497ff9e3ea33f16ad67c3903ca97ca16e175cac8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 12:30:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a36a0bae7af24e94a07b588661c6f879 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5c2b53581d44a31a752b79e606efca8 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6019f940f22c4eeaa17d6ffb44e91481 : SUCCESS in 7m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/467cd444893c4d8cbc456d3d5ed58295 : SUCCESS in 8m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/867df0f72fc04c04ba655ce410d8a175 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e84a8424b6e64126809ee90e2c0af4ed : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5ad928bfd6a54725ad4ed63e2ed13c0b : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6f584379c4a4681baeece1012f9df7b : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee5bdb60680349e5acb4c680a80b7727 : SUCCESS in 19m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a98cf21538754e29ac483d3b6ba135c1 : SUCCESS in 1h 01m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/771599152d774648bb94493cbb6318f5 : SUCCESS in 43m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/07c1136008eb4521a0c42e2c9a35d7c6 : SUCCESS in 1h 03m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cdbdb526071a4da6853c6718df979566 : SUCCESS in 15m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85c91e6b55864c9d958ba228fe22751f : SUCCESS in 14m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6655524b688849f197cbf40208a18cff : SUCCESS in 1h 45m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c24323190ecf47dcae1fe56e9dd915b8 : FAILURE in 17m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c625815c7d0146f3aa47dd108020b2b6 : FAILURE in 15m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0ccc6004b02947d386e6905720df248c : SUCCESS in 24m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8a2b95fe1cab4398a3144dfc15ea7e9e : SUCCESS in 12m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70af4cf05a044fe99f284b2eecb63cb9 : SUCCESS in 10m 12s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"5c870675bb8d5e056458694e5b0a290b28144e7c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-10 12:45:59.000000000","message":"Patch Set 1:\n\nrecheck\n\ntypo in os-images","accounts_in_message":[],"_revision_number":1},{"id":"46d3c0429766532aee2c880379a1f06ab42a5946","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-10 12:46:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"b38fc510f43c552b597c1b3f937af3536d63723f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 14:11:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66b52d420b8b4c5fb12f2262e4009b35 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4640f7b6fdb4c5ab8ea799bd5897291 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8ce75828d4b452c975a903fd305fef0 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a188665e6c7425094e71220913c2bc7 : SUCCESS in 11m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5819d5bc430c4d829839c0484d90329f : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17c67fe1f961477ea4662e4ddc67257b : SUCCESS in 10m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc3adaabef974977b42f4f58c6cfdee9 : SUCCESS in 9m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8becaf574f41445b8e2c6fc97d1b1d28 : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d617b0473f04553a4119f7ed0db6134 : SUCCESS in 18m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b292694dfbe04278aee75ec2813a00b2 : SUCCESS in 1h 02m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/039f3b4753664581a9e2165c806bf2c9 : SUCCESS in 40m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b5693d9c756043a7aafab72ef31bfd93 : SUCCESS in 1h 06m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b7efb8176b1449308a9c50f7ee447f35 : SUCCESS in 21m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2deb4933a7d4e3489c0cb57840f7ec9 : SUCCESS in 15m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2adcea4bf6fb45c494ac3e64bd3b2bc6 : SUCCESS in 1h 23m 38s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8eba33ad17b947fb967e0d1d1c341481 : SUCCESS in 30m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cba5203d8814a67ae068f8c6354e03e : SUCCESS in 18m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/932e33e4124c49bd940413c53ea693aa : SUCCESS in 24m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/98da8fb2c4dd45cd9028f78f38130f14 : SUCCESS in 10m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b334e3154e0b4a13b58a2ec180f2db10 : SUCCESS in 10m 03s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"50012d71b979071d0fac020e00ccb6dae53a587c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-10 14:29:59.000000000","message":"Patch Set 1:\n\nPRs:\n* https://github.com/stackhpc/ansible-role-os-networks/pull/28\n* https://github.com/stackhpc/ansible-role-os-images/pull/42","accounts_in_message":[],"_revision_number":1},{"id":"e113ffefaa5ac67a99b22ae23f6ac96f05c1a8f6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 20:46:20.000000000","message":"Patch Set 1: Code-Review-1\n\nNeeds an update with released roles.","accounts_in_message":[],"_revision_number":1},{"id":"fe471b4426f0dc430f42ffb0735210eda0f40c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-12 08:11:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e6553c3abc4fada1c0c28994a3c5adf97772dfaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 10:02:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ed9e77c72aa4a8f8e80ffb6bcd5f542 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ed417eea820418ba29091eb14cf7a11 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7fb1e7a91723413dae381aac61d8fb3d : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1975994d6f334796a0553dd43c4566e9 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b2430c59b9c4bb688e41018a12481b3 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f13abd8e8b4cb2b36d9fa2f8acde78 : SUCCESS in 8m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/21f843ffe2a84d48afe0374fa66aa5a9 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/429f1b3bc252491a9ed3dc01c213efa7 : SUCCESS in 8m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/538fe12c80f4432788a8a79f570fe4f7 : SUCCESS in 20m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ea35d3192a864139a19f8e6ca49db200 : SUCCESS in 1h 01m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75d4f98877624c689558f2b0517a5015 : SUCCESS in 44m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/27037c9bd7754e64a17cc96b8399cf96 : SUCCESS in 1h 00m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/de67732c315e4476bb1709c3a15e3c1f : SUCCESS in 17m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1117c99ac4794bc3a1c1a69eafa3d2cf : SUCCESS in 12m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/39aa0c9de3cc427aa170accd526f689c : SUCCESS in 1h 49m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/06dd55c0c51c446ba6c9c5827e1ae051 : SUCCESS in 28m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c9cdba241d143d08609ef08d1b7f665 : SUCCESS in 25m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e47d3a1cf005405688751d9ca4e9b4b8 : SUCCESS in 19m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2129adafb10746dc8d6b8c64404243ee : SUCCESS in 14m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae07c5fbf5d149d3b8a325f530487a7e : SUCCESS in 9m 12s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"3b6ac90862daff3eb03bf7e419d56e60a6208942","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-17 10:04:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d7c71d9a02f264bbdf17a890db660a81a0b38210","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-09 09:24:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0c9ece7635a968b877506ed7dd8b6b2d982e1002","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 09:25:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cb66e759bb477894105826d365ccdfa58253fa24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:23:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ac708fe08964e3ba154488b84e17f71 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adea21e22fd841c1966ae9610400a819 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2b8fc1461864ad7ab34a5e9c00e85f0 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0671d68a4db47179607459636b14b30 : SUCCESS in 10m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41fce4e48c1a494abd2b4fb0f55bddb6 : SUCCESS in 7m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/77f55d3649c949a5a3f38b713b71b1be : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ff0ca0ab438a480abf6734a24101474f : SUCCESS in 19m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ed4b2452bd1b4046a166b6907a6d800a : SUCCESS in 1h 07m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de5f17a4690e452b886610d88ec4d916 : SUCCESS in 48m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/261eddbe18e84d51abdadd89f2c14022 : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a2338b1420834093914b40c64c4c24c3 : SUCCESS in 20m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e20d8403aaec4448a3dcc59b7752f4b9 : SUCCESS in 13m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/12f85ea1e00d4fdc986b6b5a9e065125 : SUCCESS in 1h 44m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1559d2902149411e9488f0aac004f576 : SUCCESS in 27m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7d34583e92349b6984301835833a47a : SUCCESS in 24m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e2b0dac90dc24e1ab223268e332f9f63 : SUCCESS in 27m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5fd7cba741cc496896f1a6d2896f688c : SUCCESS in 12m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95e1ff77c6df46ff9260f565044bde79 : SUCCESS in 10m 43s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"2ee2f84b4bb786394dda86b6010ad51b414ec3f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:23:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4a59ea483bc1baf57a6423812b597147d5e24b2e","revisions":{"35a6baf0be223ecb88b7984621be1728f306597e":{"kind":"REWORK","_number":1,"created":"2021-05-10 10:43:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/790400/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/790400/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/790400/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/790400/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/790400/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/790400/1"}}},"commit":{"parents":[{"commit":"9c0bec4d3c36dfdd02f36f077f9f561d2ec83b6b","subject":"Merge \"Ubuntu: add documentation \u0026 release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c0bec4d3c36dfdd02f36f077f9f561d2ec83b6b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-10 10:33:14.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-10 10:38:36.000000000","tz":60},"subject":"Fix overcloud post configure Python interpreter issues","message":"Fix overcloud post configure Python interpreter issues\n\nSome time ago in Kayobe we switched from using set_fact to override the\nansible_python_interpreter variable to using a task variable. However,\nnot all of the galaxy roles were updated to match. This can lead to an\nissue where a fact set by the os_networks role overrides the task\nvariable set in Kayobe.\n\nNote that we haven\u0027t seen this in CI since the code path requires the\nuse of a VLAN provisioning network, and CI uses a flat network.\n\nChange-Id: I14289df5e248b5b47fb8e60c6eea1ac845f50fd3\nStory: 2008284\nTask: 41158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35a6baf0be223ecb88b7984621be1728f306597e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35a6baf0be223ecb88b7984621be1728f306597e"}]},"branch":"refs/heads/master"},"4a59ea483bc1baf57a6423812b597147d5e24b2e":{"kind":"REWORK","_number":2,"created":"2021-05-12 08:11:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/790400/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/790400/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/790400/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/790400/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/790400/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/790400/2"}}},"commit":{"parents":[{"commit":"9c0bec4d3c36dfdd02f36f077f9f561d2ec83b6b","subject":"Merge \"Ubuntu: add documentation \u0026 release note\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c0bec4d3c36dfdd02f36f077f9f561d2ec83b6b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-10 10:33:14.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-12 08:11:08.000000000","tz":0},"subject":"Fix overcloud post configure Python interpreter issues","message":"Fix overcloud post configure Python interpreter issues\n\nSome time ago in Kayobe we switched from using set_fact to override the\nansible_python_interpreter variable to using a task variable. However,\nnot all of the galaxy roles were updated to match. This can lead to an\nissue where a fact set by the os_networks role overrides the task\nvariable set in Kayobe.\n\nNote that we haven\u0027t seen this in CI since the code path requires the\nuse of a VLAN provisioning network, and CI uses a flat network.\n\nChange-Id: I14289df5e248b5b47fb8e60c6eea1ac845f50fd3\nStory: 2008284\nTask: 41158\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a59ea483bc1baf57a6423812b597147d5e24b2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a59ea483bc1baf57a6423812b597147d5e24b2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
