)]}'
{"id":"openstack%2Fkayobe~790575","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Iaf0dfb34750ba8b8d74a043587a2035b511de6ad","project":"openstack/kayobe","branch":"stable/train","topic":"dnf-centos-repos-stable/train","hashtags":[],"change_id":"Iaf0dfb34750ba8b8d74a043587a2035b511de6ad","subject":"Sync the names and content of the repo-files for CentOS8","status":"MERGED","created":"2021-05-11 13:02:27.000000000","updated":"2021-05-11 16:32:57.000000000","submitted":"2021-05-11 16:31:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790575-dnf-centos-repos-stable/train","cherry_pick_of_change":790247,"cherry_pick_of_patch_set":1,"meta_rev_id":"d9cd768a0549b0f6041a3d1352ec46154668238f","_number":790575,"virtual_id_number":790575,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-11 16:31:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2021-05-11 15:27:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2021-05-11 15:02:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-11 15:18:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-11 15:18:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 13:02:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2021-05-11 14:38:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 15:18:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"d4c3a8c0120fb90ff18f481ba25bb1f3d5c3afc3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 13:02:27.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.\n\nThe following files contain Git conflicts:\n* playbooks/kayobe-overcloud-host-configure-base/tests/test_overcloud_host_configure.py","accounts_in_message":[],"_revision_number":1},{"id":"3bd2c0f7e962d836420194d8c0511e3cee4fb9c3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 13:03:56.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"51832d3c27c71739b4e62b1c334e7964104b4269","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 13:04:13.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"380b2165dd8bcab82c028a1f775bf3f2232c7df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 14:38:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8caf180635184eafb07b84b2a2dadc09 : SUCCESS in 4m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0680672ac5b14d6aa628b25eff8c326c : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/832199f4b9fa421fb5e0cee3f8453925 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a37621e7b33d4728833d5f760c563c1a : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e0f0ba277f24580963e0c2a3410cfc0 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a6ff0d261a54592a87b7753e13d7f12 : SUCCESS in 6m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3537de0740e947468f884e5f3f3ff22c : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a5f211a7c1a94548880cf4a07eb0d01e : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c29cad9c41144c2cbdfb3cd1b3648036 : SUCCESS in 17m 57s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/70a416e6351749a8853eaa9f5b01ce65 : SUCCESS in 1h 09m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c3132635ed664f3382777c5142389457 : SUCCESS in 59m 44s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/ed2f38e6ae7a478b822b0aae061935fa : SUCCESS in 16m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/98f21c0648b8498098758b167f0172b5 : SUCCESS in 16m 26s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/999c9626632a4d6486b14e744cfa5992 : SUCCESS in 1h 33m 23s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/73e54adeff2a4d0da8e0495f12baa038 : SUCCESS in 27m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9d4d9c2f88324822b116db312bbff3c3 : SUCCESS in 29m 14s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/869c6572030943ddacce23fd3dbd74b5 : SUCCESS in 41m 46s","accounts_in_message":[],"_revision_number":2},{"id":"0812f5daa4708a46f9ee33dd9cb221554d4b6d5a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 15:02:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8ede2c4d55812aaebb897eca027726e0f8746522","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 15:18:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"89a15d79ea74748dd7853201ab2fe4f86b20dbbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:18:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4785a6e77d032e4003942eb6b752274c7b48d4d6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-11 15:27:29.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2b5ee43b8122b5e8d80d04023ad67c10d7654efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 16:31:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bcefcc3d317844dfbb71ffc862c750f7 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3162693a6894259aeb71653f7f09856 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4ffeec5de6348a6a1ffd8d3715bcf26 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8529caf5f47747c79b4889d0f923aa0f : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bba4bb3e3a3c4fa193278ea4fa1af9d8 : SUCCESS in 7m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9772493610bc4cff92059b8ae609f9dc : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a14b0ea84324720bf9338ca25a2d1d5 : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84fcfd3bf42142c3911bfde3cf516a29 : SUCCESS in 17m 58s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/23560f2c934b4d85912137b0c36697bd : SUCCESS in 59m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f051b08af5324ad7b3d5246854c9a396 : SUCCESS in 1h 00m 20s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/f96c65e923eb4835acc7be34fcf13666 : SUCCESS in 13m 04s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/15532c029b8841f2ad45428dc51fca92 : SUCCESS in 15m 37s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c98bf1a7ee1c480eb61277abd79d52f8 : SUCCESS in 1h 09m 49s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/dc8b6569684b47eaafecc5c1be07b291 : SUCCESS in 15m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/622c07579a404d83b336b08fcbe825b5 : SUCCESS in 23m 54s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3e1c034188ad43ba984cddf6a73e7827 : SUCCESS in 28m 45s","accounts_in_message":[],"_revision_number":2},{"id":"cd338e8a2925a0ee67d8e67385e934d7ee7fdc35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 16:31:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d9cd768a0549b0f6041a3d1352ec46154668238f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 16:32:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ccfc3cfe8443839d4cc92013515bf1 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c39bdcddb236484311f284afbea8a4e16eafe903","revisions":{"e162fef1736bea0117a7482231360a9e2279d3b0":{"kind":"REWORK","_number":1,"created":"2021-05-11 13:02:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/790575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/790575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/790575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/790575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/790575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/790575/1"}}},"commit":{"parents":[{"commit":"67b8e7867af32746b386f08bc7478240b72d6754","subject":"Restore upper constraints for docs and releasenotes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67b8e7867af32746b386f08bc7478240b72d6754"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-05 21:46:43.000000000","tz":180},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-11 13:02:27.000000000","tz":0},"subject":"Sync the names and content of the repo-files for CentOS8","message":"Sync the names and content of the repo-files for CentOS8\n\nSince 8.3.2011 release of CentOS the repo-files renamed and the content\nis slightly changed. This change sync the templates used in role with\nupstream changes.\n\nChange-Id: Iaf0dfb34750ba8b8d74a043587a2035b511de6ad\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit cdcb90809a2429c6e57d80a2f93fc9b6b218590c)\n(cherry picked from commit 9fdecebeff8cba152f5aa82008bd58d02875004d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e162fef1736bea0117a7482231360a9e2279d3b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e162fef1736bea0117a7482231360a9e2279d3b0"}]},"branch":"refs/heads/stable/train"},"c39bdcddb236484311f284afbea8a4e16eafe903":{"kind":"REWORK","_number":2,"created":"2021-05-11 13:03:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/790575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/790575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/790575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/790575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/790575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/790575/2"}}},"commit":{"parents":[{"commit":"67b8e7867af32746b386f08bc7478240b72d6754","subject":"Restore upper constraints for docs and releasenotes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67b8e7867af32746b386f08bc7478240b72d6754"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-05 21:46:43.000000000","tz":180},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-11 13:03:44.000000000","tz":0},"subject":"Sync the names and content of the repo-files for CentOS8","message":"Sync the names and content of the repo-files for CentOS8\n\nSince 8.3.2011 release of CentOS the repo-files renamed and the content\nis slightly changed. This change sync the templates used in role with\nupstream changes.\n\nChange-Id: Iaf0dfb34750ba8b8d74a043587a2035b511de6ad\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit cdcb90809a2429c6e57d80a2f93fc9b6b218590c)\n(cherry picked from commit 9fdecebeff8cba152f5aa82008bd58d02875004d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c39bdcddb236484311f284afbea8a4e16eafe903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c39bdcddb236484311f284afbea8a4e16eafe903"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
