)]}'
{"id":"openstack%2Fkayobe~790587","triplet_id":"openstack%2Fkayobe~master~I536a526340f6d7dea7a69372904b777292774eba","project":"openstack/kayobe","branch":"master","topic":"docker-image-prune","hashtags":[],"change_id":"I536a526340f6d7dea7a69372904b777292774eba","subject":"CI: Remove unused Docker images after overcloud upgrade","status":"MERGED","created":"2021-05-11 08:22:38.000000000","updated":"2021-05-11 17:06:48.000000000","submitted":"2021-05-11 17:05:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790587-docker-image-prune","meta_rev_id":"7a7cfaa24ec79a6df42809d93ef220f2086aa462","_number":790587,"virtual_id_number":790587,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-11 17:05:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-05-11 15:20:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-05-11 10:47:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2021-05-11 15:31:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2021-05-11 15:31:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-05-11 10:47:40.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 09:41:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 10:47:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2021-05-11 15:20:33.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-05-11 15:31:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"e4e4f0db287aac576c88b1fbfd717c57881409b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 08:22:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"66e7d795f2dca0c29312390902070221f42e31fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 09:41:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8f64138684f44d48733e563ff493b94 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01cf108326d449f9980c2c9d501cfc4f : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a61687d3ddbc4659a63cad2d693f6cf8 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77fbc982d7e84ea79f434fcbb4dc2427 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac3c0bd659254900a5d2b0efcb19f28d : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d7a4cd2fab34f14a398a66298be5aaa : SUCCESS in 8m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cb6fe72158fb4beaa0579b65da00aa6f : SUCCESS in 8m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f619b599fe204c69a8223c138e7a72b7 : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/21de880adf134473b9d9c0dd8a05f7ea : SUCCESS in 22m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7c3c12861ed54aba83dbd2c640706e98 : SUCCESS in 52m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7d3981b693c466b9b846cf463c17009 : SUCCESS in 33m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5f2e3bbad94e45909d7ba69d0bbd4222 : SUCCESS in 53m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/72be1febc87e4f909f47ded0e3facc0d : SUCCESS in 15m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4a97492d4244e389e6581aa2b70c9ad : SUCCESS in 10m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fe034f4fe2924c95a125c5d5e4f39b78 : FAILURE in 1h 17m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8917ea193ef24e30a192c5a2f995dcc7 : SUCCESS in 28m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd7f05e7114041e29237bb7cc7ecc7ae : SUCCESS in 21m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fe344c67fa204e8f9a73947345c3eca5 : SUCCESS in 27m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/67ac6328529d471b9c3d7d5513db70a1 : SUCCESS in 14m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b2b1c32d30f47abbac9bfdc164711e7 : SUCCESS in 14m 47s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"ea6b0e09a1a066ea869078ba61c5d01665aceb66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 09:45:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3f0db418006d44aa78260816b0e0195756908380","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-11 10:47:40.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9c4d9ca94944e5ec2637b169be2fef97bf7b5bee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 11:37:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19a4d0e0a7494148ba4a39f035085b35 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cecdf7ea13674b1294eae87cf4bfbf7a : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/211b0f1651534b3eb33dd19e96636b65 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faa0c1d437d84f7d9afdd281918999a9 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df6227902ce64179b8ba2fa769ae999d : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72dd93f76bca400c82cfba94dd51badb : SUCCESS in 9m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/589565bad3064b7f8d06c4881a1d799f : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8584aa65a50340bf9f450d2d82266600 : SUCCESS in 11m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a5eeba8c87e64b8f97caa89aae1f234e : SUCCESS in 18m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5885fe0e3bf64364918a1a1b5836dace : SUCCESS in 1h 02m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/baf2ac238d7e4207ba6eac36cfff2d81 : SUCCESS in 43m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b596a85c43704d89b9dd7a7a9efd94e8 : SUCCESS in 58m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/db87b454b7d54abfaf12e1e25d38513f : SUCCESS in 17m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f948e652a8a406f8b24c6667dc041a0 : SUCCESS in 12m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c078a8267a1e40e58c5099e8e7995ede : SUCCESS in 1h 50m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7c86a3bcec224dbfbde673571422a149 : SUCCESS in 27m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9ff8b00b5904e21981db90329312c69 : SUCCESS in 24m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/576c586c204042dfbc5a4a808e04244f : SUCCESS in 24m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/912617538c4742cf8f14b6d87234444f : SUCCESS in 10m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/864a1ee229bb44579c1d05ecd4c9c8d0 : SUCCESS in 10m 28s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":2},{"id":"efeb753f9738e2dda177916997a56c3ddd4523d6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-11 15:20:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c68d1334e8e6bd0b4677f3294b08462a6a846026","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 15:31:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4725d8360d32a7c0ca41872806324a3c228a7ab1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 15:32:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a397bf53dd12d313276a6b5f13d7174d1f574f66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 17:05:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d345b32af84f18b8bfaea3318c71c2 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70f38e1e83e447d7bea7a3f29274b645 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b469befa0030437596de95f867f4f135 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fe50b61daa04d7b887990edcc4e4282 : SUCCESS in 8m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/273a5ffca529442f98017defb6e8d969 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2fa3517fbf3246b9b24fa7c4075cb336 : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bfbed19042334affa5b9f13ca8ba9137 : SUCCESS in 21m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f36809b423a5478c846a6c7616c48083 : SUCCESS in 1h 01m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89612f2567a846d2b3a5770d69cbe586 : SUCCESS in 49m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c3a05d6a815244aa82412264c030463c : SUCCESS in 1h 03m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c187ba589ace429e8e83abd48b92cd1d : SUCCESS in 22m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32a1433155cb4eaa9b3abf1afb660035 : SUCCESS in 12m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/316bb139a6f441daaa8c9423ee066600 : SUCCESS in 1h 32m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/73f9059c0fcf479fa8c5b8b6e23f7986 : SUCCESS in 30m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe8453ee07b449a99969d9d5bcfd1b76 : SUCCESS in 24m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8ba034b9d4c94a6fae1ecc42f12c5563 : SUCCESS in 26m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dc175a28cd7b43e5a63b4363a4914c55 : SUCCESS in 11m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf945e167f174592a7e0269984c82d91 : SUCCESS in 11m 23s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":2},{"id":"f4d8d20711b572f62512761910832c14b6a4b752","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 17:05:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7a7cfaa24ec79a6df42809d93ef220f2086aa462","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 17:06:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c454bf00d8e44eae8b031562f760e22e : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"93a8baca93b7525c41d164c13822f60672e41bdf","revisions":{"618af40f10d954cad38ae290de29d9bb585b0696":{"kind":"REWORK","_number":1,"created":"2021-05-11 08:22:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/790587/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/790587/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/790587/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/790587/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/790587/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/790587/1"}}},"commit":{"parents":[{"commit":"bd47fd025def5ac34a79b0b52e4c8fd65ffcc51a","subject":"Merge \"docs: aio: add note with link to automated setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd47fd025def5ac34a79b0b52e4c8fd65ffcc51a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 08:11:26.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 08:11:26.000000000","tz":120},"subject":"CI: Remove unused Docker images after overcloud upgrade","message":"CI: Remove unused Docker images after overcloud upgrade\n\nOvercloud upgrade jobs often fail when running on nodes with a 38G root\ndisk. The overcloud upgrade itself generally works, but post-upgrade\ntesting fails, for example when RabbitMQ gets in alarm state due to lack\nof free disk space, which prevents Nova from functioning correctly and\nmakes it return \"Unknown Error (HTTP 504)\".\n\nChange-Id: I536a526340f6d7dea7a69372904b777292774eba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/618af40f10d954cad38ae290de29d9bb585b0696"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/618af40f10d954cad38ae290de29d9bb585b0696"}]},"branch":"refs/heads/master"},"93a8baca93b7525c41d164c13822f60672e41bdf":{"kind":"REWORK","_number":2,"created":"2021-05-11 09:45:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/790587/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/790587/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/790587/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/790587/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/790587/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/790587/2"}}},"commit":{"parents":[{"commit":"bd47fd025def5ac34a79b0b52e4c8fd65ffcc51a","subject":"Merge \"docs: aio: add note with link to automated setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd47fd025def5ac34a79b0b52e4c8fd65ffcc51a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 09:45:04.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 09:45:04.000000000","tz":120},"subject":"CI: Remove unused Docker images after overcloud upgrade","message":"CI: Remove unused Docker images after overcloud upgrade\n\nOvercloud upgrade jobs often fail when running on nodes with a 38G root\ndisk. The overcloud upgrade itself generally works, but post-upgrade\ntesting fails, for example when RabbitMQ gets in alarm state due to lack\nof free disk space, which prevents Nova from functioning correctly and\nmakes it return \"Unknown Error (HTTP 504)\".\n\nChange-Id: I536a526340f6d7dea7a69372904b777292774eba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93a8baca93b7525c41d164c13822f60672e41bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93a8baca93b7525c41d164c13822f60672e41bdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
