)]}'
{"id":"openstack%2Fkayobe~790727","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Iacb579a46b0e9769a4c404a858d17968f74dd7e0","project":"openstack/kayobe","branch":"stable/wallaby","topic":"feature/chrony-stable/wallaby","hashtags":[],"change_id":"Iacb579a46b0e9769a4c404a858d17968f74dd7e0","subject":"Adds support for configuring chrony","status":"MERGED","created":"2021-05-11 16:53:36.000000000","updated":"2021-05-12 09:55:01.000000000","submitted":"2021-05-12 09:53:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790727-feature/chrony-stable/wallaby","cherry_pick_of_change":784460,"cherry_pick_of_patch_set":16,"meta_rev_id":"f53f99d3ef095f6fcc5ee25e89e59dc8f3d6a53f","_number":790727,"virtual_id_number":790727,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 09:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-05-11 19:15:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-11 19:21:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-11 19:21:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 16:53:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-05-11 18:41:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-11 19:21:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"188570d221c13c280d2acede8eee0de1b7018c8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 16:53:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"79f1d868c90c6cf407a99363db1b3a606ede7d10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 18:41:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a641992f67f244b5a230cc78f769e648 : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d84957259fcf4032afbe002198aa3216 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da8a9f011d17414f911292b80504e8c3 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0caf2c09dc5b4d5d99316eaf44e4d172 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01657f33adae4548848b31b89fd6aa7e : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f9c024bc69f412e8030617ac7011044 : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e79572d3d15492aa4ae62bcc97a355b : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dcc16bebae274e9ebe5f2c2779e9500c : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/be19ca92e6994c1eb646d2e93e2c566e : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/127f824b2a314575bfadc49785826ba6 : SUCCESS in 22m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c70dd9ef8b044b1aa16c15e53804bcba : SUCCESS in 1h 42m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b447a3c87c6349559b497078ba0f411e : SUCCESS in 43m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3939ee26ce5c459b915d4b547061793e : SUCCESS in 1h 22m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7b35ce92be554ea7ae00f9c98d38df59 : SUCCESS in 17m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b548c43a37f4d9384ac575ccc4a5032 : SUCCESS in 12m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/299df47a51f44eafa04b55a8b59aed1d : SUCCESS in 1h 46m 32s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/92c09ca2871e4d62b95e01c53f1237d8 : SUCCESS in 42m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07ba8e60ae8f433d951a1027847ee97d : SUCCESS in 24m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b9b228bd5dba4169aa3a00c8dc380724 : SUCCESS in 25m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6f1eeb7f4dcc442fae4ca518b742b525 : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/577c1f2242f64218ba32b591f7b23ae9 : SUCCESS in 10m 15s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"a7682a425b06b81f667e6f131bd586fa8e68301e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 19:15:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e2ad6002feaa22b7d9391838ac701e194cdcc82d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 19:21:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"03851a611150e37f34916a60cd6bd7827d27a2a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 08:05:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bd24b621339ff5901c22c532d208b4a6873c64f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 09:53:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45acfc0c6cb44e36b30ddf73b248a6d8 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c6ecf4a9e1d4dddbeae68f1c951eae9 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efcae69c4a554ca2a93df17ecf752ffe : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d3a3deb83c34f33ad0a200d0ef4862e : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40a311568ac9492caa8f2c0798fcd470 : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b1bf4516f47c47fd9e2459da00c04f23 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f069af517174216af6e7e0b619bc3a1 : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/262799760dd94672974a3974e74fb8fa : SUCCESS in 20m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e26147786dc04d6ba15a7f533924cb0f : SUCCESS in 1h 05m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fec7c1ba85bc4707b4bf82c7609c6f43 : SUCCESS in 42m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d169a9251f384129be94b0eb2aab0703 : SUCCESS in 1h 07m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/07d52828cdb84a3ead3e4f46bd729e1d : SUCCESS in 17m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aedaf485ee654b68b1ebfb8c4b8d4324 : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a0bf50c1583445199f9cde5ff9ee945d : SUCCESS in 1h 44m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4877f6a654bc4b06842c7e8980b1d60c : SUCCESS in 29m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c24129d8e43e49cd86cd88a9d634f044 : SUCCESS in 24m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/67159fea8e234f67be1afeffd4261867 : SUCCESS in 25m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9dfa13c6cc464ef68020a4318a9a0e75 : SUCCESS in 12m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6924dfd00a4f4b0aa808a1127875af97 : SUCCESS in 10m 07s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c66f65cf071bfa81e5978d40362102ea53f25964","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 09:53:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f53f99d3ef095f6fcc5ee25e89e59dc8f3d6a53f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 09:55:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b528ea2bc0ee4730b809a0c3bc0ee51c : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3a630e77eaa341349a77555e8e3a441a : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"010a18334f08d173c9b8185f739324f58a37da89","revisions":{"010a18334f08d173c9b8185f739324f58a37da89":{"kind":"REWORK","_number":1,"created":"2021-05-11 16:53:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/790727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/790727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/790727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/790727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/790727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/790727/1"}}},"commit":{"parents":[{"commit":"7a56c95a4ed769512d43a41e1b8ff1d055a99a07","subject":"Ubuntu: add documentation \u0026 release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a56c95a4ed769512d43a41e1b8ff1d055a99a07"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-02-20 00:50:27.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-11 16:53:36.000000000","tz":0},"subject":"Adds support for configuring chrony","message":"Adds support for configuring chrony\n\nSome hosts in the kayobe inventory might not be in the kolla-ansible\ninventory so it makes sense for kayobe to manage NTP.\n\nChange-Id: Iacb579a46b0e9769a4c404a858d17968f74dd7e0\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/790726\nStory: 2007872\nTask: 40240\n(cherry picked from commit 085cf7d1759879169dddc69378ad784e1acaf524)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/010a18334f08d173c9b8185f739324f58a37da89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/010a18334f08d173c9b8185f739324f58a37da89"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
