)]}'
{"id":"openstack%2Fkayobe~790730","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I536a526340f6d7dea7a69372904b777292774eba","project":"openstack/kayobe","branch":"stable/wallaby","topic":"docker-image-prune-stable/wallaby","hashtags":[],"change_id":"I536a526340f6d7dea7a69372904b777292774eba","subject":"CI: Remove unused Docker images after overcloud upgrade","status":"MERGED","created":"2021-05-11 19:16:51.000000000","updated":"2021-05-18 12:32:49.000000000","submitted":"2021-05-18 12:30:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790730-docker-image-prune-stable/wallaby","cherry_pick_of_change":790587,"cherry_pick_of_patch_set":2,"meta_rev_id":"6126e819d70dcb39f4a39cb4948fbf1f5a249f62","_number":790730,"virtual_id_number":790730,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-18 12:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-05-12 08:28:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-11 19:22:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-05-12 08:28:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 19:16:51.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-05-11 20:44:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0134d332136aa0b5ce445c6ad8b7ddfb826e432e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 19:16:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3e2c7ae441fa63059a9f9fbcd1ca07643cd513c5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 19:22:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"929dc8f036e5404524e4df5bf3cfeade30de5df9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 20:44:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26819342d9504637b47c0fb1eae69cef : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6595fdb04fa4641b1e283dbf5bd9d17 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e27631ec35094f1c9f26dc10d7083b77 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1381c0573454173a63be2ed6bb047c3 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e11223c5a1ff4bd286b3f6fbd2b46bc5 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/734fae107281471f913bfdd565544b46 : SUCCESS in 8m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d9fdb76fb46b472f9c10168b368758da : SUCCESS in 7m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae0f2daf84374e89907c77b67ce73b22 : SUCCESS in 8m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/468b996d611c4500bf1d836e6b1d713b : SUCCESS in 15m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d9450a787af944caaa4021e9800e43f0 : SUCCESS in 46m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c58ad60ed12e4962810547c22717d04f : SUCCESS in 36m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6a6fa9bfb1c54bf18a41d5ab0c960c11 : SUCCESS in 50m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/42df86c56edb4ec6ba78d132129d3050 : SUCCESS in 15m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7e292eb1a16494ea438a5b199a2a52f : SUCCESS in 11m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ee8923c9353a4df898589f651bd9d98d : SUCCESS in 1h 25m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d2d7a9abda38472a87f83385400f7c32 : SUCCESS in 26m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95396b4924fb4118bd0624773272d8b7 : SUCCESS in 25m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c88059e6cce14e56878a63fb31cbc148 : SUCCESS in 22m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/569e2d6ee9c8439e9b03710540aa2331 : SUCCESS in 13m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/356728ae39bc4c5f8be66511277a584d : SUCCESS in 8m 57s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"c61e52927d04aad094fc0a2c2bc5918341ee9dec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-12 08:28:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b73165ce70dcaa8ae93b3b82bc7f07a685a9231","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 08:29:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"14492117797cb7a375c32c052462272c10197b2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 09:53:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19a9d799698241bc899ba2c669bef954 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c4a6bce8164460dbbdebb2e2ed0872c : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3b227f4fb89488cb42028a8f6c0d859 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2f2433e55a2406485c2a5cca3a819f7 : SUCCESS in 8m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/765e7c350aed474eaf1e53210b88590c : SUCCESS in 7m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a8fd4a91e9f3441bacf4b6af2db1d283 : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30080fb43f764732a65a42e7be09501e : SUCCESS in 21m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e0672cb254d145998209a8805025e104 : SUCCESS in 1h 05m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46aa84748f454ad59a4a83e100498a05 : SUCCESS in 48m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f0c71245a3cc46cebdaaa95fbb33bc9e : SUCCESS in 1h 01m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3fdc502d915d454e95ebf01b99138dce : SUCCESS in 18m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa9ff978c90a4255bc30bc34a03c857b : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/da8a1cf141564a14897507f581f0b8ed : FAILURE in 1h 18m 24s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bed44d3fb94b486ebb888bf72ef5daff : SUCCESS in 29m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0514cca13f6942f4ace5a76c055ed748 : SUCCESS in 27m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/33fef89fd0a8453e860aba76691c7bd1 : SUCCESS in 25m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b3ff7e836c0b4d0cbdd2c402db97983b : SUCCESS in 12m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fcb89fd48b64937a297caf09469818e : SUCCESS in 10m 36s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"1dcd9891655d48ae11928a20afd63293ccc89664","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-12 10:35:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b98ac1f8d7b9d9cf90f82828b19c24d149a7ba55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 12:34:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6c69afd1455410db9e0752aa501e64c : SUCCESS in 9m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f0d69b506846e4ada0b85bdd4e4af6 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45c3d49890c243fda52f17e08e1af401 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82691de00d9f452fbccc0b5c416b86f3 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d85b786f2794d54991469498f21de96 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a5e2f518e3d48668d33f01b7b4347c5 : SUCCESS in 8m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/603dee8dbef34809b3c941b3639aaf3c : SUCCESS in 7m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ccef6a96fca448a8950bd4f216a18628 : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c92fa23642b4b88915433b9a4e8e036 : SUCCESS in 21m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1f1d792a57ef479689d60b9c1410da55 : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe8916d578394359afecc3c107f94cea : SUCCESS in 44m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9174f465022c466da7d438bc9d9ccbf3 : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/68fb6cb291a84698bf6ed4fdf27c3fe2 : SUCCESS in 19m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aeb80027e37045c5ad61108c59ef8b12 : SUCCESS in 14m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a2e089b06929451581f26b976fcfb5f6 : SUCCESS in 1h 51m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0405a5a3f55e46aa9d08457c9ce1c5aa : SUCCESS in 27m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8610bf58843a4c1ba3a09019db2a0e69 : SUCCESS in 24m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/76488257320e4740b04bdb7e2f736b3a : SUCCESS in 23m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a43baff9aaf641268cb0d79a6219b806 : SUCCESS in 14m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41e1aa4450354d7e9417d9738d67f13a : SUCCESS in 10m 57s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"4e0c2e26eae2b5cb71f34a842d136b981283dffd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 12:34:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebcbcde85f80ea723999a8f86b7d7f9fef4645b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 13:58:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53d1c72839a64dec954c8c9390f861ea : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25a97922d07d436c80ad795c4e3dc444 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02172b20c35a41dd96939589b2676d37 : POST_FAILURE in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59620c99745541ffbc1a1132d62fba37 : SUCCESS in 9m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a1e4a1fdbeda4d1fa66119c29214197a : POST_FAILURE in 9m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9221727e31e4488eada26e8bbec72feb : POST_FAILURE in 14m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/672213a6030347649c8126a51bff443c : POST_FAILURE in 22m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9028c4c39c3a4924885f3913d25b4221 : SUCCESS in 1h 10m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9cd4a4a61624f9cb3a02dfa45f91f86 : SUCCESS in 49m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/360b010316514a3480882f181389fe76 : SUCCESS in 1h 10m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bfd7b8e88aec462280721ef48f9e8473 : SUCCESS in 18m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b105fa79279947058aeb0b015d94a7c3 : SUCCESS in 13m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/926d1892357248358dc635cc5232f28e : FAILURE in 1h 18m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0d286346898c4bcba0da94a3ca4a8fce : POST_FAILURE in 34m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d33c3fb42ffa438382d56a05f4955211 : POST_FAILURE in 27m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2f2b44fcccf049358bdaa50124d47b73 : POST_FAILURE in 29m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/20c8a28753944b02be343af54e72cc7e : SUCCESS in 13m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a061190ce454a67aa889e7fb8851abb : POST_FAILURE in 12m 46s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"b08a2a5d9d085e5af366d172c2ce4a30867ae901","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-12 14:03:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9e8341bcdb2de5ae4a6e5351ef80921d4f3ded1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 15:28:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dc01ea8988441bc80b5270898896cac : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8409f3e2b743451d88def6b6dc0d950b : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e61dd93abb4d4c12980f0dfd9e0a38d2 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75a3fc562e274a3fa2fb29c76ee80827 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bda7e83660894d0cb0fdfae1ef0b988d : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/213ac8b2cfea440f9377d99edcae273b : SUCCESS in 8m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/32b61a417e604b26ab14567f1e515a7b : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5437fc4f90984efd8079c2cd8ccff867 : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7749d43fd9ec48cf90cb007416b67f77 : SUCCESS in 19m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/67585cd89b5e4f38abf5980e8093d124 : SUCCESS in 1h 06m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c3b0fcb622c42008df4a252cbdb519e : SUCCESS in 45m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/28ba002587cf467b98e6125e7fe235f5 : SUCCESS in 1h 11m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/951e9e8cf4ad46be816b2bea09adf8cf : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd2906bd82434553a18456d160038b8e : SUCCESS in 14m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/22ce80d180c34a26b8caa691c7656b40 : FAILURE in 1h 20m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/84aa20adaf934c5d9abe57391fc1ff5e : SUCCESS in 30m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a082aaf6b57146d9a2055f2d43771b91 : SUCCESS in 25m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0980c39fb3054892942a57e0bdab75a8 : SUCCESS in 26m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a4854401cfa9495e8d098860a4103166 : SUCCESS in 13m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaf002b21c6f4ac0baceb507f8d3794e : SUCCESS in 10m 44s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"bfe973abad4e9ccb5c23d0e8de4f3439c292ae8c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-12 16:26:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d1e6d34ffd3452aa48b323764d593b69e85dc80d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 18:01:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbb55f7327774e32bbf3fd95ea6d37d4 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d64ff68e3174a2e97187a7184813cca : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c76285c7038e42c392bc35ab07341240 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c22bd4fe6ae24955ae0141c8e4a72105 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2b647caaa0242c1a48163f5343eb86f : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb3ad483774400faceaf36281664bac : SUCCESS in 10m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e8cb3ca3ec394b948cd48b8763ac6747 : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7cc4b03f668f4d7cbccb67ba3a9afeee : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7015200cfcd340ad86c26106a18c9e1c : SUCCESS in 18m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3c6fe4d16bd44b0d8060504b6807bbe9 : SUCCESS in 56m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cce7ae5e0af74f2287ff2011ef630fc6 : SUCCESS in 1h 07m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cec45fad2e414aafb8bf031d7e012966 : SUCCESS in 56m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e61c5fd9bfb24695a81ae506211431c6 : SUCCESS in 32m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f8e71158d7f4858add2fdf72a3d1ddf : SUCCESS in 22m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8a2508b924d848adb3b6ef69520ca299 : SUCCESS in 1h 33m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ea35f8cdba2c4fc89e2b00924043304b : SUCCESS in 27m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11e8d2f51b224ca4add4ae03504639b6 : SUCCESS in 26m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fba6f2feb2cf4a76930afe7f321b6f9b : SUCCESS in 23m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d2ae24e6606b42629c06e733c9d3da28 : SUCCESS in 12m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5856ff9a14544064be280f8ed434d735 : SUCCESS in 10m 12s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c7833abd630b181232accc54de2b403419f5af7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 18:01:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f27914104b6ca80ccf5e3a87a050e076e9089d02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 19:27:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9bd9a9cc123435692280c830f17d575 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb377e1f3f2b4937886c719e38247a38 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dade6ce621864a01a1c6ff2f5dee2b9a : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc60e5b8df04dcea3f70d2ff61b0298 : SUCCESS in 9m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cf4b5c9c99d74b0f8f5949ceb1c8e936 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c118d2b56cf4eb9a7b4f2a7d0549050 : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/288065a5b0ae47e398de840cd5dd3845 : SUCCESS in 17m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d145034a0c964da0ba6f14432b225934 : SUCCESS in 51m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad2f0ce9301647569207a38d8ddb060f : SUCCESS in 42m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9e7c3e86f4a149e0a2f0ceb28b42166c : SUCCESS in 58m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/68d24170c114409280df565ff7922daa : SUCCESS in 15m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6cdceb63f344e428ede0fe7d01bf701 : FAILURE in 13m 02s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d819994f6bf940b2bbd15e34a13c25ac : SUCCESS in 1h 24m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/32b9372d8e75496cabffc720da97bf98 : SUCCESS in 23m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b16a0edd2de342a7a0181336d775387b : SUCCESS in 21m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f824c35e1f574eceb7f98f9e1c68b4a5 : SUCCESS in 39m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/15a0a1fa6b0a493d8616da45fae49286 : SUCCESS in 11m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62e30c03f5ad488ebe3bdcae5770f835 : SUCCESS in 10m 35s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"600875331a615fdbfcc9e8c816210d473b2da379","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-13 08:27:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"57f4d236314870729a7acc84409ac1ebd3c68fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 10:14:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5083ad9957804235b13ca59645de37f1 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183557c2cc5c4a7fb8236c2136e3bda3 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edaf36501c0b4b51a932df75f0c9fd69 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24db93812add4849a21ddfd98de1c527 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f61c16ed177401cbebb593e1e3757f7 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2683be777a20427f95561132f2e05909 : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30bc7986d46c4c9ea85f8006df3f73d4 : SUCCESS in 7m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1505aafccac643fea1713ebbde655550 : SUCCESS in 10m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b2b502cde034d97b04b9d3928f9c2a8 : SUCCESS in 18m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0f81120c57444d9fb307c58840d57697 : SUCCESS in 56m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54e2677a85164eb2b9c791907d5c7089 : SUCCESS in 40m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b0bc091dfc9b403ba810cfc8ea60a7dc : SUCCESS in 59m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b3746df97fba4ddd8da869749f5d730c : SUCCESS in 15m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a284e1069114363a0eb0eae0c25ce40 : SUCCESS in 10m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b410652abeb94557934ae5c1ca17c795 : SUCCESS in 1h 45m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4bb1a21fd16d46649d4a9586cec14a95 : SUCCESS in 30m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dce37e66d404acebfce6546768bd2cc : SUCCESS in 24m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4ec0ac52f2444dca89d0b8eda732d72d : SUCCESS in 23m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8da38ded125248fe8ce95f4f441a520b : SUCCESS in 12m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e8cb5a3dc9e412a83d25d6ca1a403a2 : SUCCESS in 10m 18s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"42b3b7f1ac0a3ebc2ad57b0495f7fab432f01cbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 10:14:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"51017624bcf09c444f3173f9fcf63a9237c8da64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 12:03:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7fd187773e8466d8c47b1f6bdefdd1a : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaf43a31a20c4fd1b5fd41d9b60947c2 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cf12b027f414e3a95e45755d9f489d9 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f53cd0aa81f34d248a0057100763c689 : SUCCESS in 10m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61f23ac52b1e492796fa7e15676b3fe4 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b5f982394d04e23aaa13f477a4490d4 : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/301b269d174542ba98c1f379dcdcd3d5 : SUCCESS in 16m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ae3121560e0342adaf02e2e75de26e1f : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a37f049f6ab540a3b40b203bc3135ba0 : SUCCESS in 38m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/82ebcd05375740a6adec1dfb253aa8a7 : SUCCESS in 1h 16m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/080589b403b846cf94c2c9e4276d3366 : SUCCESS in 13m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04c4aa1f26da48fd96fd8e0fec547a68 : FAILURE in 12m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8964dad6e2614c898867198290eee0db : SUCCESS in 1h 48m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a11896ad17b641a59c108ed7bac4d747 : SUCCESS in 31m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10b8dbb638db48a1ae501440d133afeb : SUCCESS in 23m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/26e48cadb1304248a4c07a40356fb8f6 : SUCCESS in 24m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/40b5e50f146441189317a836dd68b09e : SUCCESS in 10m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/679f607a21e24ce4bfa303662698f1bc : SUCCESS in 10m 17s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"cc5bcdec7ff7c6ee361042045d576bf3ac393270","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-13 13:51:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bdd33cca95b982a7813c69733fecb433d1d3ccca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 15:57:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e4ba54722bd404583a5037279067a1f : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38ac14492b95414490d4981359e39099 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/414eef4a9cc8450bb159667d82459244 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7a7c789a51546f9b4c03dc0615b35f1 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab8d49b56f004f5fa620f52cb8ad70e3 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f890fe3085b4469987ffb479d2733c47 : SUCCESS in 7m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36115ed6bf7441b0a9da6f3babdf6a53 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bc6fac2850224b2fa9e194943841c31b : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4304fc5cbeea44868b78445bf1416c6c : SUCCESS in 20m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/41524f0774a240389f27f9f2c790bcb4 : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f8c8ce953dd4545bac7a47f30fa5ea0 : SUCCESS in 47m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/26490b8de87f40ab9e15b527f9f11ba8 : SUCCESS in 1h 08m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/25db558f965d477e8ac492cae8f9291b : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5aa654084f7f4af4ad9a40917c04e216 : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b0993ea1c09241d983d53aaeef635f33 : SUCCESS in 1h 59m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0f370aa33e4e47609a1ff36ec3dea9b1 : SUCCESS in 31m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3239c98b46bc4ffaaed04d0d63465cfd : SUCCESS in 34m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fc7ab4dec5cd457f9ea7b322a277e208 : SUCCESS in 26m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/03b4af0787ac4a909e2bcf06d51c6162 : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbbdb5e06bf0406fb937ab2b17d6a628 : SUCCESS in 10m 56s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c9abd0a38ea88c094d2d3c4648637e50a53aaf39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 15:58:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5b36fe52c163e1affe13c1b418b3536d748d7bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 17:59:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1be6d616c67e4efba40cc06e66c93594 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37b0ae7de2614cc7b9dd84216327804e : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/119e98ce69bc46e4812cf3202725c2bb : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/173e7d9eb6e24822b7867773c5a4d3e1 : SUCCESS in 8m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c8564bfeb06a43749343f9a3f8f5a1fa : SUCCESS in 6m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/75094de854a14e5ab56b70f0931fee61 : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4c9c611bead4c11b3700062800b998f : SUCCESS in 20m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d71f77c4b03d4135a5eb1efb85fcdd63 : SUCCESS in 57m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80c3d366521c477094c382cdede4123d : FAILURE in 34m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/975b21bc3b844af3bf28be74a013494e : SUCCESS in 1h 00m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8ff24e3edad349b5837b0c53b4c25db1 : SUCCESS in 18m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8de37449d08494cb1b7459b9e081114 : SUCCESS in 13m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/22821f8a022e423fa1226a7b818674df : SUCCESS in 1h 52m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3e3cc3e2b1d9461fbb10579aaacfb583 : SUCCESS in 29m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/715d4a00be7c41bea1b49071347989e0 : SUCCESS in 23m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9a6c0c686dcf4a79b9dc4764a08fd434 : SUCCESS in 23m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/280956c0f6934c359ced29c10bd69762 : SUCCESS in 12m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30e015f31eb4447fa780cf7917fb379e : SUCCESS in 11m 07s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"71f6fbb57ac8c4d5d6912560d603cafda97160bf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-14 08:00:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1061d78ad5a8a1a09fffd489661ffa8d350a8bb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 09:42:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/970ebcf94d894a629405e648015efafb : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28c5be212881455db36d694d70644223 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6810b8621c914a1c92c2f58af6890602 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/743f534fd7454140bec07346d71872ca : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3213e948395a4538b8eeb87ac7835083 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26aed94df32942948dbdbc3c239d8209 : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af168bd765a24cf58f5094e28b74d7d1 : SUCCESS in 7m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2ad873c0bb8044479b9049ce9834c6b9 : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e4150544be514372ae941b8b7098d1c9 : SUCCESS in 19m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/46f4d98e82e34142b504d5d064791c9b : SUCCESS in 1h 01m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/420dec24776f4dc49d3f5dd247d2b33b : FAILURE in 41m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dbb8e69abc9243d39903709f939ed316 : SUCCESS in 1h 11m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3f7ac9acd29b42b6a7ce00be3052fbd6 : SUCCESS in 16m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee8662a5f4434a3583874aa63a01d0ab : SUCCESS in 13m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/33c6de751f484275ba8eb83dd3369e27 : SUCCESS in 1h 39m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a1357c00e4804b96abec1d5ab7527974 : SUCCESS in 30m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a484775462d24583bac783427d682655 : SUCCESS in 23m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/45a2e5a6aacd4a0c856e4051484e1676 : SUCCESS in 28m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/023871b071d04e94ac9feee1b02c07ab : SUCCESS in 11m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c14c75b3a40475095ab8ebb50d2bddb : SUCCESS in 10m 10s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"da6d2dd98f4e8e7a33c808a10fdab37171b88c2e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-17 09:08:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8719792c7f653c6f07678c8c739dae07f85e5ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 11:05:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27f6cc36edc246ad9fff10293984c2b1 : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e3ba0e0e5845f29b2a69cb5e8c5b52 : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d508b862c39b4cc69ea057f8f1b3b575 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35047f6c6745485a95cf92e3d1b8c520 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a09b2cd269b74d20bb6536b63d7126c6 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bff11f713e644d0ab87b8c2f473f7678 : SUCCESS in 8m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b335a6d8ae6b47288dc29824a8a06c66 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e29d60eff75240a19c26d629c6620739 : SUCCESS in 14m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ec12df0e3e6645c4af5534364a5b18df : SUCCESS in 22m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/209058be1e4c46678847f17539e18849 : SUCCESS in 59m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c745bd5ae8bb4858837c43649d1036bc : FAILURE in 37m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9452e107ba084c9b84bae7c359955b82 : SUCCESS in 59m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/acab368c0ba14e0fa61cc1a782845931 : SUCCESS in 18m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d97d47c2cbe498aa237c4e30993634a : SUCCESS in 13m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/24d3e3497e644b0b84f1b45192bdd59b : SUCCESS in 1h 48m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c132a1a9236c46d080fb96ee1a3a928b : SUCCESS in 32m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58d7ef2c1ac149b3a62618c741670c66 : SUCCESS in 25m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/add81fc73c2a441f867915c54b59e330 : SUCCESS in 26m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c7315fe1304843adb91b8cb0da2c356b : SUCCESS in 14m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d102c3aff8c49e7b4ab8fd3d6ef3fe3 : SUCCESS in 9m 47s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"0559485b81d0e75ddefe79e8336f21f7d483e2b1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-17 21:02:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"64686214b1656b8f28af418abc1bb214e6a56dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 23:59:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6c8f1aa89e645cd97fa6d136c6a89fe : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a1fcc303a5d48be9eaa492fb9d6a1ff : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab8f32ca9eaf4a9fbed65d3669ea26ec : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/504cc382e7414b55a268a63a73eb6791 : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34770d7aea664ed2ab4c247e9600e0c5 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dd3fbe548b349fa9bf17bb59742afc1 : SUCCESS in 9m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c3c1feec077946fa8db987042790b48b : SUCCESS in 7m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c11c8ce66544db3b2ff729c5561397f : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ec72c2615d4f4a4bb7962a65ae9e591a : SUCCESS in 21m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a370b482f91a48388e2d129597efcdf7 : SUCCESS in 53m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/800b3ceab3494069bd660eb40a42286b : SUCCESS in 42m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f78c61bb42ec4854b73455c8bfc0f65b : SUCCESS in 1h 06m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/737767597d0747168fc1ba71545e6e81 : SUCCESS in 17m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36e9520028eb44d9ab5b5e9e21e62dd8 : SUCCESS in 12m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0e789597a8f340c69d13a348e50a924c : SUCCESS in 1h 44m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5bcaad64664e4c3282898ab9e1c4f825 : SUCCESS in 30m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2abc2b03fde44b60bcbe7cd805b22131 : SUCCESS in 22m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/170df1cd8bb74f2f9f15a0426673f58a : SUCCESS in 25m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f383829a55fe4a32903cd57bebe2e6b5 : SUCCESS in 13m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0632e72e652406aa1ed7cb4117164eb : SUCCESS in 11m 05s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"7f9120e25dbf89e1bba3e899ba2f67e1f87f69a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 23:59:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5478bfeb8d6f381eb4948d2022ed8b1e1ad24f80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 01:35:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d21109e79042e5b2c8354dabdabe7b : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f739581a4d0455c8136e157a82ffc6b : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fadb87e7b24341438ceb0b3e83e8080d : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e96b1ef57a44feda99a08392ce59226 : SUCCESS in 7m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/468553657ca34b56847d1a04a29e2b5d : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7654b8520e3749549017977e10ee4122 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/756a1199921341b6ba8ce5404c81d9a4 : SUCCESS in 16m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e37192ff636b4dcd9fe4ad9fffc0d497 : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0336132a4b1146429c6add76c3a4591b : SUCCESS in 45m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7f0c3ec42e56476da38a8dbcce36708f : SUCCESS in 1h 10m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6ca574681a464d97bd6b507d982ae147 : SUCCESS in 14m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d1ff81079e642aa851d2f21dfcca581 : FAILURE in 12m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0fbc54f4af494d41aae2137315ddc4f8 : SUCCESS in 1h 33m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/654d33bba0f74e948c2c7ac7ff9bc64e : SUCCESS in 24m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb0f85c7d7fa4c0aaf9b75275bfdaf7a : SUCCESS in 26m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aad1b7284850418c8eb775ca9415b64e : SUCCESS in 25m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/41afbfaa1b7e4ce6838725db57657b3c : SUCCESS in 12m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/915956d48c324e8e8ec35078f9bb0acd : SUCCESS in 9m 43s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"48077180408f58b417694ddca83959932d62de53","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-18 06:31:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a9e552daf2f128d8dbe4bcd4e4bac1cda9e1cc58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 08:29:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9c8fd6e9d864524b20f123d02aace6d : SUCCESS in 8m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a4e9e6651b4b60ab57fde4c9a0ab39 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/769c5b9221c641afa68f4abd21690a31 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80933b01c7044a9581607408b442e10f : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e07b8d1f6e734d8b9229aa2ef54bd686 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d3c8091d0d1415c8ae0687848f8f68d : SUCCESS in 11m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a0a19a0b319146a8846b53d84b5a2245 : SUCCESS in 7m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/63094c621773487eb8d5734f622c3ca5 : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/80407b35f48342cbaae7e464e1132803 : SUCCESS in 20m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4ac36377cfdd43be913cf0233fdbe727 : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddd1625509a648c7aa908e2cf27274a2 : SUCCESS in 47m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d65cef042ac24cbba5db8ba86ba432e4 : FAILURE in 56m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1214929b12714bfe9685dcf500e28556 : SUCCESS in 19m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b23da77347814c5f8993f131b7894aa2 : SUCCESS in 16m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1de2df2c16574030a3c2b0fa8d2e23f5 : SUCCESS in 1h 28m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/967fb8a93b9a44a589065d13dd9cd79f : SUCCESS in 26m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/894b50f3743b4f3b8185cd7e101511aa : SUCCESS in 34m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7356cff6cf4f4cf389cb0ca4c9c78ce2 : SUCCESS in 31m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/82b916dacacd41ebbdff2a1a7af1d270 : SUCCESS in 14m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/859403ec866b4dc4b4ae601bc2ed7731 : SUCCESS in 13m 07s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"e81b67755a28194e758d5d22c16a4eb4aea0fee2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-18 09:05:30.000000000","message":"Patch Set 1:\n\nrecheck\n\nFailed to connect to the vbmcd server on port 50891, error: Server response timed out","accounts_in_message":[],"_revision_number":1},{"id":"64d1b514a463c686930bed88293e8339ccbad09d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 10:52:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/106743b530ba44208ab947c24814584b : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d01df56a4979480c8c0b3bf7495076cd : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7555c679812c4fe18867b99aeac08c6f : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd85bcb991a349ceb4bb8b328636a873 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ef738f0d0bb4e0392f07638d4bca075 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/796e5192f8094b8088412d96aee58828 : SUCCESS in 9m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e7e103284ff47db8f3c0732297f57a4 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4748375eea244c19bfaad9048ae8d864 : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/55193a15b5ed42a690249a7cf1ebb4b9 : SUCCESS in 21m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6fc7bdab791844b7ae439119c911e943 : SUCCESS in 1h 01m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b385681187047bfb4fd56c9736785cd : SUCCESS in 45m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/aa20a5eb16ef4e429d23220c64cf6079 : SUCCESS in 1h 05m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8a34305faa704e21b4da859670196d86 : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8291bb0483314bccb4ecceae6121ca48 : SUCCESS in 13m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/867f796ee9004d34827afd3a80f839f1 : SUCCESS in 1h 46m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bdf72e40fa0e416795205e4d1fcccc29 : SUCCESS in 29m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/732255f10ac74eb493522d94e2737bc9 : SUCCESS in 30m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1883d54bcbcc4a4b8bf6f363c5db8eb2 : SUCCESS in 24m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/24bfd0c346574e7bbc79302260739176 : SUCCESS in 12m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b1d958da71d4325a4bb3c6dc5a0b801 : SUCCESS in 10m 40s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"fdcabf7185e53eaf13ef09ddb815608a3489cd5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 10:52:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7720dba1ad50750a72de90042dad122a0b7bfb31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:30:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31637861c1b9473b8e69a3001fb706d7 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6e5bcd6a2444cec9b22533ae94b7417 : SUCCESS in 8m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32585ecdd5f247e9b97ce3e571ab9b1e : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e87afc170c9442685cb29619cda9db2 : SUCCESS in 15m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c545a50b6af14440b7dfcedfdea22a97 : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81a182e5bb6d4e708a2975b9ada683b1 : SUCCESS in 7m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/96ca46ac353d4794a7772ea0c88b2356 : SUCCESS in 15m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dc2b151db8714577a95b719b01e9fa12 : SUCCESS in 46m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66ff710ec8d748d0bb46e87d0169c1f8 : SUCCESS in 45m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/885245aaded547ecb68745cee07b9365 : SUCCESS in 55m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5a2971eb2437445f85644becdea6f662 : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e628f0be736942389d043a869a52b4bc : SUCCESS in 10m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f42b9f1bc9f44ab58cc73d7695eab72c : SUCCESS in 1h 24m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1187c81e28fb47609f5b6040eacfe2ae : SUCCESS in 24m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a467ab9e647d41e298e19ff3d47b9b73 : SUCCESS in 27m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/99a7b971b8da4450812ba515c52f480d : SUCCESS in 23m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/758bd552c36543208fbcf6c1bcd620b0 : SUCCESS in 11m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/220568e2e0dd43b1a194bc55b10e4c32 : SUCCESS in 8m 23s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"8fb3202b149993bbfdd6ca30661b6b324c120c7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:30:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6126e819d70dcb39f4a39cb4948fbf1f5a249f62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:32:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/414df8c6d2014a76930a0cdc565ac43b : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5a319aa2a0309176b3b30ce762e2fee525974b8","revisions":{"e5a319aa2a0309176b3b30ce762e2fee525974b8":{"kind":"REWORK","_number":1,"created":"2021-05-11 19:16:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/790730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/790730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/790730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/790730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/790730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/790730/1"}}},"commit":{"parents":[{"commit":"7a56c95a4ed769512d43a41e1b8ff1d055a99a07","subject":"Ubuntu: add documentation \u0026 release note","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a56c95a4ed769512d43a41e1b8ff1d055a99a07"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 09:45:04.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-11 19:16:51.000000000","tz":0},"subject":"CI: Remove unused Docker images after overcloud upgrade","message":"CI: Remove unused Docker images after overcloud upgrade\n\nOvercloud upgrade jobs often fail when running on nodes with a 38G root\ndisk. The overcloud upgrade itself generally works, but post-upgrade\ntesting fails, for example when RabbitMQ gets in alarm state due to lack\nof free disk space, which prevents Nova from functioning correctly and\nmakes it return \"Unknown Error (HTTP 504)\".\n\nChange-Id: I536a526340f6d7dea7a69372904b777292774eba\n(cherry picked from commit 93a8baca93b7525c41d164c13822f60672e41bdf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5a319aa2a0309176b3b30ce762e2fee525974b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e5a319aa2a0309176b3b30ce762e2fee525974b8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
