)]}'
{"id":"openstack%2Fkayobe~790732","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I536a526340f6d7dea7a69372904b777292774eba","project":"openstack/kayobe","branch":"stable/victoria","topic":"docker-image-prune-stable/victoria","hashtags":[],"change_id":"I536a526340f6d7dea7a69372904b777292774eba","subject":"CI: Remove unused Docker images after overcloud upgrade","status":"MERGED","created":"2021-05-11 19:18:13.000000000","updated":"2021-05-12 11:38:47.000000000","submitted":"2021-05-12 11:37:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790732-docker-image-prune-stable/victoria","cherry_pick_of_change":790587,"cherry_pick_of_patch_set":2,"meta_rev_id":"ab8fce03c9d92e041fe0d0972bea3976870dc63b","_number":790732,"virtual_id_number":790732,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 11:37:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-05-12 08:28:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-11 19:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-05-12 08:28:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 19:18:13.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-05-11 21:24:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d3f46fa96555fb34739d05614f1b28972577e0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 19:18:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3821c6da563eb49f94c93d4a9632919a3842feca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 19:22:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d653fe2be38cb9da882e92c429f03e2a192928f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 21:24:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d27619fe09eb41af958b62ba03f4ad8e : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/092ae3bfc87f448d936959303febb64a : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f9e1b0b9ea04495b992213170fab74b : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f784d62b2125483da263fe15a0094c04 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c4abd64c05548579827cf0b5c07fc48 : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/881b5045c5494537bb356f9f2cf2e8c3 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd54f9fcb317487db5d6fe9f14704443 : SUCCESS in 8m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/73c7b12833184303af367984d611068b : SUCCESS in 18m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/78ca264a3d09462fa525ea339f2c5743 : SUCCESS in 1h 01m 05s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/aaaf663421cb4a81b92705b6406ec587 : SUCCESS in 1h 03m 49s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f358aadd9c5f414f964710165aaa9a31 : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9f876fe9497f4cb199e3d252725e5632 : SUCCESS in 2h 05m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/84785e3cc6814fcc995227b2be5fc28d : SUCCESS in 26m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8dc521ba7b894233875fe502a1b8a412 : SUCCESS in 24m 33s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d3fc11ddd24e4af182ff776155b15add : SUCCESS in 10m 26s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"6dfaff06e30d6232e76c56dcf4fd47281d5c89a7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-12 08:28:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4bcd0b45ce532f9d215dd7c855efc1868e44ae71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 08:29:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a67944b10e081b644d47ed611ad33ae6a9564a96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:37:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9aa2573f24a4fd9856956ec204d89d9 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0ad1ec9497445ff845b10a1e99c9226 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c7184a128b64450aa62251003e95c53 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8adf4b7aad50434abdc2f01cdd08faee : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81e1f6cc7c484b02b5e574b2b16ea746 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db44c031756648dd8c8cc515c5a3c6cc : SUCCESS in 7m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4aeeca3973549999233479a3ffd0cad : SUCCESS in 15m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/29cc051b71b24723bf779c9509dac2d4 : SUCCESS in 46m 03s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f1366acdaf7640a688a635edad7979db : SUCCESS in 48m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/48bf9659842c41068c89a3c329689d5c : SUCCESS in 13m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/048441a30d4f45c598d8de8382d7de70 : SUCCESS in 1h 43m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/19b7445e90b448bd9c8f188a26feb51d : SUCCESS in 19m 36s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5420711c10294fcaa9dbec911b4c46aa : SUCCESS in 22m 53s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a473d627cfba4a91bc89e346d7da057a : SUCCESS in 8m 14s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"5e2d3343fdff4295b448024a1fd6e977e87f1e0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:37:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ab8fce03c9d92e041fe0d0972bea3976870dc63b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:38:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6b8b7ef557c45df813cb8a2a5d9b87d : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37ab622cde3ad7b0e1b3b703f688cd24a0899ba2","revisions":{"37ab622cde3ad7b0e1b3b703f688cd24a0899ba2":{"kind":"REWORK","_number":1,"created":"2021-05-11 19:18:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/790732/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/790732/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/790732/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/790732/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/790732/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/790732/1"}}},"commit":{"parents":[{"commit":"cdcb90809a2429c6e57d80a2f93fc9b6b218590c","subject":"Sync the names and content of the repo-files for CentOS8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cdcb90809a2429c6e57d80a2f93fc9b6b218590c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 09:45:04.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-11 19:18:13.000000000","tz":0},"subject":"CI: Remove unused Docker images after overcloud upgrade","message":"CI: Remove unused Docker images after overcloud upgrade\n\nOvercloud upgrade jobs often fail when running on nodes with a 38G root\ndisk. The overcloud upgrade itself generally works, but post-upgrade\ntesting fails, for example when RabbitMQ gets in alarm state due to lack\nof free disk space, which prevents Nova from functioning correctly and\nmakes it return \"Unknown Error (HTTP 504)\".\n\nChange-Id: I536a526340f6d7dea7a69372904b777292774eba\n(cherry picked from commit 93a8baca93b7525c41d164c13822f60672e41bdf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37ab622cde3ad7b0e1b3b703f688cd24a0899ba2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/37ab622cde3ad7b0e1b3b703f688cd24a0899ba2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
