)]}'
{"id":"openstack%2Fkayobe~790733","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I536a526340f6d7dea7a69372904b777292774eba","project":"openstack/kayobe","branch":"stable/train","topic":"docker-image-prune-stable/train","hashtags":[],"change_id":"I536a526340f6d7dea7a69372904b777292774eba","subject":"CI: Remove unused Docker images after overcloud upgrade","status":"MERGED","created":"2021-05-11 19:18:18.000000000","updated":"2021-05-12 11:58:59.000000000","submitted":"2021-05-12 11:55:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790733-docker-image-prune-stable/train","cherry_pick_of_change":790587,"cherry_pick_of_patch_set":2,"meta_rev_id":"2c6a4ea99f44437f92094c5a1ed1ca03198bf9b7","_number":790733,"virtual_id_number":790733,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 11:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-05-12 08:29:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-11 19:23:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-05-12 08:29:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 19:18:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-05-11 20:47:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"104e757c6ec9d4588238d48aa6c190315292eb70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-11 19:18:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"281a77f836fac6ab7ee9a9878549e8a37b459aac","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 19:23:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"264999a929303b5a13fe1edc1dd86b4023130035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 20:47:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59fd73c74d314751bbca08e9b49aeb65 : SUCCESS in 5m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5d5247a9f63e472a8f4afb3583ffb7a1 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5fbbb0c98624cdfbb585199acebbe34 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05915963f19c444eac72ae00221ac8d7 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d702ec434404f82aeca7324368d95ed : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e9c7f34d054e28b84b5eecc0744135 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c65b0f0c27e451fbbfdd55a74adf6ee : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/670a69f7a0984c14b90baf9b1f7e3f91 : SUCCESS in 8m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1eb3fb8323d744049d92371553c802d2 : SUCCESS in 17m 20s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/38e796d3e10044fcb654a0b84b4e50cb : SUCCESS in 57m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1638a6fa8dfa4b82bd7128d092f9bae5 : SUCCESS in 58m 06s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/b9d66457cf7547a2900042158aa665c0 : SUCCESS in 13m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/61123ea6961e423ab7fe4b06b1c79886 : SUCCESS in 14m 48s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0042b67fe1264d4ebbc00bb019cb0127 : SUCCESS in 1h 27m 51s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/7508b6d539924f7ba4d7c20b2f69b029 : SUCCESS in 20m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6cdec2dd07964d138928b89dccd60a17 : SUCCESS in 23m 50s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/025a610c97e24e19b1dbe15e09fdaac5 : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e78f57f5fad45edcc704a62580562516b29e83cf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-12 08:29:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ff650f9e625d46d6f62367f8fb4ddffa8e41f311","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 08:29:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84d47b8b1833879b2b53cb54565144ab3ee582c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:55:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/85785113cdc44415a015ebfc990a67ab : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05ac3808927044278289571568f8c32d : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34f65cf76e754ef3ad6d3fbb481ff20b : SUCCESS in 4m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d39c9e1d86f48929f79e03b00f1a1b5 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54bb6bc3bc634e65b05e26f9fcb7e842 : SUCCESS in 6m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/967c98399b75420f9a9531983b5bad70 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4e66bbd9b9a43e0ba44639a45f390d8 : SUCCESS in 8m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d98e43fe78e6407d8a6678420f5653cf : SUCCESS in 15m 12s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/ee2e599f6fb7426aa00e070ab46332aa : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/533ee38043a5471e9bfdcca9a6fd322a : SUCCESS in 42m 21s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/dabf6a32ce0c4449ae2e61e03af387a7 : SUCCESS in 15m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d126b5413d824d8394b1e7e27dd14195 : SUCCESS in 12m 38s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2d245f35e244437b993040e141e5d786 : SUCCESS in 1h 38m 00s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/56958eb6cc54470ca40c4318d4352235 : SUCCESS in 19m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f6c101efaddb4519a2e78eab77a9e96d : SUCCESS in 30m 46s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0c94ac774b1343a89de803bee1b1e3ba : SUCCESS in 23m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e96b1b9e815492adabc5eda25e8c42ca6c0088e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:55:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c6a4ea99f44437f92094c5a1ed1ca03198bf9b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 11:58:59.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f3d16b9189c48babf73488ce5b09e7d : POST_FAILURE in 2m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3f95e6c54d87d10e668f6e9c239f707add00a93a","revisions":{"3f95e6c54d87d10e668f6e9c239f707add00a93a":{"kind":"REWORK","_number":1,"created":"2021-05-11 19:18:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/790733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/790733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/790733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/790733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/790733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/790733/1"}}},"commit":{"parents":[{"commit":"17c0e213105c653aa269de9e193006cc7c09108c","subject":"Merge \"CI: switch to quay.io for container images\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17c0e213105c653aa269de9e193006cc7c09108c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 09:45:04.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-11 19:18:18.000000000","tz":0},"subject":"CI: Remove unused Docker images after overcloud upgrade","message":"CI: Remove unused Docker images after overcloud upgrade\n\nOvercloud upgrade jobs often fail when running on nodes with a 38G root\ndisk. The overcloud upgrade itself generally works, but post-upgrade\ntesting fails, for example when RabbitMQ gets in alarm state due to lack\nof free disk space, which prevents Nova from functioning correctly and\nmakes it return \"Unknown Error (HTTP 504)\".\n\nChange-Id: I536a526340f6d7dea7a69372904b777292774eba\n(cherry picked from commit 93a8baca93b7525c41d164c13822f60672e41bdf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f95e6c54d87d10e668f6e9c239f707add00a93a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f95e6c54d87d10e668f6e9c239f707add00a93a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
