)]}'
{"id":"openstack%2Fkayobe~790819","triplet_id":"openstack%2Fkayobe~master~I3c15c6976296abf9c78f1480c1779b854145c7ca","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I3c15c6976296abf9c78f1480c1779b854145c7ca","subject":"Remove default time sources from customisable configuration","status":"MERGED","created":"2021-05-11 20:31:57.000000000","updated":"2021-05-18 02:17:30.000000000","submitted":"2021-05-18 02:16:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"790819","meta_rev_id":"19861bf5230784c6a3e4889ef4fdbe0f19997d47","_number":790819,"virtual_id_number":790819,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-18 02:16:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-05-17 09:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-17 10:00:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-17 10:00:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-11 22:18:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 08:37:17.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-05-12 14:02:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2021-05-13 15:36:01.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-05-17 09:16:58.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"e23a27166fae56627758baf56cda5f77a08b0d0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-11 20:31:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"87e54ffb2562affe0302a91487d1e51a3d4f22b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 22:18:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70cd0f4d95554df0baf431354be1366b : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e97b5cb8542b4b208de5c8529420118f : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/075866f4ea3341db96b4cc8e241fa56e : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9bd5091a35c40af938073c66b02552b : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d10dee99a88f4b84bb533adb7e00c7ca : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/630bec670f8745699bb95b200a7d8b0c : SUCCESS in 8m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3bd657e72be34a70a4bf94191aaf4362 : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4cf4e7a6e91438685fad5c7a9e3f583 : SUCCESS in 10m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c7f7cc6333ea4f1a966972fbd3896cc3 : SUCCESS in 19m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/99b52b8a9f2d4c2293053682a9a25e32 : SUCCESS in 1h 01m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42d4e86fafac4ab09b21fd167eb56524 : SUCCESS in 45m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7d4e1315950c4456b30060bf353c7cae : SUCCESS in 1h 04m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/70ec2216cdd746b5b6bd50de53c0c902 : SUCCESS in 17m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82aa82ebb463471db9bf619bf8dec223 : SUCCESS in 11m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f6e8fb964fe34f4fa509f4fc668a8695 : SUCCESS in 1h 45m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/51d26fd16b834c8387433429f02640bc : SUCCESS in 29m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ba37f67b33f499f8a47d9d344f8c1be : SUCCESS in 43m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/387e43dee9974edd84242f01e329ab0f : SUCCESS in 24m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0bcd7cf2b5e041d9bc46fa63090c0d52 : SUCCESS in 11m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d17c6390eecb46f18f1a5a9868b5f35f : SUCCESS in 11m 04s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"26ca2502fb0b789b7c0f6dad435ea7af2c793ef9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-12 08:37:17.000000000","message":"Patch Set 1:\n\nWhat\u0027s the reasoning? I think we do normally include the defaults in etc/kayobe/. Should we omit them?","accounts_in_message":[],"_revision_number":1},{"id":"6b7923c6ea0adb2ab6d1d0b69fd515ea77ee9bae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-12 08:43:45.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e What\u0027s the reasoning? I think we do normally include the defaults in etc/kayobe/. Should we omit them?\n\nI thought this one was an internal variable that had been exposed by mistake. I think we have exposed variables named \"default\" when they are combined with extras ones. In this instance, why not set them directly in chrony_ntp_servers, like we do for dns servers?","accounts_in_message":[],"_revision_number":1},{"id":"aa7d6756b2be7fce9d9770e6281d2f5174aa4f62","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-12 08:49:59.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e What\u0027s the reasoning? I think we do normally include the defaults in etc/kayobe/. Should we omit them?\n\u003e \n\u003e I thought this one was an internal variable that had been exposed by mistake. I think we have exposed variables named \"default\" when they are combined with extras ones. In this instance, why not set them directly in chrony_ntp_servers, like we do for dns servers?\n\nI think the idea of having a separate default is that you can extend it:\n\n chrony_ntp_servers: \"{{ chrony_ntp_servers_default + [{\u0027server\u0027: \u0027foo\u0027}] }}\"\n\nWhether users would do that in practice I\u0027m not sure. If the default is not useful we could remove it, but if it is, we should probably make it visible.","accounts_in_message":[],"_revision_number":1},{"id":"af2855bf2ec09706436f8466e44a88b61a064fd9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-12 14:02:44.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e What\u0027s the reasoning? I think we do normally include the defaults in etc/kayobe/. Should we omit them?\n\u003e \u003e \n\u003e \u003e I thought this one was an internal variable that had been exposed by mistake. I think we have exposed variables named \"default\" when they are combined with extras ones. In this instance, why not set them directly in chrony_ntp_servers, like we do for dns servers?\n\u003e \n\u003e I think the idea of having a separate default is that you can extend it:\n\u003e \n\u003e  chrony_ntp_servers: \"{{ chrony_ntp_servers_default + [{\u0027server\u0027: \u0027foo\u0027}] }}\"\n\u003e \n\u003e Whether users would do that in practice I\u0027m not sure. If the default is not useful we could remove it, but if it is, we should probably make it visible.\n\nI am not sure how useful it is to combine defaults with local customisations, but if we want it I would suggest using our usual default + extra pattern. The reason I thought this was included by mistake is that the default variable is poorly documented, it doesn\u0027t say what servers it uses or that chrony_ntp_servers is set to it.\n\nWill, what do you think?","accounts_in_message":[],"_revision_number":1},{"id":"61305cc70674c8c4919241154d1a314670a38d40","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-12 14:35:28.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \u003e \n\u003e \u003e \u003e \u003e What\u0027s the reasoning? I think we do normally include the defaults in etc/kayobe/. Should we omit them?\n\u003e \u003e \u003e \n\u003e \u003e \u003e I thought this one was an internal variable that had been exposed by mistake. I think we have exposed variables named \"default\" when they are combined with extras ones. In this instance, why not set them directly in chrony_ntp_servers, like we do for dns servers?\n\u003e \u003e \n\u003e \u003e I think the idea of having a separate default is that you can extend it:\n\u003e \u003e \n\u003e \u003e  chrony_ntp_servers: \"{{ chrony_ntp_servers_default + [{\u0027server\u0027: \u0027foo\u0027}] }}\"\n\u003e \u003e \n\u003e \u003e Whether users would do that in practice I\u0027m not sure. If the default is not useful we could remove it, but if it is, we should probably make it visible.\n\u003e \n\u003e I am not sure how useful it is to combine defaults with local customisations, but if we want it I would suggest using our usual default + extra pattern. The reason I thought this was included by mistake is that the default variable is poorly documented, it doesn\u0027t say what servers it uses or that chrony_ntp_servers is set to it.\n\u003e \n\u003e Will, what do you think?\n\nThe original idea was so that you could extend it. You are right though... I should have added the _extra variable as well to be consistent with our established pattern. It does seem like something you\u0027d rarely override in practice, so not opposed to removing it.","accounts_in_message":[],"_revision_number":1},{"id":"96df1516613c3db32c1271dbe57f94cfda52960e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-13 15:36:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"05a93316fe722d8e4db3d95637e63526664f4a38","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-13 15:41:16.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nShould we get rid of `chrony_ntp_servers_default` from group_vars too?","accounts_in_message":[],"_revision_number":2},{"id":"42249fcd0a35d4f273a221e46211520d5cda961a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 17:27:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14966858121d4509abd25dfa5786e0c2 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c5236816e14c54aabfbd1d70ce02e7 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/932be749e9a849fc9296a5fb1f0d3fc5 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f11381b0b8764d8ab11771943a1fc813 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0481eab9784a4868a1c548644a9af531 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6ae94672def43ac9abd5df64606f9b8 : SUCCESS in 9m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a046229aac54b7ea8be457715ac5559 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/647c5f22d1a4456caf4ca5974860a39a : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4393eabfd1f04a4e8c889b0fcfad9cf7 : SUCCESS in 19m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fedc36b9069947ff9270a8dd2afae5d9 : SUCCESS in 1h 04m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d88d3d42458486998cf3de4a48ae93f : SUCCESS in 46m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/06e9b1d9bb95454893346fe8e2b5f90f : SUCCESS in 1h 06m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2c5349c7868a4dfe9802c46bd672faea : SUCCESS in 18m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6b60b7f0bfa4a178de770c2e819f44e : SUCCESS in 14m 19s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/42a5cfb4f0364fc38511c59660de269e : SUCCESS in 1h 49m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fadf0a8e2b5144d29c27bb25cd70386f : SUCCESS in 32m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60a47f899bc74a69ab85eaf1a3851317 : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dc60dd7e8d0f4797a8b5a9e5f8d9428f : SUCCESS in 28m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c53585a558da41b68bd015b47a20cc14 : SUCCESS in 13m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa90ddf5afef4254a57cae00b8a72195 : SUCCESS in 10m 17s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"94a7d019ed088fc1542f8acb3a18f183341324bc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-17 09:16:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2e02a32d961826d3e9cce690da233ec7fad4a10b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-17 10:00:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks Will.","accounts_in_message":[],"_revision_number":2},{"id":"b8751ef7541e18e48c7421230583d992c787d5af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 10:00:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e6917cd85c85b98a0dce69621cd0cf4fd386477","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 11:31:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f54da96f75d45ffbf4dbc2be5202a31 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/207f657872e64af3b8f113c09a5c3467 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0399281723994ea2a25b62f653f3116b : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1994c7c23ab4b90ac957dd26df5efb8 : SUCCESS in 8m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f228f487df64cf1a1fd6d5edcd61acb : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b25609b9be644824932cf363e5aad5e7 : FAILURE in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d644e7e7f8441f3bd26f92262be3e1b : SUCCESS in 19m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fd3cda2ebbf34ceab5afd550d5115351 : SUCCESS in 1h 12m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f94adbe5c1f14832acc52d8b3013e28e : FAILURE in 40m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bdc3c93a81ea4d809e0451e86a559493 : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2b1ccd22ce08418792a64f0a26aa05b3 : SUCCESS in 18m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ac83e4ae226423aba8f8fbcc384ff4d : SUCCESS in 11m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b10a2099168448fe97c19a84472fe54e : SUCCESS in 1h 28m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4de49135487e4f40bc01edf3a93e6674 : SUCCESS in 31m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d2dd1c0db664cda871128ff95c2c52e : SUCCESS in 27m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5a4a9fb9f9c84de3b03dc6baf5d54495 : SUCCESS in 26m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/613651892ee44533981bae0ba842f6e5 : SUCCESS in 13m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f7ab4e72ffd42aeaba24f5f1283cfb3 : SUCCESS in 11m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":2},{"id":"1111acae6a93ac0278c5b7e8de94134c8629a147","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-17 21:01:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a3ecb6b40a1df48eb826537a45570ad9e0bee2fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 23:59:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c64c71c055ca47589f53afeb49358631 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f036d862c8444e06bd6f568137a49e3e : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d075e620c2e449379f5783200bf5cc44 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59376d9c16fa43fd821916ea0a3f61e6 : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b895695abba540d0b3a494f1c1d2ce92 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9a5eb6457db4c3ea8b29b91d6755ca5 : SUCCESS in 9m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ece44ce3924b4931a427b8b1535f9f31 : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/feeea88fcea34144bc2405289e3a572b : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8797a6745162489da9479c9edd8f4931 : SUCCESS in 20m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8982a3e2f3f54085b3c7a5ea508d0ad7 : SUCCESS in 1h 00m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87a66410096d4987ae79ec3f018eda86 : SUCCESS in 45m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a2b2e3abfb834b2d9e64ae204c772312 : SUCCESS in 1h 02m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1d556303db6645ef999e4b93439a6a7e : SUCCESS in 17m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/244cc5a5455c44f6870ad9e17e422dc7 : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c6e57bad278420ea0041beaebac27a6 : SUCCESS in 1h 44m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4e51219b78d94c07981ceab90ff88e9f : SUCCESS in 29m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ae22ae6d0f64119b40607155185203b : SUCCESS in 24m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ddb5dd4b8cd4444cb48a58afa81f1b7d : SUCCESS in 25m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4a20709325cb47b9a05e2162cadcb7c8 : SUCCESS in 13m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8b972b784a04ccfb236227e2fcfe791 : SUCCESS in 11m 20s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":2},{"id":"8318b57ef66d81bd96e3783a56ffe26d4ecac772","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 23:59:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b43cd17bb8a417619f481d606ee54d3413517bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 02:16:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f74b7b552c034162b60b44738d50f666 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7251a79e9ff4fd58c9b425111c5b4da : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a8c261d6aab4aa2856b2a1f54683332 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce29703bd5314ae0b5585c9dd0539922 : SUCCESS in 8m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d4d0c9e77114144ae273bc3ceb0f607 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8214d994dd194d7d858ae2d3202bb386 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/808620186f8b463c9904de617d018496 : SUCCESS in 21m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ef3a409586624a9fac872206c03edc1a : SUCCESS in 1h 02m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc8a8de629c045b3a359b21eb0abc21b : SUCCESS in 45m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cfebf89e25d0464a95501b7bc8b3de06 : SUCCESS in 45m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/31b347cd50da4345a14e77eacc98eb4e : SUCCESS in 14m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fbc5304ee424c51ae3478046cb622b4 : SUCCESS in 13m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b3729678d3cd4044934eaeb8d04352db : SUCCESS in 1h 58m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e6ef999b3b1944c2b35486112db79ba0 : SUCCESS in 32m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a1f05cce87344db8b8005553e0727ee : SUCCESS in 33m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dba9b935073c4426b281ed801aa1b9ae : SUCCESS in 24m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9801fb23e05047ffbb13827cca2df2de : SUCCESS in 10m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e073447647674aa4a4cc3242e109ffe3 : SUCCESS in 8m 13s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":2},{"id":"07985f9bf0417b8f423cc623374ee6e87224c578","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 02:16:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"19861bf5230784c6a3e4889ef4fdbe0f19997d47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 02:17:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ca1607c53c4ca5bdfb62585390f847 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"223e310bc2338bb5f86c3a3de31f69361eba2f73","revisions":{"f14130a613368a67d7f3afaf6535b52333d09a0d":{"kind":"REWORK","_number":1,"created":"2021-05-11 20:31:57.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/19/790819/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/790819/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/790819/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/790819/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/790819/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/790819/1"}}},"commit":{"parents":[{"commit":"4daf293e7824a9444906fd3510c2def0a736801a","subject":"Merge \"CI: Remove unused Docker images after overcloud upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daf293e7824a9444906fd3510c2def0a736801a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 20:29:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 20:29:48.000000000","tz":120},"subject":"Remove default time sources from customisable configuration","message":"Remove default time sources from customisable configuration\n\nChange-Id: I3c15c6976296abf9c78f1480c1779b854145c7ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f14130a613368a67d7f3afaf6535b52333d09a0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f14130a613368a67d7f3afaf6535b52333d09a0d"}]},"branch":"refs/heads/master"},"223e310bc2338bb5f86c3a3de31f69361eba2f73":{"kind":"REWORK","_number":2,"created":"2021-05-13 15:36:01.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/19/790819/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/790819/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/790819/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/790819/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/790819/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/790819/2"}}},"commit":{"parents":[{"commit":"4daf293e7824a9444906fd3510c2def0a736801a","subject":"Merge \"CI: Remove unused Docker images after overcloud upgrade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4daf293e7824a9444906fd3510c2def0a736801a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 20:29:48.000000000","tz":120},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-13 15:35:57.000000000","tz":60},"subject":"Remove default time sources from customisable configuration","message":"Remove default time sources from customisable configuration\n\nIt seems unlikely that anyone would want to extend the time servers\nvariable, so let\u0027s keep it simple and not add an additional variable\nwhere one isn\u0027t necessary. This is also consistent with how we configure\nDNS.\n\nThese variables haven\u0027t made it into an official release yet, so now\nis the time to clean up.\n\nChange-Id: I3c15c6976296abf9c78f1480c1779b854145c7ca\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/223e310bc2338bb5f86c3a3de31f69361eba2f73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/223e310bc2338bb5f86c3a3de31f69361eba2f73"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
