)]}'
{"id":"openstack%2Fkayobe~791099","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I6a801e8bb5171bebd2b3f20d6140329d38ea36a5","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"I6a801e8bb5171bebd2b3f20d6140329d38ea36a5","subject":"Fixes for Ansible 2.10","status":"MERGED","created":"2021-05-13 11:48:36.000000000","updated":"2021-05-18 12:31:22.000000000","submitted":"2021-05-18 12:29:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791099","cherry_pick_of_change":790394,"cherry_pick_of_patch_set":3,"meta_rev_id":"478ee2ff37a5a4f3c48911b0e256c34044d2c13a","_number":791099,"virtual_id_number":791099,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-18 12:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-05-13 13:51:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-18 06:32:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-18 06:32:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 13:12:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 06:32:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"a40c8428660f00171fa024692c6a56a73bcb70f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-13 11:48:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f4a7d8b5ee1c7f716aa8d014692c3b8c6059b76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 13:12:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ebb838daf604d5d81cb575347c3ef00 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e983e12cfe74da99223f8e7151735f3 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3edb2bb373540aa8ca6609f3107b769 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8cc30f69de94730838e1078c3701d77 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5637d0e265b8459da9b8eda7155f95f0 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef7c8e07a7c44d98a7eb9b463d8ada3 : SUCCESS in 9m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ed038e9dc92a40e496bff9d7c2574d7f : SUCCESS in 6m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5db2c065375149caa5e30066e839588b : SUCCESS in 9m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/feda79317234470a9eed5d89cf920991 : SUCCESS in 20m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/558cca781e8946c39cbe97d2f9750568 : SUCCESS in 43m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c168fa37d6444322a68130eba8951793 : SUCCESS in 34m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8f174c2d27af4844bc55686ee8c16038 : SUCCESS in 45m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7a7910331d164f12ad4d59fe852d8c18 : SUCCESS in 13m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cf8a825af8e45b4b4c048172c3664cf : SUCCESS in 9m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/44e506525e67499a8870a5b083b15f24 : SUCCESS in 1h 19m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9b3408cffcfb4a699ed337ebf84edb3a : SUCCESS in 23m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fd59f3710094363824d3bb264f494de : SUCCESS in 19m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f72505d790784f42bf18da6051ad795f : SUCCESS in 19m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b063f4662fe248e08a9754ac593cf700 : SUCCESS in 9m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/955610ccb44d457cb60fc83dfb9f9871 : SUCCESS in 8m 26s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"9e550954014ca0ad1e7520d40576fba41e094cdb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-13 13:51:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5c0ac6530bdf3633b9a99410f678c6c8913bd60d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-18 06:32:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"63b17696e93582de10dfb074a494136fd93a1252","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 06:33:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca69a62f1f479fb3262d4229c4b00e5c60e7e34b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 07:48:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71b4ab46e2a947cb9d84a97a2304bb70 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f75de442f284032825eb2b43651f56f : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c080771822d43e5aeaf0b5b7b5a0090 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ccc810d985346b88f4d98b09b087934 : SUCCESS in 9m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d7313d34f24e4760bea4dc55ad0c7b16 : SUCCESS in 8m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c20ce49c6c7f4da2b482a1dd14ed586b : SUCCESS in 12m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/da9e5bbf1ea54913aad973594e1b9e39 : SUCCESS in 20m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2e01c2ba93d44da0899d43aed2d2ebad : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04d678718763400fb93765b5e42c1200 : SUCCESS in 42m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b0cb6889222147cd9636d42afc276987 : SUCCESS in 1h 09m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/222bbb5f2d3d4175bfcbe03e88eeea69 : SUCCESS in 17m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea816e18068046388225b5ba42e32b73 : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f2a93d0dae6d4d87813c82c4f74e805b : FAILURE in 1h 11m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dd917430d1d242c0a8bd9ec1fd84ff5a : SUCCESS in 29m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f516491102a84a0d85a9c9be9e7fdc71 : SUCCESS in 23m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dc0261f3cd6c4ecabd752f64a7519410 : SUCCESS in 26m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/011e601d5db241b6bc7f3d76334a1d7a : SUCCESS in 14m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/028f153837164ba9a3380fd5caeffd08 : SUCCESS in 11m 31s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"d0f807cbb6116e37bcd3ee65d30215bf397224cb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-18 09:10:19.000000000","message":"Patch Set 1:\n\nrecheck\n\nNo space left on device during upgrade","accounts_in_message":[],"_revision_number":1},{"id":"80b89ff6243bbced28fc3932530cc34e3796dc75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 10:41:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a5e5c6ff1614f43ae0c9fb529b34da7 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c71a5efc4b43ef81f53965b6c4409c : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e205468f6e74f13b9f74a0b524a8a5e : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff8d7084badd4ef9bf7be109e54d4b9a : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df4862fee6b94ba09f00167fc35c2a16 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35e384c658c844e58f8dceeb153d00c7 : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/edfe3f6f6591435c941128b94495d2f3 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a090e855b24d4301b549972afe1945fc : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b4948af23fba47ecb798e159aa8efcd0 : SUCCESS in 17m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/79efb04da5234afa9998e738a54f545d : SUCCESS in 52m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/685ab068bc0f4a9baaa9bf006366fe6f : SUCCESS in 42m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/51aca95815f64dd7b459c7f834d76678 : SUCCESS in 1h 29m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/89ceb745384d4ff2a532f0916dff0140 : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f4970c86f4f4c69ad0ac62578cd4b23 : SUCCESS in 12m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/35f00c36284a4ec0a008b7db86006e72 : SUCCESS in 1h 27m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8edb7f14a5274d038cf901622fb1cfc4 : SUCCESS in 31m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a03f5a0dc4b74b4f88671630914fa1d3 : SUCCESS in 26m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5df4ed0c420c44998ffde54c18bd13c3 : SUCCESS in 25m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/47af6753f23041a0bb0f86d61f24b1c1 : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2954778628064a65b38bd1b005f9e52a : SUCCESS in 10m 36s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"7f070720e04b9775ff6dce7d47bb21564c8990fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 10:41:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2cda71705dfd6482c0ad0f3ded0ecc718beb315","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:29:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d677d55948c40eca78c30c638003613 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d618fa2146ea4905b65754c5abb88b9d : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88f27b1618cb4852b1dcf6e3f14ab08f : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad55074a8eb6461186fca242b74a6923 : SUCCESS in 8m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8c23d9ff2a642cfb7eb325b5def8a47 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1dde36d12cc4b2d8a095dd1b17fa361 : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8f5f00b84aaa480398386b192c6e2315 : SUCCESS in 19m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/811f0f9e1b424557b5d89a251a8c4246 : SUCCESS in 59m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93d97fed7ecb4f92b1f3f5ba3f8f8d7c : SUCCESS in 47m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4f5281b4136c4b198dcc54a8e3a4f518 : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/575b97193ead4feca9aaf240aa24f06b : SUCCESS in 16m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54d4c4e0d09c4868a1f02711aab646a3 : SUCCESS in 13m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2b0e96c0f2f0497ca740838f9ac8dc89 : SUCCESS in 1h 44m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/071dfa55a37d421faa8c240c630266bb : SUCCESS in 29m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6bb6ec0d6a241a890b8dedca8fe1ca9 : SUCCESS in 23m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f10354ae50da4538841e0df0af0df27c : SUCCESS in 24m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c204325442604fb7a914f0f36e4546a8 : SUCCESS in 13m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a019877db5e4452fb3879877c8a71bbb : SUCCESS in 9m 49s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"2aa45f2da9ded7b23d19d2d8c83c3b5eeb5d2d30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:29:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"478ee2ff37a5a4f3c48911b0e256c34044d2c13a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 12:31:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d85db0b1f0064dd6be0398b74fe3df35 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73dfeb2960ae502e4f60b50ac7b5aca7cdf2d164","revisions":{"73dfeb2960ae502e4f60b50ac7b5aca7cdf2d164":{"kind":"REWORK","_number":1,"created":"2021-05-13 11:48:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/791099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/791099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/791099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/791099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/791099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/791099/1"}}},"commit":{"parents":[{"commit":"010a18334f08d173c9b8185f739324f58a37da89","subject":"Adds support for configuring chrony","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/010a18334f08d173c9b8185f739324f58a37da89"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-10 09:53:48.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 11:48:36.000000000","tz":0},"subject":"Fixes for Ansible 2.10","message":"Fixes for Ansible 2.10\n\nos_image_facts renamed to os_image_info, and no longer sets a fact.\nInstead we use the registered result. The same is true for\nos_networks_facts and os_networks_info.\n\nWe did not catch the networks issue in CI because the code only runs\nwith a VLAN provisioning network, and we use a flat network.\n\nChange-Id: I6a801e8bb5171bebd2b3f20d6140329d38ea36a5\n(cherry picked from commit c44f4815cf22955a2b68937b03b50c81f81fade6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73dfeb2960ae502e4f60b50ac7b5aca7cdf2d164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73dfeb2960ae502e4f60b50ac7b5aca7cdf2d164"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
