)]}'
{"id":"openstack%2Fkayobe~791304","triplet_id":"openstack%2Fkayobe~master~I14db53ed6e57d37bbd28dd5819e432e3fe6628b2","project":"openstack/kayobe","branch":"master","topic":"ansible-facts","hashtags":[],"change_id":"I14db53ed6e57d37bbd28dd5819e432e3fe6628b2","subject":"Use ansible_facts to reference facts","status":"MERGED","created":"2021-05-13 18:43:03.000000000","updated":"2021-08-31 14:49:57.000000000","submitted":"2021-08-31 14:48:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"791304-ansible-facts","meta_rev_id":"074cc4bfcceda76e978d58e9ead7387aaa36d0c1","_number":791304,"virtual_id_number":791304,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-31 14:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-08-26 12:44:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-21 20:36:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-08-26 12:44:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 19:20:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 14:03:53.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-05-21 14:06:22.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-08-26 12:44:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"784ae10724cd5b66c5e1cce30b4659bbbca2c492","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-13 18:43:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afa7d3e1c9efa6a59797af13594659747ad585f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 19:20:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38b41d953b42434fbbe91aba00ac46b0 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/516d2dc13f8f49b5877501f8f8cefcf3 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d990f4d54ee4e878fd6a46b2881d738 : SUCCESS in 8m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8d372763c244d5bb4d46dac17060ece : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ddeae41041345b7ab2ad914268c98db : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49f5915c2f1c43a483146e61ebd7ea6c : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/989bff6667fe42b1ac7110dcbbb9ab54 : SUCCESS in 6m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/26ca1b2742a9431e91f2a529b14ef41b : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/256bbb8255fb45819a0d2b89a9ad7d04 : SUCCESS in 12m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dad6de25d57c45649535e465a1cbd3ae : SUCCESS in 18m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/95e350ec21ad4a119853fcc27474b887 : RETRY_LIMIT in 3m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8876a1ef0caf4d0aae08494251263e62 : RETRY_LIMIT in 2m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/16556182160843568e9acdcd01d87578 : RETRY_LIMIT in 2m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/47251616f52c4b0babdb8ccc92fbfb19 : RETRY_LIMIT in 2m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e3e23a097514a3eab86ee28bb98764c : RETRY_LIMIT in 3m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4d7c295260cf4e1e8d5ad47d333e4bc3 : RETRY_LIMIT in 2m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3f67e6c9ce70401ebc1b3275aca8be23 : RETRY_LIMIT in 3m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1a37d19c61e48469a11b7ed749cc103 : RETRY_LIMIT in 2m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ebb5993eddad4c97a9ed45d0165a40fc : RETRY_LIMIT in 3m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1460cfae3df0480396964c26c87bcc2b : RETRY_LIMIT in 3m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a3a5048f61943f08c193e83b3b2a9a6 : RETRY_LIMIT in 2m 56s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"3168aa0a8294f38dc8129736b5eabe78406a85cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-13 19:32:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2a6aa7d0d482bf787de761928bd4a9808b3d446d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 19:52:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/527e4747017649919baefa2e1d2a97c0 : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04558f1354744a85bbc884034b7e578b : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58ea41e8a15a421f95a9fe1600879341 : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/292806d3d5f74365915bccc11b848192 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef51a638aab3466ebadcc30e16a6e207 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170238f66ba848c3ad6fde4af30deb60 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82c08c36b86a46b098b3d11fa4ce2c4d : SUCCESS in 5m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/641e94bf00a0408584dcc8c503fae05f : SUCCESS in 7m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f0d3dfb4b035454184ea5b7960852276 : SUCCESS in 9m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5645d3c69b0a445b9794bcad611f9528 : SUCCESS in 17m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/46f4cf59a963401a9368cd74c3bfd680 : FAILURE in 9m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1498afc39034f25bad28635d5ccdd41 : FAILURE in 8m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6ca3aae06ac94495921f836ac7dd69c6 : FAILURE in 10m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/51be7e7040c74fdf862b865e1c643465 : FAILURE in 11m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43da4edecec04447bb4665962486e797 : FAILURE in 8m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/45422dab38d346a3a8ae545c648524f1 : FAILURE in 6m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bf281efe07ee4c08b62091f9ea205626 : FAILURE in 10m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fd4eb94271e470084d284b49ed70bca : FAILURE in 7m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6a89cd5d3d8547c6b771543c007b7a94 : FAILURE in 6m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7b57d1938c5b4261a3cc3396cf73dd04 : FAILURE in 8m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/245011c1425d41b784cd4d3168e4fe22 : FAILURE in 7m 24s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"7c5e15d5b7c9b116ae9ba12ccf194d0c39ffefe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-14 08:07:05.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"0c5aaf298eca7fbba0215ab25bef1011bcf04827","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-14 08:09:46.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"6b6935697ea2487b94b800011a146ce0d6e39754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 09:00:55.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4c26acf715d4771910c4a5611a4ec4e : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff50ec5904ab4fcf891a72b08b49a926 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a78c4b7e256243998fe3cc4278ac832a : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c03f1542a601439a9e203f61b0860818 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43130a72e9a844e2b9bf3223ab855a96 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41019e2fc12d4bdd8bca70864bdd5131 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d1155ee07cf4fafba358a22681f6909 : SUCCESS in 5m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b8cfae6ff9b497f839f2c9af10d6881 : SUCCESS in 7m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/99cbb024c6b54efe922f911467b53190 : SUCCESS in 9m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bba702945abe4e56b4c87f481612d626 : SUCCESS in 15m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/24d8f427e4384a559deb08c59ba5705e : FAILURE in 10m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8e22d7157d34bf2872db71a6e7736b9 : FAILURE in 6m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/56e483e2f42d45af8aeca040d905c885 : FAILURE in 10m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8cd8771a283947c79c4b719452fbaa94 : FAILURE in 8m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58a2503cddb54615a016dea6df516694 : FAILURE in 8m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cd6c7ac5a0d24d72b46cdde251a55e92 : FAILURE in 49m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/79999dd742b34d59bee67f78ee3ea95e : FAILURE in 7m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae3136dd4ca0409b83c4616d77cee9e1 : FAILURE in 8m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/db4b5d9f6cba47469318688bafc017e7 : FAILURE in 16m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/530a44b54ca048e4a0311cfbab6999cb : FAILURE in 7m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd81b16151a84a1e9fe1dc6f57eaf267 : FAILURE in 7m 58s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":4},{"id":"781ecb1d2649f5996255c28707598e9a2a8a4103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-14 11:40:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"22062baac800045adc90b79133f9b9a0f500f366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 12:03:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b9ca71ea61a4a97ae007501b6c30a48 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f70b02c86c9432893361199dd1df6b8 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c72316bbf3f4f7abaad9a1c83754b3b : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2120f8ebb7784d88b33a566d53911279 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7bc7f4da05d4976b1386b4d04002d2e : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea77a6d882794a8cb196965040a2b8f9 : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d981961419c4c65836043eff486b176 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c96301aff2a747cfb04fe6f5322aa601 : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d12b9ba5a8864e1bb1f5a12f3b80e082 : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1cf3ad1553ec420e9315553af3f1e8fb : SUCCESS in 18m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9936ec1e70a94a48b6064afeb1974c8a : FAILURE in 9m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46f48b4fffec44c5893f3f0e6f27bda9 : FAILURE in 8m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/284484a8bd8241c69c1b9ea1890930b9 : FAILURE in 8m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0b9fdf18f49f4758a3712d44492b4649 : FAILURE in 15m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/931bbdd2b8bd4ea5b6bcd3d588f3f0bb : FAILURE in 11m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/97f9011d3931426c81937b1b0cbf0a53 : FAILURE in 9m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de972b2b3143484c8fd21eb39cab0b39 : FAILURE in 8m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/27b4a7a1a9ac43cfafaa7fca67c0f46f : FAILURE in 19m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d54a75fdfba04901a245e45316905153 : SUCCESS in 10m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afcf8162be5c48a0a6c05c5b2b1d415c : SUCCESS in 10m 06s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":5},{"id":"727c7f7c2117984eaa1a67ea5d62e363d1cbbbed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-17 15:25:15.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"af7b67b634e8ac96dbb273e8bf78a44c6bc8db3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 15:57:56.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/018670959e024d7883e79b343ad2ec7d : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df9d00f4da9d47a9ae0019be85b73d48 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46efe1bee36947159a9c6048c38b714c : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1baefb57ba74d5388831cc7da919d5f : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4514bb3aca7945f28014d5cfd4954702 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d69205e14ee40d3b44b034ecf0740d0 : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e80ed6db4a94a6e97e0915774e6b7c4 : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/013ff5589b264d129f1f7f8b15a343e8 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2728cbe37e2545a3b0c0a5ce8be84553 : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1efc4932f684af4b13ebbbec0c07185 : SUCCESS in 20m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/326540f540e04da69ac15c399bab6063 : FAILURE in 12m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be52cc124f9e4f60be51b73d60385aeb : FAILURE in 9m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/706918d34fab4eeba82af683f0dfb0fc : FAILURE in 11m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/599181004d224e03af9d89561837aa5c : FAILURE in 13m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/786efb2b0fc74301a54365d627386565 : FAILURE in 11m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fc98c750b43c4a758355d767ff1ed815 : FAILURE in 9m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16d40208e1154f179a6ad411e2e36143 : FAILURE in 8m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f4a20fbb19d543ea97ca9eee2f595f69 : SUCCESS in 26m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/52450c3f9a1e4f42bce5737339080900 : SUCCESS in 12m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95df9419d5e04b96bfd1c6503b17336c : SUCCESS in 9m 21s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":6},{"id":"c08352393109b6a544f4f22fda134125c1c21416","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-17 16:12:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"de4aac4d00b695abdff8d06ff4b57002a6f88cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 17:38:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fe7c0c43738451f8b9294959df06bd2 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35befffefe434c2bab56b56b2a9cf553 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2ad398d4af24d2b82a560ce84fb5582 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78bacf38a91c45ac864a7802738bb707 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e36b7bd0e584dedbec37e66ef8cc51e : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf8bfb54ecc462fbff61ea6076728b9 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e73142b7724344babc0bfd349110ff45 : SUCCESS in 4m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/980a7ad1ff064f93b023a12a2f4f9c09 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1bbd29dca0924b71a6da6287dff3e200 : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/21dca955ada846b2b6ff6a6bfa99b739 : SUCCESS in 18m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b0d40f1ae6f84a3194e7278863c39210 : FAILURE in 11m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/756fba5859c6452bab94ec268133a67b : FAILURE in 9m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/78c3558aaa504ac598be20e6b54d43c9 : FAILURE in 11m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/77fdf3fcdbd44967afc3770d7b139895 : FAILURE in 15m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/006771ff1444463f89dc0f37adb4c503 : FAILURE in 10m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ca1c6a2b9c2649fd98363f5750b26500 : FAILURE in 1h 21m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/51b38a3b91864ab5b393286e40549828 : FAILURE in 13m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7039a6bc6cb9433892dd98b9d6923988 : FAILURE in 10m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/713085a233ac4d8b8b060c6503ce421e : SUCCESS in 23m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2a9711bfff2245ce92b70ff7d344fae1 : SUCCESS in 12m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93548638060d44b39790eaa1d937fc8b : SUCCESS in 10m 22s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":7},{"id":"ec4c840dc4930fa055de9479ae61513ce1730c79","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-18 09:12:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":7},{"id":"2720060ec06454d627c3c6329cb1ce58e06d0870","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-18 15:07:13.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"46884719c09f7545452fb308e3a0edb5c2b26241","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-18 15:12:41.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"421661883bfd665c7e59180cb6661daf8b68bbaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 17:07:59.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f63a699a59d24d5b912d7ecf47f80ea9 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a7b95fc04b4e469e1c2e1e77bbd1ae : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b759e9e22e604b86860b63d567d6fec7 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89d3c7b34f4c4b15b4f48819f4a03ac1 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4f22dc641344776b2879b0e58a8490d : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd1a4333ce0d4d5db918af566c5b1957 : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a87a0102fc014904a2805ca6ec73a144 : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b05abc848e1341d0b3e4cc6b94d330e5 : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/59798b42f31f40e4bde8d4b17abec8ce : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/20614e1ebc83447cabf1b79b58c18c6f : SUCCESS in 20m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/53c4422b185b4c658bf27afe562e5eb3 : SUCCESS in 57m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/553913b05a1e444a8d91178d412c7a28 : SUCCESS in 44m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b825483fd3874b829e3df1b19849e5b0 : SUCCESS in 1h 07m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1b64190aec8a4e1bbd62e6663789a988 : SUCCESS in 15m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66196caddd2342829aabc3523adf1fb8 : SUCCESS in 11m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/989d8bc67adb4c4dab92d10b57912b94 : SUCCESS in 1h 49m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/882e95cbb86340728d5fe1b2883e4cab : SUCCESS in 26m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/913390c077ff4ab2877a003463709e4a : SUCCESS in 26m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/130dacc20c8044138777c91172102e26 : SUCCESS in 22m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/072bb9b8a52c40e2836b044ff7f482c5 : SUCCESS in 9m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93e4d9823fde420d9447eba7d3fe0c98 : SUCCESS in 9m 22s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":9},{"id":"4fe8530c3808a9e365f86f18e6fb2e294fc5ff21","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-19 12:38:42.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"8a73a28ac9778495375059e1d5089838a517e151","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 12:42:41.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"a738c4d9f2c88a53620d329dcae41d194550747c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-19 15:02:03.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"18013454375dcfde4b4a017f96d08b404f324a7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 16:43:31.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f8a2b965bc0407b8223751e3ab8b0c8 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/538cb71e0875414e9066ae59edca2bcd : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/754cfdfa530a4b6aa1734c52c06a153e : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/903e087782944a92ad86f97f460adf78 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db4574beb89d465aad415e4d0ab531ac : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d76be33675d434c86ebf040a1acd3e4 : SUCCESS in 8m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfb0fdf4dd70494ba648b9decd961f7e : SUCCESS in 5m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/80f2d88fcff04b5ca9b4bb1e096d5724 : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2544d69924f64c79a5a71a1831cc0475 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/09d12ed0296e42d1ae9b7674824b6c71 : SUCCESS in 21m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/baf77ec5e4f3482d855e3d593f69b673 : SUCCESS in 47m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb71c744b4764d8381f6d34cf3098557 : SUCCESS in 42m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/116c3f7df2144e3f9f8cc8816720f01e : SUCCESS in 51m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8326153c04ab46a2948012f39a99b052 : SUCCESS in 16m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6b850adfb8e44309c46e1b49cebb522 : SUCCESS in 12m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d06236f1081244c1976227e893dfb901 : SUCCESS in 1h 39m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2b94387fdc3e4e84b922ac4eef464a70 : SUCCESS in 27m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f40366390c5c495f8f8263d786acf852 : SUCCESS in 22m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c4e4b3ed468e40c78b460b2e0a98f635 : SUCCESS in 24m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/644ce46672f54d73b5a23dc1e9bed650 : SUCCESS in 12m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bf441a0b16945d4be9d95e54b0476fb : SUCCESS in 10m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":10},{"id":"7cc71832ec52ded5e4735d7e23d15db86c2cc5a9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-21 14:03:53.000000000","message":"Patch Set 10:\n\nI only fixed a markup error in the reno.\n\nDo we want to get all of these changes merged in upstream roles before merging this?","accounts_in_message":[],"_revision_number":10},{"id":"fe0fc3963f8103130f9ff2d311da0870cbf1c049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-21 14:06:22.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"66e079557d29181a2f34da07d559b7a29e1ccdca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 15:02:22.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/860f8c6fc9204d26b54807a5b94056a0 : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/987954088b62472682873fd9aa1a2ef2 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a176aa6a2eff4d9eaca4e8be89e7b364 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af36ebcf246c47e58d5f97e7854b89a1 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4be375efe8442fca80d48e2ef99c473 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08a608b9430b44b398dff0c7b919259c : SUCCESS in 11m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a5fea8ba8b041ffab0e3c4c87fba859 : SUCCESS in 5m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/21ff88b3ebd848b48e2814636b5f3722 : FAILURE in 7m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72493054a73e41f59787255f112f12e3 : FAILURE in 8m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d836fcb0d15044408509874e49dd3955 : FAILURE in 9m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4b860c0056854c9f826d7a38ed2d7541 : FAILURE in 9m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8748d3f89cf8440ba8d6f894473adaab : FAILURE in 7m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e880a8ad4df6491fae0412bd1313a7d9 : FAILURE in 8m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e51d830b59f940d49aa4b24a5c3dfc48 : FAILURE in 8m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a089c1e75bea410190df7b70acedc846 : FAILURE in 7m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/733076292ea24305a273de539bf76b57 : FAILURE in 54m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4244bce2e886439f8109121d160d51cd : FAILURE in 7m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7eae13eaf6de4288a57c9105a89a1c4d : FAILURE in 7m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/66e5b8b720014c35a08e3fc3ebcfdccd : FAILURE in 18m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3ccfc859bd3d41f888e53097fd669b36 : FAILURE in 7m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adab703ad20f4025ab0abd8820ff839b : FAILURE in 8m 28s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":11},{"id":"14ac613a6bbf358db7a7ea1eacefd916a04434b4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-21 15:30:34.000000000","message":"Patch Set 11:\n\n\u003e Patch Set 10:\n\u003e \n\u003e I only fixed a markup error in the reno.\n\u003e \n\u003e Do we want to get all of these changes merged in upstream roles before merging this?\n\nIdeally, yes.","accounts_in_message":[],"_revision_number":11},{"id":"9ea9d4ae25639d8cdeda5d5474de6ea440cf1352","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-21 15:30:47.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":11},{"id":"f41c985cd8022aaf481aee550083833a2a59b3a7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 14:20:21.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"edfeaeb6c6199ef49f487e7dd2e3859758d059ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 15:42:21.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70241086fea140b6b91d9bc221711aeb : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd3af8c05226450580e6cdb4c743a02c : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c567011e96e7422da98551da916bc0dd : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca41c16713f74ee6b26bd1ff64ab47a6 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/713a0bc73a9c42d18457255383d64bea : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76d1dafb1be14f779811d6862c6322d8 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14f9b01c40b247c6bb1c70e5515d58eb : SUCCESS in 4m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/99a48fcff0fb445d9f72070dd50be553 : FAILURE in 8m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8609a22fbbab474284263f22e542fc2d : FAILURE in 7m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17d69c8c77e6446899344cfdb2a71f6f : FAILURE in 7m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2ffc0a6b74244feda5fb51aa093df49d : FAILURE in 7m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2481f5b8c8347bd88a3fa68edb99f49 : FAILURE in 7m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/73e0b48310984dca996913a481f335fb : FAILURE in 7m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/dd07f65582784b7fbc9f19ff5a1063f3 : FAILURE in 10m 07s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dd3ab62820342eca3d467de472bd967 : FAILURE in 7m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7794bc73e5fa4051bad4834a4473490c : FAILURE in 1h 13m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6062b742d6694e8b905770653c38f337 : FAILURE in 8m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72bbfd24931e41c1835eeba010652094 : FAILURE in 6m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5a37f045b7b44520beac9c617e3e92b8 : FAILURE in 17m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2646c43df87f4ea3811df942b3b15c82 : FAILURE in 8m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae74087c59c34df690a5ddc5d60e8128 : FAILURE in 6m 52s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":12},{"id":"14d3479ed12b25d3f147c94780b6d34bde5ca0cf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 14:35:30.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"da17d1635729fa0b28dc132d80c390eb430a147f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 14:39:34.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"0a753a4021f380b4193d50b5030f047395104210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 16:27:04.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/662e0f237987495cace5569bbb783670 : POST_FAILURE in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd8ed49c00843588adcea3d0ffc6b06 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ff0e1f740b94d8f9a6f6387890dcd74 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fd6577f206541228966c5761607d7bf : POST_FAILURE in 8m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88ad8ffd5e1b4b6591b65bfb44684793 : POST_FAILURE in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51c214fc1fa6463b9ff2e8e93eea92e2 : POST_FAILURE in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8ce57ac675f437fb5a7004c82c1db54 : POST_FAILURE in 7m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5fd8d2aaf7444acb8facbb595f8613e2 : SUCCESS in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78b53177845e4fcba1e93444f86fe19f : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c5d0a16dc7ee4e528790f64e6fb93224 : SUCCESS in 18m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2ad7f2e03f884c60be0c3952985a268a : FAILURE in 9m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32b62aa06a8744a4bf1279aa5a6485d8 : SUCCESS in 47m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9afafbf0d1264019b01f8e5f6a286b6d : POST_FAILURE in 11m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5cf8900c404049ea90d90f1ef0c212a9 : POST_FAILURE in 12m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2fa005e894f43c28f9a7844a1eef22d : SUCCESS in 13m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f361e4363234455980e336bd4b0afe14 : SUCCESS in 1h 41m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b66f3d29d06e4e47ac470f10b98443f8 : FAILURE in 9m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d96e8d0e93948b0a2b5ce67836a54f2 : SUCCESS in 25m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/46775ce5a0c74b30b095007a10e9681e : SUCCESS in 27m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ab8ca548a41d42e8a52b838f4fe83a78 : POST_FAILURE in 11m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e4f7579b14d4c978bd0aa5214831fc6 : SUCCESS in 11m 40s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":14},{"id":"fd3484d88eed753878e8002969dc48e1b205718f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 13:07:47.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"b6f3724ed577489c6a6143fccd62edda91992bce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:49:32.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5c7f673cf2042d18e2947da2f547e08 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/023008e7c03f4047bdfe87d198abded1 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d65bfc1ee0dc450f84a1ffde5e5dac98 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2b5409535af45429aafd03ff85027d4 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1abc38845cf24d909a262ec3ca3ee70f : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0342cc1f34414264aa05139b675b0a10 : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0195f385adb4dfbb2402a1020af17de : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bfb7746b45aa4c34ae03385703a59bb4 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25866ec10a6c406cb3f8b83a44c2a5d6 : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/759b1e0c2332404e8a854a762d14cb84 : SUCCESS in 20m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/566a8b0b51af4ec99f628531335f2b93 : FAILURE in 43m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad511251967b480b8c2117be905a15bd : FAILURE in 37m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b7eab7c15c7c4379916a3eeef1eba528 : FAILURE in 45m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4396b6f756dc4c2ab6c6783b5191ee2a : SUCCESS in 18m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20dd24f8e56a44e7bc7b269a42ac4811 : SUCCESS in 12m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f5f5439d6dc744029a24b1abfeee5f02 : SUCCESS in 1h 40m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/29e5db3ae2a849b790157c9e35c8d8de : SUCCESS in 26m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8df2503ab5364ad4bef33a8b1e3c3f56 : SUCCESS in 22m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b2c3a08b5b404b579a9c8e3dd77fb2f6 : SUCCESS in 25m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d7e6270bd69a4ca582acf4aeb0582f47 : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de1c52646637462fa68e3fe417faa5b7 : SUCCESS in 9m 41s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":15},{"id":"7984485841c1b0b7d272c9abc0ad014bf65c7881","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 18:50:56.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"7921b33323d7f13f9a61a841c78f204d17513fdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 20:59:17.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2569b27d7c5b4feea32fd163895fabb0 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/576fd05c0aef4ac8892b3a50212dbbf1 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c08fd8a807f49bfae1eeee170121ac7 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f03bb98f38247a89dfc225a5cdb1f99 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c816e8caa4349208764d3547cf3dfda : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a3564ae8e241aabd801b4aecbe63a9 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e83e87257474c649bd45d3d00871a89 : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/258eeef9b9f04503b6368f37db5f6f5e : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dc64f1f64f164fdb980ed202b17b001a : SUCCESS in 10m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12879578c3bb4fd2885c9e9e753d6a38 : SUCCESS in 20m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d064324cffda45878b561cfa867e1f18 : SUCCESS in 56m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9afc6fdac76248e1b37f6ec94632d454 : SUCCESS in 51m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c303b69780d245d7842a2fcf32c64042 : SUCCESS in 58m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/858169e958fd4d57b0a359251ab42512 : SUCCESS in 18m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48d95606f16b469b95ea7bdbb46f90e4 : SUCCESS in 12m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eef19c0c440649668a92dfc04e876da5 : SUCCESS in 1h 50m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/da50f3843368489294e0a812b7e5999e : SUCCESS in 30m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53cd95c5400f4c6fbca11aea04949360 : SUCCESS in 27m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/026de3a87cc5409ebcf8fd32076e999c : SUCCESS in 24m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6b632c12dccd433e9749f360367a5771 : SUCCESS in 12m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/051962212f0346d3ba55a2191c8b25a9 : SUCCESS in 9m 38s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":16},{"id":"d5dd439dcedfd0a0f527d8cc4bd685d167107948","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 15:33:56.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":16},{"id":"65da8288ff6c9050fa8157e9799c9890a93a8478","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 15:41:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":16},{"id":"8733ea79b14d0f548d9ea6fd9a6d43bf470f96da","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 15:57:40.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"47d125ffa0f8428d8fde378842d826def4402f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 17:01:19.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/231ede69b0954d47899a7339cefe6f79 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6435e071af6e461b86c3224ef99a8b53 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/721bec12ee514eeba8b984e968630d0d : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27a20d33aebc4df8877d8c1f6005729b : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f475e36f5c9439d8af9568b8d434bc0 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/829ee57e06da4bd483ea8d07fea7e5c0 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30d6eaab06514f758ae124aa03bff150 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4bbe7efd7a64efbafcdd8429356c7df : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ceeae8ec6659477c8d923f7c63932e18 : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/31c82a47d2554b5f849362750726e245 : SUCCESS in 19m 04s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5464b09c81a247038fd0da3b916b0ac9 : SUCCESS in 59m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae78bf364a914524bc47651699821c3c : SUCCESS in 42m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/10c1b3b1ef584e35ac20ceb98c632051 : FAILURE in 22m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/91861dbc86bd4faa9c0c2b8e9bacd608 : SUCCESS in 17m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41c9d7167b2149be895200a0658e3de0 : SUCCESS in 14m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c9634c8ad5ce4334a62fe6495353a653 : FAILURE in 22m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8318e46abf434b32b6b8dde40bb757e0 : SUCCESS in 29m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/422a9fa60e0745fb8c2e00e87c381dd8 : SUCCESS in 26m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2ca386efa1d647649a03f8b94f0fdade : FAILURE in 16m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae76667140cf4c2d8d09af92a7499263 : SUCCESS in 11m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eff5ced6451b4caa98c5252e2ec0e941 : SUCCESS in 11m 05s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":17},{"id":"700d8f235eba70d9cc29162e260f3e227af64c91","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 08:22:21.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"0c531cc609a2078bde72ba77c3c9d1b5da0e9ebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 10:12:09.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/623a94cb4eba46cea7fb22388b39b5a7 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d1059c5ded74d81994fb6675c7b2442 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d15fb28555b44a385e58c932a4994c7 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c62d6470ce584e7ca8b812e30601cc3c : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9be002591d64a70af2ff24af2becd67 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77e63d80a4045d68021bcefe85b9449 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f69c64fd7454dd3900adb06abe35c27 : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c1111b88e2f347549c741f7c13e57eca : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6cc10822f0674624bb4aac9fda3924b0 : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa889fdb85db490f83a5bab8df760d01 : SUCCESS in 18m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ebf870b9e8954d66860f640fc6e15ca4 : SUCCESS in 58m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56494b51b1e94ac88dcc37299766c5ce : SUCCESS in 41m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c06f4f13a8f34f1d923f23e8d3ac4707 : FAILURE in 14m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e02c7bc9d2a642b6954f52af746f96fb : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff28c5d90cc44ba5bf4fb90e9d5bbfb4 : SUCCESS in 13m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/87e6241df7c94a4ebfe27e0948f8a104 : SUCCESS in 1h 40m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e7f70a2294ed4444aec0c0979c1e7b7c : SUCCESS in 34m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9adbe91c5314f1c9ed47c7d2c9be81b : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c5b49b2bed664cd09fb935cad4ae53be : SUCCESS in 24m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/85d3de3919cf4da58efe37639b8985cc : SUCCESS in 13m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7486dd6c32aa44d6969cec3628f92ed2 : SUCCESS in 11m 27s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":17},{"id":"6a4719d8bba772b07ed99c7bc0de42b09603c116","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-28 13:06:54.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"0d22df98789492923ba8ae146c879da9bd9fef5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 14:52:27.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b378312141b24351a72c7dd675d1f8e6 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4c0a38188d44d468e3b1caa9cfdcd03 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c93c8f0882e94db299c2230eb1a83402 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55dde883b2454c4884007346ec9adc82 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba1d2aedbd044c939734ac511db8555a : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23ffe7afc2f4c71a54d87a4051b102e : SUCCESS in 9m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c3541eba9de49ed97c5d73bc1c75868 : SUCCESS in 4m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/08bd8a6cb2bf41c3a95b14734a4893af : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e00116a1d20442fc903ed93902b5ce62 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a9515b8f891448739b661d0129341389 : SUCCESS in 20m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fed5fa0c6f6b4f3ca29477a2d46d3e43 : SUCCESS in 1h 03m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8d8b1fa53e34be98d5dde547304e2de : SUCCESS in 47m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1235359f0cf44781bda6bbe8e902de2d : SUCCESS in 1h 04m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3b3aeb14ccc54b59aa8c14ceb6336856 : SUCCESS in 16m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c89bdbedf484eed804ee4ae79c6a74b : SUCCESS in 15m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f9e36c126d844f5ea789bc63c9170def : SUCCESS in 1h 44m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2a3bf3aeb7cc4f1caf92577871430b98 : SUCCESS in 30m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efb8ad9bd60b4bcabfa855b2c9731027 : SUCCESS in 26m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/12ba0f90635d44d28c64a4f20954098e : SUCCESS in 25m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/193609433e574a9f9272cc886a168726 : SUCCESS in 12m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc566f408a2a49e9bccb66a76c40fee5 : SUCCESS in 11m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":17},{"id":"1f6460d9ede35ca4fbd0fa2a0454a4af7648beb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-17 15:49:22.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"8311a5e97cf7b6c476ac8a28a3745400291652eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:56:17.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0211f7bf0f048d582e96425c1460531 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f31e1b27f70347c6a60df69a31185ca8 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99b64b2faa4646988ad2b05319be4394 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9eec903d352643c2928fc80229074196 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7990b51072de4ba0b86c73bc4b2b96b2 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eeb2dfc0f23494f8c5523f9c1b56c22 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bb9dacbcf2c49e6a2a06243c88a2d12 : SUCCESS in 4m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f405bff9c0149ce9c18f86d861430f3 : FAILURE in 8m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/37ba5ee99b8f48638a76a8fd369df9af : FAILURE in 8m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f502b372fc55495993d5e1a3889a694e : FAILURE in 7m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/52b3c082ae2d435bb7028c825820da32 : FAILURE in 9m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f0502d58a7c4deaa2dc0db22edfc979 : FAILURE in 6m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/74d72c432f2e4faebbd062a4823ea9d1 : FAILURE in 9m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/26a75525916242d1ae15c98086a48edd : FAILURE in 8m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/146e4f62107f4b458f3eacfb619954e9 : FAILURE in 6m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3d58c0bf5a8d40b7a2758dd0c9fa7b2f : FAILURE in 1h 02m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/632605e0bd9741328294146d01c0a1cb : FAILURE in 8m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/597a5c46636341b3a6918077f9cc4b9b : FAILURE in 7m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4d2554767de1434185f3a8f15353a39c : FAILURE in 21m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7fffc262dbdb49e0b54e63b258ea68da : FAILURE in 8m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d85c8cd80e55426688846e3edca1aa25 : FAILURE in 6m 44s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":18},{"id":"44643c4667c2e7b293339fe37210b3db23e5c2b5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-17 21:03:37.000000000","message":"Patch Set 18:\n\nNew releases of mrlesmithjr roles are missing from Galaxy. I emailed mrlesmithjr about it.","accounts_in_message":[],"_revision_number":18},{"id":"a4a671dfb6b1c793e01652191fbacf22e670a436","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-17 21:03:41.000000000","message":"Patch Set 18: Workflow-1","accounts_in_message":[],"_revision_number":18},{"id":"11af63e5916c6979c81ffe8ae0048550b3c0fedb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-19 13:22:31.000000000","message":"Patch Set 18:\n\nrecheck\n\nmrlesmithjr roles are on Galaxy now","accounts_in_message":[],"_revision_number":18},{"id":"a04f263141f5532c57b62c8032bddf5e56584388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:30:14.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7b0302599f54c5f873a1a2a576b6dcc : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f0b41953d940649b7f4e04bb7fbaa0 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1962f76de66349a79e827a56852ad86e : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51fee4aae2de40c5a2216c383be02473 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/143e54dba70e4658a4975e407b3c6d25 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d424409eea48449f8192991ffebcda : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0007b8e9b244b6dae6aaa4abf2393eb : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c14bd4beda643f7bc7b1b6c770035d5 : SUCCESS in 7m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b2f756d630a400db993c91e4b3186b1 : SUCCESS in 11m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04caac9ea1e84efbaaff8e2ad5af0fc2 : SUCCESS in 20m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8c72f0e9a19949ff8068dd61a31d7c20 : SUCCESS in 1h 03m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffcb1559dfd143bf9d52e044c6224032 : SUCCESS in 43m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ab236bccc7b94276b943977f9f4a9868 : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c452434dc8984b44a8a73c8975200cc8 : SUCCESS in 18m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c8ac5b6a45f40d99dba5c66b7b33c38 : SUCCESS in 13m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/394983ab1f484027b5dd79084f3350ea : SUCCESS in 2h 03m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/085bf8dde92348189537e125bdf5cd55 : SUCCESS in 36m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a57da0c0fcff4065a39d1716834e3623 : SUCCESS in 26m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7b1a4c4aadaf40329798b7d6d7f1ce17 : SUCCESS in 25m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/02f5ce201f57449c9c91d991eda0d736 : SUCCESS in 13m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23c3f1fe2658490a81a72b7712b28e31 : SUCCESS in 10m 57s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":18},{"id":"8f4deb70c893b1f8ef85e165b946cf9197a29a2c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-19 15:44:23.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"5bea159c8fc66d92a4ec6444e1f2bcd512542f19","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-20 16:00:01.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"ff05f72319527cfa1ace73fbbfeb31a3d71c90a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-20 16:11:31.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"99d5b11c9b540649d7bf2d4cb9da29f1e00ba2b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-20 16:44:32.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"575fc57935a8abf622fffb402a3c7e062bad31e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 18:17:59.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae980313409f453f8d2968b01c34489e : FAILURE in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2efc5e8963406182bdab18d9b66f5f : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bef0128b2d444d88db8af94e9eefd3c : FAILURE in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21de0e9595cb4a2880a5a4873333dfd4 : FAILURE in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01903455ed60446484dc21d49c0a0a3b : FAILURE in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da4e55c7c1fa418bb2046a4c8e9e002b : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/761e332e225049949fa5f2bb3f44ad15 : SUCCESS in 5m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2000c99ee01b4429b7b9414fcf722599 : FAILURE in 8m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72f084984d004154a101158138e10103 : SUCCESS in 13m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/18f994a2b5b0439b82de9d6758b08b2d : SUCCESS in 21m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6c570353e957478a9f942a28b6cc68ad : FAILURE in 12m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6973c6bd6f604e1e893e7d76ef53cdc1 : FAILURE in 10m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0cfe5bffd6b04a3fa69602f2413a61ea : FAILURE in 12m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6eff9f6df3294787a78066f044f8d03a : FAILURE in 11m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61b904c914bc4327a857a7f4b3e90808 : FAILURE in 11m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e47c322b0f3945f1ba0e4d3fc14fa49e : FAILURE in 1h 24m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7c1e47ff00f441d2b37d193201234e4b : FAILURE in 11m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de41412131514a37add935c89c90ff35 : FAILURE in 8m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/24336fa311db4d6988a6c9e910e7c236 : SUCCESS in 27m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b9aa78f1a9324f1b8b5a0293396aeb20 : FAILURE in 11m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d100b395a3041e68e97ecb3d262beb0 : FAILURE in 9m 21s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":20},{"id":"a995fd6ca4f53e43e950e6e7470c25740db35a74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-20 20:52:53.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"29bcfc6ce0da9cfab66a71ee29f0e508184204fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 22:45:48.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/852f397ca91248ea94f67fd6a8363b52 : FAILURE in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f60362c782c14fca8e53bf11c8d8d6c0 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdd42d38e05248a6a99691357ccb5856 : FAILURE in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f12eaa4fbe1a48eeb8083ec9379cb053 : FAILURE in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2888426f311a4f8c8ae0b15992f01630 : FAILURE in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17fa03770c254fea99c906af60e38f25 : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ac9961e739846c8b275c34b6e5735b0 : SUCCESS in 5m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d8c065e29cfa40bf9d57ce7e37b69b1e : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f835b78fa4e4475a209d745c62eee48 : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b45fba4fe234d4384ca1aaa8b1d54da : SUCCESS in 19m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a7b43240447b41ce8f298d9ea61f7386 : SUCCESS in 1h 02m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d58b8b9f569b4d5384cdc489fce1ba29 : SUCCESS in 45m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/883a4aa0223f4dad86101cf61b7f6da9 : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/61874e2d4b25437b821f34f3b8a75002 : SUCCESS in 18m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8565aec297d4ef8ae56c150abde9ced : SUCCESS in 13m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0818094798cb48f5aba104c2010e3451 : SUCCESS in 1h 47m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6252f8072d794f32855174fed1c3d61a : SUCCESS in 33m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f68c94effc746b5bffd809fe1229584 : SUCCESS in 27m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a4651b03accb44f5998baf01cc4b4c95 : SUCCESS in 26m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ff9beb1e88174a4394f3e07fbe690b29 : SUCCESS in 14m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67590fcc12c4402fbaf46f1bb5237453 : SUCCESS in 11m 46s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":21},{"id":"affed44b02d7a0eed03fb9eb14ae0d7a4a2b8b4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-21 07:57:35.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"897133618fa3e540a82e5c0d59f0f4c44392f001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-21 09:47:47.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef4a754da7054b66a325e8646c3c5808 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a82a1a6a0a66418da2a1011cea473c22 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/359d6f0fb34e4faa8a2c7b948dc90379 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/434ca8380872422f98fbe24e43575448 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c0aabb5bf274eda8542b544a86ff9b0 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e95aa5d8b24164ba4c4ac64d47d3ad : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4fe22f79d8e4b0dacf621d79badedc7 : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/669a6d965d5a4272baa38c49937ed070 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/32f7a2dc5575428293f05a710fd4094e : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b3f511bc5e34d65bd02ab18782a020e : SUCCESS in 18m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bdfbf197fe234af3b53ed37e845fdf2e : SUCCESS in 1h 03m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c9d00554e6d44f1817c895b0de376e5 : SUCCESS in 44m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6aee70f6ba89401e9f6a39fbd222d75b : SUCCESS in 1h 01m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f3cdcad4d05942e38630be349bb1fb85 : SUCCESS in 16m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b2382fbce1f44e39679893b9509dcbd : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b46057a560ac4ed38702a5cd006054f2 : SUCCESS in 1h 44m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2f426a21fc914236b9ffe4b8527dc5eb : SUCCESS in 32m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dd557c6d12b439fb83f53c40c85fc74 : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c1da8903e70945859a670f268969ddf8 : SUCCESS in 32m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b17c179deaef46d19ef85aa02812cde1 : SUCCESS in 12m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46421c490917450fae5aa34e56fb6a3e : SUCCESS in 13m 07s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":22},{"id":"68a1c9d6b2202611c6ee56ed03e37125298baaf8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-21 20:36:36.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"70dbeba09078e9696c55f13e4fa528617180e59f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-26 12:44:32.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":22},{"id":"5514a5ccc326deae2ded66f5a589e7f6f58b10e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 12:45:05.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"cb8fa08dd178b99e27aaacc278cec1337718cdc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-26 16:18:00.000000000","message":"Patch Set 22: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b4da5e1e754478909e256e49b3ee1f : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/886e611fa65c48c2a8e6bf835e9352ac : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfcc56f6be57477a9d6809d35361f04d : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d8cfb537724ab1b1365afb808c3ffa : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/426acbcfe68d4d828c27fc5deaff9a85 : SUCCESS in 5m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f35772f2024415ca6c297de0c90e4ae : SUCCESS in 7m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/34f2fe99baf7449aa5f5ca46305832e4 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/71be4093ed0947819ad110fd9c5cc9cd : SUCCESS in 33m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5e9c0a343a50464f88bb8dfeabdc5cfe : SUCCESS in 1h 03m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bc48d63b3fe4bb4a386e4dd3244553a : FAILURE in 41m 45s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fce675eddc3a4670a69a72411734ccf7 : SUCCESS in 1h 02m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/140a8f455a4d4661abc35dec08a9627a : SUCCESS in 21m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c580927dcade426d9571c425784a6074 : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/598a308a5227489ca39d6a5ed9ea0e84 : SUCCESS in 1h 43m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/90c71ac2b7354fb0b690a70ed4eafc68 : SUCCESS in 33m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f3697ab51164d2db4796fae92582976 : FAILURE in 26m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7fa21b69c0f140289a7492247b045d2a : SUCCESS in 28m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/004fdffdb25946b9857325df4bac676d : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fabccc5cf12433c99809ce0721cd3f2 : SUCCESS in 11m 36s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":22},{"id":"87db801a81c6ccba945e6ac51f68c8fff913098c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-30 09:38:31.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"bc22dc25dbf5ad8cd221a389dd8124d98209c546","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 11:29:57.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a06baf392f18403d90e2f32ed707c178 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6017465c2db4489594993b1f349bce73 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2f75068fdb04c5da26682e25f6cc872 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01331b8479b140708cda84a18f1dd43c : SUCCESS in 8m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c324446d2476433888a49d5faa81631a : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c561cba3ab7f44c397fe383e22864480 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abd32b10f9e64e6f94a4fe84695a6798 : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f41ddcbecd5545cf87ed9dccca2095c8 : SUCCESS in 7m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3773e01f3ef945f88f0e12bd2ab82fde : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c3ef41c6183d42ddb424571f6509a3e0 : SUCCESS in 32m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5b041b0331b24639bc8eed78445df95c : SUCCESS in 1h 20m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b00b44e3e4524256b517cc198b466a49 : FAILURE in 41m 50s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/33de6535fde8434a96a3fecae9c4df85 : SUCCESS in 58m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/507f4b59baea4d9c8d2ee3c4cbdfc9b3 : SUCCESS in 18m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51bb1403e1214293947bc83bb798d584 : SUCCESS in 13m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a13c24145cbb4d4dbffce5f7f5b569b9 : SUCCESS in 1h 44m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8aa5fcd03aa44244920ea931cd38d3c8 : SUCCESS in 33m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba076ae4df1d43f798cc2b4bb23d94c4 : FAILURE in 30m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a92254051fed46a7995a49b476d7b162 : SUCCESS in 30m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/14740e26aec54affa24e3370b24daf9f : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f5a2157cd104312bb659334386e7cc2 : SUCCESS in 12m 04s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":22},{"id":"036833c6d88b971374d6a88d0edbcbd6376813c2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-31 08:11:37.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"6ccfada1a6ddf323cf6d4f24cb3a2b25caac67d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 10:21:05.000000000","message":"Patch Set 22:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5e26915c402477fb37230d656909181 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3069119e3c45288c36687041e2f8ea : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60b6b1329b9444ae965a56988c13ffab : SUCCESS in 9m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8601899eee64e4db476d19714126a12 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd2f13139b5743f19916ca96e4736da0 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9099bd0aa30246c69f723edaa4a610fb : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccde316f3fe144298d4edc8951200c49 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2e89ba6aae3a43ea83efeb38e8f2cc95 : SUCCESS in 7m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bb3d3f6f20c247aaa80388efa6cfae8f : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c9b8ce5014bf432b959d6bc2cf50eed4 : SUCCESS in 30m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cfde06831f2d41a8afe602180012a22c : SUCCESS in 1h 03m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/743ed7ec3d124c18a1586132aeaaf86a : FAILURE in 44m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7a3df9dc5a904cf89578404068e1ba24 : SUCCESS in 1h 05m 35s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d3088ef559bf4811a94c7e97ac5def52 : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2481eb762ce842079047c0bc002e0c13 : SUCCESS in 12m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3dd9ae6862ba4778b078d2c89d6e26e7 : SUCCESS in 2h 02m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a2c367e3cd94487486832ff90fb1cc70 : SUCCESS in 36m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2bb2c2c7a2854d7aaeccad9efcd7bb64 : SUCCESS in 29m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1a47d473d2e2428dbd3d4200c1345d6b : SUCCESS in 25m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/20dca019e5cd40fab3a1ffc68a9a41f0 : SUCCESS in 15m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4206a683d4b3441998510105f4485a37 : SUCCESS in 11m 44s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":22},{"id":"4813577a5ad99219337c989f6a6ad85aa3f9f147","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-08-31 10:41:32.000000000","message":"Patch Set 22:\n\nrecheck","accounts_in_message":[],"_revision_number":22},{"id":"25bff127c51fe20ef42133c0924a1dc99b5c9f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 12:57:47.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c31d75aedaaa49d78e51a3154b218502 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9ba0a4127fb49848b19066529a2850f : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f247ea7705149d7bf67b218e07e8612 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03c9de578c1c4310be13d87da57b6ab6 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/213e6a951d8c48d6ac51ae919545dfae : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6442a752fab4e21b787c85b09225361 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0be3d6d04bda492192ca64736e74f921 : SUCCESS in 4m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c8272b44dafe4d719426aff1bc967f2d : SUCCESS in 7m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca387fdf8f2e49e8a611197943a66882 : SUCCESS in 14m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a7c92154f60d4c5c8c3a84e6ab665c02 : SUCCESS in 32m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/90de4639ecd9464989c2156638755078 : SUCCESS in 1h 12m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db6535fc37094483ac7be2127a11635c : SUCCESS in 41m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/73d3f9a7e2c9448b9b2f41ac2f7ac284 : SUCCESS in 1h 05m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6fd206e8096d4779a4c16ab0dacbddb9 : SUCCESS in 18m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/609d383c34a74b54b5e741aca2aad450 : SUCCESS in 16m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b8d2d62a9121404fa869b33e2205e072 : SUCCESS in 2h 11m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dc6f961b3683448eb76ec9ada4008211 : SUCCESS in 33m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5f0f13471fa48518754f78a053afb55 : SUCCESS in 28m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a78fa3aaebff4b5381384b7c29cfae84 : SUCCESS in 26m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ba17f2f4b788449f9fdae10bbd00bec1 : SUCCESS in 13m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/741fd62355294730a5305ae79bcb6a03 : SUCCESS in 11m 01s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":22},{"id":"8c56a50fc3e60ca4d8a1c0003430540c3b00f8db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 12:58:15.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"424f024ae0eb65a55c7af2d449985533a975294d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 14:48:03.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f8c7d93673c454293187f58357b594c : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15775ff3830b4377abd60e58e602f1b5 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20d305dc67e34c60ae32029c2389d9b5 : SUCCESS in 8m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbbab327adf04964bcd25158e39c55e8 : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7f44eb7c10143b8acb78ff78d315327 : SUCCESS in 4m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a495ed1b97248b9a74db57bb7d9712b : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2cecb574c1ee41a0b58236e1a7d71b50 : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75123ddbc2bd4b58b68f4395611f6e56 : SUCCESS in 30m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d847a207305047f2b18d16580cc3cc42 : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa290714b2184bf5ba6322bd5f1e4070 : SUCCESS in 40m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3edd4035589f49ef8c9a6d679d646ce7 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/943a6e3a96094fea8d0d1a0203729346 : SUCCESS in 18m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14b497d69773483db9fc34c6cf680a60 : SUCCESS in 14m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/149f8f4d304a4125afbdafb648e1a111 : SUCCESS in 1h 48m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eb96b0387a164e99a12199ace86e3c83 : SUCCESS in 36m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d58c051d9314090aa2a290a7fda769b : SUCCESS in 30m 17s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e75ea0acf6d40c9b07558a97d0c351a : SUCCESS in 29m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ddde9ec4f3e4474e9b7f0f58b3c54fc0 : SUCCESS in 14m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56c86d8e36d94ae4b85893ff192b421d : SUCCESS in 11m 24s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":22},{"id":"430d309a511735199a242b2baed41cc81a622d55","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 14:48:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":22},{"id":"074cc4bfcceda76e978d58e9ead7387aaa36d0c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 14:49:57.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5af1892757ac401a9090c124e357f80d : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dec5a02b738441be8f8420e100b90bce : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"f639ad0b3514c3a6a485b2e75082ac0b3b850454","revisions":{"53297f68cf23737f0dfa4e1b629e1d4220389782":{"kind":"REWORK","_number":1,"created":"2021-05-13 18:43:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/1"}}},"commit":{"parents":[{"commit":"928d0a5cb31fb6225680951f9120411561ecab51","subject":"Merge \"Fixes for Ansible 2.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/928d0a5cb31fb6225680951f9120411561ecab51"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 18:42:52.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53297f68cf23737f0dfa4e1b629e1d4220389782"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53297f68cf23737f0dfa4e1b629e1d4220389782"}]},"branch":"refs/heads/master"},"22275602fe5b773c368780f9a7cd170bc54bd4ab":{"kind":"REWORK","_number":2,"created":"2021-05-13 19:32:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/2"}}},"commit":{"parents":[{"commit":"928d0a5cb31fb6225680951f9120411561ecab51","subject":"Merge \"Fixes for Ansible 2.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/928d0a5cb31fb6225680951f9120411561ecab51"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 19:32:24.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22275602fe5b773c368780f9a7cd170bc54bd4ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22275602fe5b773c368780f9a7cd170bc54bd4ab"}]},"branch":"refs/heads/master"},"acd0f858d667b47062fd881c983dd12c1b2e99ae":{"kind":"REWORK","_number":3,"created":"2021-05-14 08:07:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/3"}}},"commit":{"parents":[{"commit":"928d0a5cb31fb6225680951f9120411561ecab51","subject":"Merge \"Fixes for Ansible 2.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/928d0a5cb31fb6225680951f9120411561ecab51"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-14 08:07:01.000000000","tz":0},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/acd0f858d667b47062fd881c983dd12c1b2e99ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/acd0f858d667b47062fd881c983dd12c1b2e99ae"}]},"branch":"refs/heads/master"},"42693c34333a3ffb21177ba9afb910578c0ef722":{"kind":"REWORK","_number":4,"created":"2021-05-14 08:09:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/4"}}},"commit":{"parents":[{"commit":"928d0a5cb31fb6225680951f9120411561ecab51","subject":"Merge \"Fixes for Ansible 2.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/928d0a5cb31fb6225680951f9120411561ecab51"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-14 08:09:41.000000000","tz":0},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42693c34333a3ffb21177ba9afb910578c0ef722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42693c34333a3ffb21177ba9afb910578c0ef722"}]},"branch":"refs/heads/master"},"aee1407a818a74ec26f3445c87a3b91dd7ddd037":{"kind":"REWORK","_number":5,"created":"2021-05-14 11:40:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/5"}}},"commit":{"parents":[{"commit":"928d0a5cb31fb6225680951f9120411561ecab51","subject":"Merge \"Fixes for Ansible 2.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/928d0a5cb31fb6225680951f9120411561ecab51"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-14 11:40:24.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aee1407a818a74ec26f3445c87a3b91dd7ddd037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aee1407a818a74ec26f3445c87a3b91dd7ddd037"}]},"branch":"refs/heads/master"},"5f87c6726719e2a0225d30c84d2a9b53de8a0c25":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-05-17 15:25:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/6"}}},"commit":{"parents":[{"commit":"b3e535d9bc99aadc6b1339b4c903fe5143b03f61","subject":"Revert \"Use OpenStack Wallaby release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3e535d9bc99aadc6b1339b4c903fe5143b03f61"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-17 15:25:02.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f87c6726719e2a0225d30c84d2a9b53de8a0c25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f87c6726719e2a0225d30c84d2a9b53de8a0c25"}]},"branch":"refs/heads/master"},"666f477db34e2d8e0b732ca8f18fc8497b4c7e8f":{"kind":"REWORK","_number":7,"created":"2021-05-17 16:12:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/7"}}},"commit":{"parents":[{"commit":"b3e535d9bc99aadc6b1339b4c903fe5143b03f61","subject":"Revert \"Use OpenStack Wallaby release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3e535d9bc99aadc6b1339b4c903fe5143b03f61"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-17 16:11:29.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/666f477db34e2d8e0b732ca8f18fc8497b4c7e8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/666f477db34e2d8e0b732ca8f18fc8497b4c7e8f"}]},"branch":"refs/heads/master"},"5ee8933a53e38c9cbec49adfee3549aafe0315c7":{"kind":"REWORK","_number":8,"created":"2021-05-18 15:07:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/8"}}},"commit":{"parents":[{"commit":"adac2d2997a5cc6086f01e79f94320f996c18ce1","subject":"Revert \"Use OpenStack Wallaby release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adac2d2997a5cc6086f01e79f94320f996c18ce1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-18 15:05:42.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee8933a53e38c9cbec49adfee3549aafe0315c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ee8933a53e38c9cbec49adfee3549aafe0315c7"}]},"branch":"refs/heads/master"},"36ac1084e9ac5c00c6f559c6cfa2508c62005dd5":{"kind":"REWORK","_number":9,"created":"2021-05-18 15:12:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/9"}}},"commit":{"parents":[{"commit":"adac2d2997a5cc6086f01e79f94320f996c18ce1","subject":"Revert \"Use OpenStack Wallaby release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adac2d2997a5cc6086f01e79f94320f996c18ce1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-18 15:09:09.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36ac1084e9ac5c00c6f559c6cfa2508c62005dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/36ac1084e9ac5c00c6f559c6cfa2508c62005dd5"}]},"branch":"refs/heads/master"},"0c530a8e74fa5a0677610f33fef079027c48f8c9":{"kind":"REWORK","_number":10,"created":"2021-05-19 12:38:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/10"}}},"commit":{"parents":[{"commit":"adac2d2997a5cc6086f01e79f94320f996c18ce1","subject":"Revert \"Use OpenStack Wallaby release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adac2d2997a5cc6086f01e79f94320f996c18ce1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-19 12:38:32.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c530a8e74fa5a0677610f33fef079027c48f8c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c530a8e74fa5a0677610f33fef079027c48f8c9"}]},"branch":"refs/heads/master"},"5958b785dd9b06fb1fc11c1899435a5a9f214afa":{"kind":"REWORK","_number":11,"created":"2021-05-21 14:06:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/791304/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/11"}}},"commit":{"parents":[{"commit":"adac2d2997a5cc6086f01e79f94320f996c18ce1","subject":"Revert \"Use OpenStack Wallaby release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adac2d2997a5cc6086f01e79f94320f996c18ce1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-21 14:06:14.000000000","tz":120},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5958b785dd9b06fb1fc11c1899435a5a9f214afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5958b785dd9b06fb1fc11c1899435a5a9f214afa"}]},"branch":"refs/heads/master"},"8f2b33d789b45f785731715cfb43a6e8d0cefbe0":{"kind":"REWORK","_number":12,"created":"2021-06-07 14:20:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/12"}}},"commit":{"parents":[{"commit":"1dc6d98be73a4f5c795d66a9c8bd35c779e1f682","subject":"Revert \"Use OpenStack Wallaby release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1dc6d98be73a4f5c795d66a9c8bd35c779e1f682"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 14:19:52.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f2b33d789b45f785731715cfb43a6e8d0cefbe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f2b33d789b45f785731715cfb43a6e8d0cefbe0"}]},"branch":"refs/heads/master"},"fbaf4d57b203394ba55837ef8a025aa456ba3c3a":{"kind":"REWORK","_number":13,"created":"2021-06-08 14:35:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/13"}}},"commit":{"parents":[{"commit":"1dc6d98be73a4f5c795d66a9c8bd35c779e1f682","subject":"Revert \"Use OpenStack Wallaby release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1dc6d98be73a4f5c795d66a9c8bd35c779e1f682"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-08 14:35:14.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbaf4d57b203394ba55837ef8a025aa456ba3c3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbaf4d57b203394ba55837ef8a025aa456ba3c3a"}]},"branch":"refs/heads/master"},"de72ff289783b33f8e37c1b3b8eaf1b3f84e0e22":{"kind":"REWORK","_number":14,"created":"2021-06-08 14:39:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/14"}}},"commit":{"parents":[{"commit":"1dc6d98be73a4f5c795d66a9c8bd35c779e1f682","subject":"Revert \"Use OpenStack Wallaby release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1dc6d98be73a4f5c795d66a9c8bd35c779e1f682"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-08 14:39:00.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de72ff289783b33f8e37c1b3b8eaf1b3f84e0e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de72ff289783b33f8e37c1b3b8eaf1b3f84e0e22"}]},"branch":"refs/heads/master"},"5cf30afab9c12944d45069777d08f515ce8ff6d0":{"kind":"REWORK","_number":15,"created":"2021-06-09 13:07:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/15"}}},"commit":{"parents":[{"commit":"e03c7d7c57ff3c73280be0cd4b9f5fb09f999444","subject":"Merge \"Ubuntu: CI: rely on os_distribution, remove interfaces pause\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e03c7d7c57ff3c73280be0cd4b9f5fb09f999444"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-09 13:04:39.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cf30afab9c12944d45069777d08f515ce8ff6d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5cf30afab9c12944d45069777d08f515ce8ff6d0"}]},"branch":"refs/heads/master"},"7e8c670e8723da49b87e096d8d8e2a519d1e38ca":{"kind":"REWORK","_number":16,"created":"2021-06-09 18:50:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/16"}}},"commit":{"parents":[{"commit":"e03c7d7c57ff3c73280be0cd4b9f5fb09f999444","subject":"Merge \"Ubuntu: CI: rely on os_distribution, remove interfaces pause\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e03c7d7c57ff3c73280be0cd4b9f5fb09f999444"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-09 18:50:52.000000000","tz":0},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e8c670e8723da49b87e096d8d8e2a519d1e38ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e8c670e8723da49b87e096d8d8e2a519d1e38ca"}]},"branch":"refs/heads/master"},"87bbb1adc7d6059b1f82daf812a80e514e8fb0e8":{"kind":"REWORK","_number":17,"created":"2021-07-26 15:57:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/17"}}},"commit":{"parents":[{"commit":"e9a3733e24c5280dda241aec839097d005014c99","subject":"Merge \"Prevent matching unrelated images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9a3733e24c5280dda241aec839097d005014c99"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-26 15:57:20.000000000","tz":60},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87bbb1adc7d6059b1f82daf812a80e514e8fb0e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87bbb1adc7d6059b1f82daf812a80e514e8fb0e8"}]},"branch":"refs/heads/master"},"287a05e3ae17e57297a2981cf82a571d45f5bd63":{"kind":"REWORK","_number":18,"created":"2021-08-17 15:49:22.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/791304/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/18"}}},"commit":{"parents":[{"commit":"a48cc2498859f5949d304266d32ce3e620ad7f36","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a48cc2498859f5949d304266d32ce3e620ad7f36"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-17 15:48:53.000000000","tz":120},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/287a05e3ae17e57297a2981cf82a571d45f5bd63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/287a05e3ae17e57297a2981cf82a571d45f5bd63"}]},"branch":"refs/heads/master"},"b3078dce7ad87cabdf6057b8869295dab053c79c":{"kind":"REWORK","_number":19,"created":"2021-08-20 16:11:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/791304/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/19"}}},"commit":{"parents":[{"commit":"a48cc2498859f5949d304266d32ce3e620ad7f36","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a48cc2498859f5949d304266d32ce3e620ad7f36"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-20 16:11:18.000000000","tz":120},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3078dce7ad87cabdf6057b8869295dab053c79c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b3078dce7ad87cabdf6057b8869295dab053c79c"}]},"branch":"refs/heads/master"},"26e802d5746641606f156f1ff3ffa4705cb816e7":{"kind":"REWORK","_number":20,"created":"2021-08-20 16:44:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/04/791304/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/20"}}},"commit":{"parents":[{"commit":"a48cc2498859f5949d304266d32ce3e620ad7f36","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a48cc2498859f5949d304266d32ce3e620ad7f36"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-20 16:44:27.000000000","tz":0},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26e802d5746641606f156f1ff3ffa4705cb816e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26e802d5746641606f156f1ff3ffa4705cb816e7"}]},"branch":"refs/heads/master"},"431ed0c5211b0d898768cfaeaba8af0c31bb652d":{"kind":"REWORK","_number":21,"created":"2021-08-20 20:52:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/791304/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/21"}}},"commit":{"parents":[{"commit":"a48cc2498859f5949d304266d32ce3e620ad7f36","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a48cc2498859f5949d304266d32ce3e620ad7f36"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-20 20:52:38.000000000","tz":120},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/431ed0c5211b0d898768cfaeaba8af0c31bb652d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/431ed0c5211b0d898768cfaeaba8af0c31bb652d"}]},"branch":"refs/heads/master"},"f639ad0b3514c3a6a485b2e75082ac0b3b850454":{"kind":"REWORK","_number":22,"created":"2021-08-21 07:57:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/04/791304/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/04/791304/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/04/791304/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/04/791304/22"}}},"commit":{"parents":[{"commit":"a48cc2498859f5949d304266d32ce3e620ad7f36","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a48cc2498859f5949d304266d32ce3e620ad7f36"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-13 14:33:17.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-08-21 07:57:29.000000000","tz":120},"subject":"Use ansible_facts to reference facts","message":"Use ansible_facts to reference facts\n\nBy default, Ansible injects a variable for every fact, prefixed with\nansible_. This can result in a large number of variables for each host,\nwhich at scale can incur a performance penalty. Ansible provides a\nconfiguration option [0] that can be set to False to prevent this\ninjection of facts. In this case, facts should be referenced via\nansible_facts.\u003cfact\u003e.\n\nThis change updates all references to Ansible facts within Kayobe\nfrom using individual fact variables to using the items in the\nansible_facts dictionary. This allows users to disable fact variable\ninjection in their Ansible configuration, which may provide some\nperformance improvement.\n\nThis change disables fact variable injection in the ansible\nconfiguration used in CI, to catch any attempts to use the injected\nvariables.\n\n[0] https://docs.ansible.com/ansible/latest/reference_appendices/config.html#inject-facts-as-vars\n\nStory: 2007993\nTask: 42464\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/791276\n\nChange-Id: I14db53ed6e57d37bbd28dd5819e432e3fe6628b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f639ad0b3514c3a6a485b2e75082ac0b3b850454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f639ad0b3514c3a6a485b2e75082ac0b3b850454"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
