)]}'
{"id":"openstack%2Fkayobe~791752","triplet_id":"openstack%2Fkayobe~master~Ide0639bfc2f1bcd57c027c59db33081c4de780b5","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ide0639bfc2f1bcd57c027c59db33081c4de780b5","subject":"[DNM] Test Tenks fix","status":"ABANDONED","created":"2021-05-17 13:16:28.000000000","updated":"2021-05-18 09:48:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c891eabe6425313cbaeb2e5a4c4d3b886b760f24","_number":791752,"virtual_id_number":791752,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-17 16:13:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 13:33:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"522205d9299a7285dd9a2cb971ab51a27d1abd95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-17 13:16:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ad3e268693926ff442bf0a2fefc3e3100ca739c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 13:33:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f64768e679c743baa834de38f48caf40 : SUCCESS in 8m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f0ab0304904cd985679c579644b7bb : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98b555ce6b234563a7a06e0aa48757a2 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a2fadbc4b394eb5a0704374d775021e : SUCCESS in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f0dfee01d2f4a5fbdb3cc3b3ba9310f : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e8519679f80434bb0033fa814dfceb2 : SUCCESS in 9m 26s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"83542d9a540a6fb1cda3fd42f2e635cff4ed7c67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-17 13:54:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5179f95aaabd0c3582560bceec429ac73d97af81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 14:43:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a994d65129d4f86a9fe90e7eb9082db : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7fc47dd8fa54b348fb5fce39f32d077 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00420c1a3ce54e8385b736e7b714335f : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10f8b4bde7d340d8af131001d7d2b97c : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/088737185a9541f58ea0ecb24772d185 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b0d41627474d18915c5d758c61dceb : SUCCESS in 8m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9723918f02124c6d846d504e83254506 : FAILURE in 47m 01s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":2},{"id":"7861436b19f3fa867e61f02f23dc88d6adbd25c5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-17 15:00:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a9d988e36297ce18abcbc637ef985a29beb33068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 15:48:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2d97c81ee944f2f8032b78a933e57fa : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fc0f8ac39e14c93946fcd5eeb97b834 : SUCCESS in 7m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70c891925e6c4d02b0993049ef302276 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/375a3870b89d4cd7bf077bf0e65f9a79 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf77b6eec5e446299e9de00ed1e018f3 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cafc8c86079b4fc89a1a96fb24c451d0 : SUCCESS in 11m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7111534883414208b89fffa63c39ff44 : SUCCESS in 43m 40s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":2},{"id":"27371c6ca7b110d649bc34db7b7603c9fc408a08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-17 16:07:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4f405f8a88685dfff306b471e2447bf53d4c35e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 16:13:45.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"c891eabe6425313cbaeb2e5a4c4d3b886b760f24","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-18 09:48:27.000000000","message":"Abandoned\n\nThe related fix was merged in Tenks.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3b6525d1ddc7cb0d72fb0583003f881a9e1abc65","revisions":{"a160fb955798835425020ec0517e3ae1258b77f4":{"kind":"REWORK","_number":1,"created":"2021-05-17 13:16:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/791752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/791752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/791752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/791752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/791752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/791752/1"}}},"commit":{"parents":[{"commit":"928d0a5cb31fb6225680951f9120411561ecab51","subject":"Merge \"Fixes for Ansible 2.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/928d0a5cb31fb6225680951f9120411561ecab51"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-17 13:16:16.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-17 13:16:16.000000000","tz":120},"subject":"[DNM] Test Tenks fix","message":"[DNM] Test Tenks fix\n\nChange-Id: Ide0639bfc2f1bcd57c027c59db33081c4de780b5\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/791751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a160fb955798835425020ec0517e3ae1258b77f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a160fb955798835425020ec0517e3ae1258b77f4"}]},"branch":"refs/heads/master"},"caad8369ac03b64b94eb772f46a918aeb4eb182a":{"kind":"REWORK","_number":2,"created":"2021-05-17 13:54:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/791752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/791752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/791752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/791752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/791752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/791752/2"}}},"commit":{"parents":[{"commit":"928d0a5cb31fb6225680951f9120411561ecab51","subject":"Merge \"Fixes for Ansible 2.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/928d0a5cb31fb6225680951f9120411561ecab51"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-17 13:54:40.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-17 13:54:40.000000000","tz":120},"subject":"[DNM] Test Tenks fix","message":"[DNM] Test Tenks fix\n\nChange-Id: Ide0639bfc2f1bcd57c027c59db33081c4de780b5\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/791751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/caad8369ac03b64b94eb772f46a918aeb4eb182a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/caad8369ac03b64b94eb772f46a918aeb4eb182a"}]},"branch":"refs/heads/master"},"3b6525d1ddc7cb0d72fb0583003f881a9e1abc65":{"kind":"REWORK","_number":3,"created":"2021-05-17 16:07:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/52/791752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/791752/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/791752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/791752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/791752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/791752/3"}}},"commit":{"parents":[{"commit":"928d0a5cb31fb6225680951f9120411561ecab51","subject":"Merge \"Fixes for Ansible 2.10\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/928d0a5cb31fb6225680951f9120411561ecab51"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-17 16:07:11.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-17 16:07:11.000000000","tz":120},"subject":"[DNM] Test Tenks fix","message":"[DNM] Test Tenks fix\n\nChange-Id: Ide0639bfc2f1bcd57c027c59db33081c4de780b5\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/791751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b6525d1ddc7cb0d72fb0583003f881a9e1abc65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b6525d1ddc7cb0d72fb0583003f881a9e1abc65"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
