)]}'
{"id":"openstack%2Fkayobe~791787","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I3c15c6976296abf9c78f1480c1779b854145c7ca","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"I3c15c6976296abf9c78f1480c1779b854145c7ca","subject":"Remove default time sources from customisable configuration","status":"MERGED","created":"2021-05-18 06:31:43.000000000","updated":"2021-05-18 10:34:29.000000000","submitted":"2021-05-18 10:32:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791787","cherry_pick_of_change":790819,"cherry_pick_of_patch_set":2,"meta_rev_id":"763375cfe13c796820412e555b2dbc8eff78e031","_number":791787,"virtual_id_number":791787,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-18 10:32:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-05-18 08:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-18 09:06:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-18 09:06:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 08:37:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 08:44:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"da69d32e97ed72e8d99caa7308255ef66d6ae49a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-18 06:31:43.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c66e4cf25e2f26e998cc78a612309dc399c1728e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 08:37:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78d90bd0636148e5a071644ce03e84fb : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60b64315e3264fa1a40c941e632cf817 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a66715765f545f8b1a5b265ea0944b7 : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40f2f7462eaa45a6b7b11e9779a7154c : SUCCESS in 9m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d051b9ac7154479ca654505f12117dd5 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cb447e17d6f4bf2bda5cdc9fe668486 : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/03212b7054bd4b999854990c2c8d7eaa : SUCCESS in 7m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/22301b728b02479db0f2ecf6336e4c42 : SUCCESS in 13m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ce4892721b6e4621a53f83839356afce : SUCCESS in 19m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e00eb1edd3bf48269de7581bd1fcd60c : SUCCESS in 1h 07m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe0df6f689b14869a2f45924a0f4656e : SUCCESS in 59m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/60ec53f4b812404da1bd7d54099eed0a : SUCCESS in 1h 17m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/546a3635d0404d7eaf99143d1b6644b4 : SUCCESS in 18m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b459303476746ada36fa225d9301df7 : SUCCESS in 16m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3a79fb40bab143e39a98a497e93b8ade : SUCCESS in 1h 57m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/eb752210fd254cbe8540ec013320e80d : SUCCESS in 32m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47fcf5193f484ac7af3543d7c9bff008 : SUCCESS in 26m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/839f0fb79be04700b75eb96ef964edc8 : SUCCESS in 24m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/991e2938d4554078be3c11d103ca863a : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07b08a3d05fc43e6bf90ac963c60b1a7 : SUCCESS in 11m 15s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"c10b38fdd463e554ce71d52ded92d7d2a4f1a178","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-18 08:44:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dee0836ccdaa480f3d2095e8b459799160aaf750","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-18 09:06:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47042d21b0ec3f7068622b9ae961e9d461cdce20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 09:07:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf4c9acf47762a67e13cf224d114099d846f1a81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 10:32:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eb970bb64024cb484df25a2516a5145 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26bac400cc1d4e05b6488f0acf95b237 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db05ab611efc434ca44eb3deb49871f1 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06f8ccbfbf3345f0b9ff19f74a7e03b6 : SUCCESS in 9m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c91619d4e349499097750c0d423d6be0 : SUCCESS in 7m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e77ef9ce1c64a19b61b07285eb09635 : SUCCESS in 10m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/06bacc92965a4c17bd0902979514f078 : SUCCESS in 18m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8896300a154a437e9dbf9a70ede11ad7 : SUCCESS in 51m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8421432144c47fe89a18fcefcc41814 : SUCCESS in 37m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d08d1eb837014935b669921fa5ee23cf : SUCCESS in 55m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/044f85fcb4814211b2b5bd510f30aebd : SUCCESS in 15m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/857f0dca212a41a18caa0b0fd678e61e : SUCCESS in 12m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fa9b72c762e04011aa53cc2d30476a82 : SUCCESS in 1h 23m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/608d4f84e27d416c9c649afa07ca9e9c : SUCCESS in 27m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99d110dce80a4404b35ebfa0c1752490 : SUCCESS in 23m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f279f6ac480d438db7a87ae4919a3404 : SUCCESS in 22m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cf0b24ea3f1849198b9de8b955b66945 : SUCCESS in 12m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/663d65feab354a04abaa0105570bfbf1 : SUCCESS in 11m 05s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"5687a309b73aed9d5451ae8dd69c18f298b8a644","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 10:32:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"763375cfe13c796820412e555b2dbc8eff78e031","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 10:34:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb3fdaaa8994970b4e30b75d49d1a69 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c01e5f0a6b4b8ccbb111439e9b2a6cac17388ef6","revisions":{"c01e5f0a6b4b8ccbb111439e9b2a6cac17388ef6":{"kind":"REWORK","_number":1,"created":"2021-05-18 06:31:43.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/791787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/791787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/791787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/791787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/791787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/791787/1"}}},"commit":{"parents":[{"commit":"010a18334f08d173c9b8185f739324f58a37da89","subject":"Adds support for configuring chrony","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/010a18334f08d173c9b8185f739324f58a37da89"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-11 20:29:48.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-18 06:31:43.000000000","tz":0},"subject":"Remove default time sources from customisable configuration","message":"Remove default time sources from customisable configuration\n\nIt seems unlikely that anyone would want to extend the time servers\nvariable, so let\u0027s keep it simple and not add an additional variable\nwhere one isn\u0027t necessary. This is also consistent with how we configure\nDNS.\n\nThese variables haven\u0027t made it into an official release yet, so now\nis the time to clean up.\n\nChange-Id: I3c15c6976296abf9c78f1480c1779b854145c7ca\n(cherry picked from commit 223e310bc2338bb5f86c3a3de31f69361eba2f73)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c01e5f0a6b4b8ccbb111439e9b2a6cac17388ef6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c01e5f0a6b4b8ccbb111439e9b2a6cac17388ef6"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
