)]}'
{"id":"openstack%2Fkayobe~792243","triplet_id":"openstack%2Fkayobe~master~I275102ec6b5bab6982577b52fd29654c874446ce","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I275102ec6b5bab6982577b52fd29654c874446ce","subject":"chrony: cleanup during overcloud host upgrade","status":"MERGED","created":"2021-05-19 18:48:04.000000000","updated":"2021-06-07 18:07:08.000000000","submitted":"2021-06-07 18:05:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792243","meta_rev_id":"6f41726d3ded280419d2fbb96ce412d5b9054ece","_number":792243,"virtual_id_number":792243,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-07 18:05:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-07 07:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-07 15:22:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-07 15:22:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-19 19:54:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 07:55:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-06-07 09:09:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"72340d48f08d32e06a9f635045431a19b290fed0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-19 18:48:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"145721b787911394277cbfeb4b4c7e047407c29a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 19:54:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf55bf89d8164d32b80fd42fe80acb4c : FAILURE in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01d18b3317d4448b33dd7c13f02c3ac : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57a085097f93405a8749bb885cf86f6a : FAILURE in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7204e9cee8824fb0931df42492fa72e3 : FAILURE in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae849ae623e448c2a99f56c2005d09db : FAILURE in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66b48724285346d3a98f3bc5db019ac0 : SUCCESS in 9m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c6c61d48257c4eb7a73cbfaa1221954a : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9392cd24b5f64efd9c5913a04ab27ddd : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/07988002ed7d4d49a555f2b2e323566e : SUCCESS in 16m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4d9c47e7690e4aa4a6d20f5fbb0e6ae7 : FAILURE in 16m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14bb8e4f6d554c4da2226c617df5f518 : FAILURE in 10m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d88cb4bb781d47d3b4fc704df6790fd2 : FAILURE in 13m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0175655319a24e6695c198c54904d641 : FAILURE in 18m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4e60a87b3e141a2aae0924abead97bd : FAILURE in 13m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fd65585f090f4c298efd61dfa8b98785 : FAILURE in 1h 02m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/39669fde344549c79d0da16ba64ecfa2 : POST_FAILURE in 57m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fbf3b4186344759bbcc079b0f64c9f7 : SUCCESS in 21m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ba1f12210ae24bc29b1258a942d3c67e : SUCCESS in 24m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/542529fae5fe4722ae6dac2296569c9e : SUCCESS in 13m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d98d47b4cdde4e8eada70abae50d557a : SUCCESS in 11m 31s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"a59edcb579ccf60cbf7c5ca13b3f1d0f5fb4a29f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-21 09:52:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a8e48d705ffb229ace9de7dbca981458c906e44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-21 10:16:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"18ab7825cc2ca6e4f5aca6f4af74ed7d0ce8580c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:21:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d930325a733942399abc73d56fab5f88 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4002494057144157a80ae20f423b3a8d : SUCCESS in 11m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e2c6bede23a4bad9c0a88dff82df441 : SUCCESS in 7m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22c4150a91994aa380e31195bcbd0f2d : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/134e73e1246c451ca4a138239d9dba84 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deab8c0aa4a44211a2f1fbfd3ddae711 : SUCCESS in 9m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3ab6a17da3fc49bd8bc99ccf48df4953 : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4ba9da16b2446a799c166a7c9183fbe : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b22ec14722804176abde1b8928374e72 : SUCCESS in 21m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/428ea491bd2f4433a9078440a9834eaf : SUCCESS in 1h 12m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee2d25c9251f49a0a6badcc36e6e0ada : SUCCESS in 43m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9998a8ebc08a403bbeb5c22bc98f63ad : SUCCESS in 1h 12m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c995b87a1b554d37affae988c4f3ca60 : SUCCESS in 20m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4361a25480d4d208fa002a46266e479 : SUCCESS in 13m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c28680e16a144811ac90df4b2a1d5b1c : SUCCESS in 1h 32m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/04e0a335f0de432f8bd72296e0fc69e2 : SUCCESS in 40m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4b15809a4154b3da1bb61e66c5f40bc : SUCCESS in 23m 17s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/41afdf1b0d554f0692f8f9f0fa5524c8 : SUCCESS in 21m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3c7d713df38a46598a30af01020a95fe : SUCCESS in 13m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ae74798a1f843499df311698cd72a43 : SUCCESS in 11m 01s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"9a86a630ef6826f90e850239cf4fc5b22e2901d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-24 10:00:45.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f454e5c0b6387e4e77ba46e983015673e58f4edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 11:40:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37fa53203c3743b4b7a4af0338251ad9 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb759f200e245778a1559ad02f63fee : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29d5a1d8c1a34a3eae122128d0e0d8c2 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1aca6c810e2b4334955b74dfc8d831e9 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b65bd516ea2439195e12c512fc135c1 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/768e8c1648394451bc872880723b3521 : SUCCESS in 7m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e6c27a5c26e44f5bb5d542f9aca8bda : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a79d4ac28e5840f2ad0578dc45dbbb3e : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4f1d8d23c9ef4e6189cfa4083297730f : SUCCESS in 18m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/528161ab17694272aa95a00292deb962 : SUCCESS in 53m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d4c693e908a4d22a7b524c13195b7d4 : SUCCESS in 41m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/40821c8dd48f44b8be1946e506e9344c : SUCCESS in 58m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2a3ca7978b1a4f078ca80d5edf11f37f : SUCCESS in 15m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bfd8a2253bd4f2a81f595f2b46e137b : SUCCESS in 11m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6e408bae35034f95ba4863972739c648 : SUCCESS in 1h 37m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0a2ca6c2c2c34970a5b583dbda71dc45 : SUCCESS in 25m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f5eac34740248feab396ef49ce9d264 : SUCCESS in 22m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eb77c3725355490ea2482c7b34adea66 : FAILURE in 16m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4f54d03c01dd4d8b97b23b8b1f4f84a8 : SUCCESS in 15m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c6ba807bc1f48bcb219a970e98f52b2 : SUCCESS in 8m 47s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"902e097c0361796117ec6d6b3c26741fa1cbce2c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-25 09:07:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9dd0d27fb05b02f496fab91d7fe2d0bc2a0eab9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 11:10:06.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd26bbd3b7964743bd79f6edc5a7bfcc : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee0910d04c98422d8a585c9250b5d0cd : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47115ab3d6db4980a9c3b6c3dd8bc4cc : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2144343f78dd4e9ab32ffd55a2042e87 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65d8a853ffd64722bc4307c80cce87a6 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03eee4df39ab4c5ab70f88317e86fef7 : SUCCESS in 9m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/91a1814a6160460085efcb84c07a1b22 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6870a5cbcb164e42837c6134a78344f2 : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c8f28697169749359ae58444a8d198d4 : SUCCESS in 20m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1decac634dfb44e187d162ebd601ccdd : FAILURE in 50m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3107627af34b4c9daf3701b5c879494c : SUCCESS in 1h 00m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6678edaf474f44c296183334fa554c7e : FAILURE in 57m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a444cf0028e4443ea169c0694fc67f6c : SUCCESS in 19m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92c5ef0a410b4786966cc7f4f9d9b0fa : SUCCESS in 14m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3b8ec31cbb524aa6bf5f4b2fd8366343 : FAILURE in 2h 01m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/191f17bfab584a21bb08730e68b8cdf5 : SUCCESS in 31m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cbd58d8a35c44329d9c15a3952043ba : SUCCESS in 23m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a43d27ca79684a55a324f6d7bf7a969c : SUCCESS in 28m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e98adbc0aae14403a64da231fe639d61 : FAILURE in 12m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcd7759d8c7d45038d26670933c42db4 : SUCCESS in 12m 05s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"45d2077c36c4038b8e42c5400d90b6125dcd24a6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-03 12:12:57.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"14f3631f16d516832b0f41a07ab9a6d3f7f0dbdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 14:38:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/132f93c40e9943988da4bf90c0a06c0b : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25fb5455ad5040f0a31263be0e6ace81 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73dbf9c126be4088b10a6c0f8fa6885a : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/118c8d2d7e8944668928059407ff3728 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e23831d01694bb68d00a96441d4d199 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709a0f5503394b37b0880b26cc38ca65 : SUCCESS in 9m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f8f15386aec34ef192503588db5994e6 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c5ee5eb02974b0f9f43592a2d8486f4 : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eea5759ee4b147dab3dc81fe9e310d5b : SUCCESS in 19m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d7f3c35e6ea147c38834735a81012e0a : SUCCESS in 1h 09m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebf181dde5ae4fcbac7914f719b3b9cc : SUCCESS in 46m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cc96dd30e36c4f8ba6ef6790a627e02f : SUCCESS in 55m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/96b9d1e41ca148f499717a5dd091266a : SUCCESS in 19m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5d65fc676bc451aba70d66a38e29d61 : SUCCESS in 13m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/daae3a0e948641a0837f869db447a1b9 : SUCCESS in 2h 23m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/81c6d02bf88b47928ca7c1fddd69c4ac : SUCCESS in 31m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dd8049ee65243aa8e466998f729efd4 : SUCCESS in 24m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d51d1487fd504ee3959f24a059db8829 : SUCCESS in 27m 03s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4c1bab63a5c7428eb0354d813a0e0af6 : SUCCESS in 12m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dc438986d78450da84048c4d3ff8a3b : SUCCESS in 10m 40s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"14f983bb1ca97b1d79b9eb9bc2d7922367a55670","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-07 07:55:12.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e6f2af4549373282486c40408aab4938569f9479","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-07 15:22:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f3c9923ceb87cf553416dcfa0496d21e1d957ca1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 15:24:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9e16b670c0acf22d37b52abe6e05553236da23ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 18:05:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0123d944d165416e84779b5f45894280 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc557e4e9f134b77b859a1c36fcfc753 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6b71c29878e4a3689074af74c34fb14 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9dd688371a342d0992d3b7d600b9224 : SUCCESS in 9m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/efee0e39308845ad96ef0cef40cb99cf : SUCCESS in 7m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25bcf47ed9b2438eabea6da52cb5bc59 : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2eacf8a5aab44f3929520da923153d0 : SUCCESS in 21m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a919acceff954650bd988588633cf43c : SUCCESS in 1h 05m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1977cc3f38404d389995fb706cc0ef40 : SUCCESS in 47m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/deea63da26dd45aeb7ae334625a369bc : SUCCESS in 1h 04m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/34365e510bca481fbc25b10d96cb4d2f : SUCCESS in 20m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/545a9952f23a4dafb173e5edeb806130 : SUCCESS in 13m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/acbae222f158446f82578e7fcb590afa : SUCCESS in 2h 28m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3ef98c22523948598f66d5199f609b3d : SUCCESS in 25m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/880cb097d5e34d1f831a263d994ef285 : SUCCESS in 25m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/002ac2f438b347dc99e8f2bc67416687 : SUCCESS in 30m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bfc5976d0616472999b4711a30f7bbfb : SUCCESS in 14m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0a737b54fae40cda5eb59b339be6a5b : SUCCESS in 10m 58s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"e3bca96510ad94ffc3c5ab7c43d0f38059579a1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 18:05:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"6f41726d3ded280419d2fbb96ce412d5b9054ece","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 18:07:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d35e93f8fb11458b8e3fab93376df47a : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dc53d29eaa0e5f67335aef3db23fc11a17dc21b0","revisions":{"c1c331ab903c91126c7cdfdb6329d3744c2464f0":{"kind":"REWORK","_number":1,"created":"2021-05-19 18:48:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/792243/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/792243/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/792243/1"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-19 18:47:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-19 18:47:36.000000000","tz":60},"subject":"WIP: chrony: cleanup during overcloud host upgrade","message":"WIP: chrony: cleanup during overcloud host upgrade\n\nChange-Id: I275102ec6b5bab6982577b52fd29654c874446ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1c331ab903c91126c7cdfdb6329d3744c2464f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1c331ab903c91126c7cdfdb6329d3744c2464f0"}]},"branch":"refs/heads/master"},"709591ccd215eca7cc3d4a65b9f8c92abbdc56aa":{"kind":"REWORK","_number":2,"created":"2021-05-21 09:52:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/792243/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/792243/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/792243/2"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-19 18:47:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-21 09:49:32.000000000","tz":60},"subject":"WIP: chrony: cleanup during overcloud host upgrade","message":"WIP: chrony: cleanup during overcloud host upgrade\n\nChange-Id: I275102ec6b5bab6982577b52fd29654c874446ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/709591ccd215eca7cc3d4a65b9f8c92abbdc56aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/709591ccd215eca7cc3d4a65b9f8c92abbdc56aa"}]},"branch":"refs/heads/master"},"9220605fbcb1c199a8d9b329f8cc6b2f46c36b29":{"kind":"REWORK","_number":3,"created":"2021-05-21 10:16:18.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/792243/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/792243/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/792243/3"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-19 18:47:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-21 10:14:17.000000000","tz":60},"subject":"WIP: chrony: cleanup during overcloud host upgrade","message":"WIP: chrony: cleanup during overcloud host upgrade\n\nChange-Id: I275102ec6b5bab6982577b52fd29654c874446ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9220605fbcb1c199a8d9b329f8cc6b2f46c36b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9220605fbcb1c199a8d9b329f8cc6b2f46c36b29"}]},"branch":"refs/heads/master"},"dc53d29eaa0e5f67335aef3db23fc11a17dc21b0":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-05-24 10:00:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/792243/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/43/792243/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/43/792243/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/43/792243/4"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-19 18:47:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-24 09:55:38.000000000","tz":60},"subject":"chrony: cleanup during overcloud host upgrade","message":"chrony: cleanup during overcloud host upgrade\n\nThe Kolla Ansible chrony container is disabled by default in the Wallaby\nrelease. A new kolla-ansible chrony-cleanup command can be used to clean\nup the container.\n\nThis change extends the \u0027kayobe overcloud host upgrade\u0027 command to cover\ncleaning up the chrony container (if disabled) and deploying a host\nchrony daemon.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/792119\n\nChange-Id: I275102ec6b5bab6982577b52fd29654c874446ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc53d29eaa0e5f67335aef3db23fc11a17dc21b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc53d29eaa0e5f67335aef3db23fc11a17dc21b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
