)]}'
{"id":"openstack%2Fkayobe~792281","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I567e83cac4d5e80bd097b30ddefa0cc188118ad2","project":"openstack/kayobe","branch":"stable/wallaby","topic":"throttle-stable/wallaby","hashtags":[],"change_id":"I567e83cac4d5e80bd097b30ddefa0cc188118ad2","subject":"Replace custom serial loop with throttle","status":"MERGED","created":"2021-05-20 11:13:05.000000000","updated":"2021-05-21 10:24:20.000000000","submitted":"2021-05-21 10:22:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792281-throttle-stable/wallaby","cherry_pick_of_change":788683,"cherry_pick_of_patch_set":2,"meta_rev_id":"db160f4ff3035ae9bfc37836ef4d3d2c60e962c0","_number":792281,"virtual_id_number":792281,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-21 10:22:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2021-05-20 14:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-05-21 08:15:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2021-05-21 08:15:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 13:18:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 14:36:02.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-05-21 08:15:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6f2fc69a3a73ed88ee4f41378971c50db63bc1ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-20 11:13:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9c3cdae608ceaa43881ab5969911b3b85fcbe054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 13:18:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d4577bb8fdf42f6a888ffbcf9ffde91 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccfedda904ea40d3a59b1e1e865f7b39 : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eebe42ea7fdc4252942b77128511a1a3 : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcbc4b707eb04785bbf5f47629f21a8d : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13deb92927a54dc79fa6f9dfb2f3ef7a : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2b0437685b41c5ae1d560a3ea7c9ea : SUCCESS in 8m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe536c82df774d58ac22bdff20ee9a64 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6563276dfbb24d73b8f986c2de4349bd : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65fa423ad9434e7f954a58c6a7dc2414 : SUCCESS in 18m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ef2a08840db54ad0b9701f23c99dbd25 : SUCCESS in 58m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4204d6decdd14d4b8c835f03ab798f0d : SUCCESS in 44m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1f80adce419e405184b3adf0d37c2ca6 : SUCCESS in 1h 02m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b9974540071d492aad819e8193c02a0c : SUCCESS in 15m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76030a8732b84b91a7c27f2e8d7dbf61 : SUCCESS in 13m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c83b6499724c4bcabe028957eb2bccc7 : SUCCESS in 1h 56m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dd2ed4b8a0844c19a30b92c7c4c113c8 : SUCCESS in 28m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9dc4fd40ebf4e3ab0e1be740fefebd7 : SUCCESS in 24m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0c72b5d1918e44b4927742529a9fc816 : SUCCESS in 19m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3bb8f35285dd482c93608b5aeae7a592 : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eac0254f4c0f4ae38f24c7cce60bcfb1 : SUCCESS in 10m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"3072c093ba96b2b18165887ff395a5199852d2e5","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-05-20 14:36:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad972d64c91ecb5d49df2567da237484e7aa5012","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-21 08:15:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e26d653d9f6bde6e0336ce60daf31c5bfb6be66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 08:15:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10f4a200bfa4324c5cde184f0797f592b2ef80a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 10:22:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8319ae4915824ea382b3b0ad1b41d24d : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/265904c3c0624ae2b9dbea85646b54fb : SUCCESS in 15m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52f52f00eef44288812618ebe3bfd66e : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2151d813444b4b3296ba5a2a2d8e1657 : SUCCESS in 10m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85a68abe81e040d5a39927fbbadf5ec2 : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8e74579d23ba4d9c81225f832ea3ff43 : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/32ae9b40f1724d118d4bc91df8bdebe2 : SUCCESS in 18m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bb9a42141b814fde92ea1e0ef5ca3895 : SUCCESS in 1h 02m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a51b45b631894f91817c55e289ffc13e : SUCCESS in 45m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/26574f93494145f7ae82ffe94d69f1fc : SUCCESS in 1h 09m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/197e6ee3d1514e298b6d0ec2626979e9 : SUCCESS in 17m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/236a8f5e328646b785622c84ad398f1f : SUCCESS in 13m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/19cd74c8c6904f06964d94b996363ab1 : SUCCESS in 2h 05m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f3db8d41a5bd4a479b90e537216e6737 : SUCCESS in 31m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c05351c05a34ccdb9e3fd472c22f0c7 : SUCCESS in 26m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/67aaf68a235b4b43937e69580e4cff5e : SUCCESS in 25m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6dc639c4d8d241939dab3ae364d2ccfd : SUCCESS in 13m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b86f09ed707413584986a24273f0d31 : SUCCESS in 11m 54s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"ebe78b9ff089f866d27cde8a7ebf44b647cd73f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 10:22:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"db160f4ff3035ae9bfc37836ef4d3d2c60e962c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 10:24:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab8273ede284a5686b017453ebbff82 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9a2f382002b8e202590ccb761a144ffdcb20fc7","revisions":{"a9a2f382002b8e202590ccb761a144ffdcb20fc7":{"kind":"REWORK","_number":1,"created":"2021-05-20 11:13:05.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/81/792281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/81/792281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/792281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/792281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/792281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/81/792281/1"}}},"commit":{"parents":[{"commit":"8dced46161f42c9766f272aa58794186e1dd5979","subject":"Merge \"Make docker registry network mode configurable\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8dced46161f42c9766f272aa58794186e1dd5979"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-29 09:38:29.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-20 11:13:05.000000000","tz":0},"subject":"Replace custom serial loop with throttle","message":"Replace custom serial loop with throttle\n\nWe can use throttle now that we require Ansible 2.9.\n\nChange-Id: I567e83cac4d5e80bd097b30ddefa0cc188118ad2\n(cherry picked from commit 9033d28435ecfbacc941045ab470e4de3869e4e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9a2f382002b8e202590ccb761a144ffdcb20fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9a2f382002b8e202590ccb761a144ffdcb20fc7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
