)]}'
{"id":"openstack%2Fkayobe~792571","triplet_id":"openstack%2Fkayobe~master~I3806b707593fc6ac9d7125e1b12d716ad40a2808","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I3806b707593fc6ac9d7125e1b12d716ad40a2808","subject":"Add kayobe overcloud service prechecks command","status":"MERGED","created":"2021-05-21 10:22:25.000000000","updated":"2021-06-09 11:25:11.000000000","submitted":"2021-06-09 11:23:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792571","meta_rev_id":"a2bb3f6cfac87edf914aab98052c9ab5e6ec231c","_number":792571,"virtual_id_number":792571,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 11:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-09 09:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-05-22 09:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-21 13:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-09 09:24:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 12:12:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-21 13:50:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-05-22 09:48:03.000000000","updated_by":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"reviewer":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"state":"REVIEWER"},{"updated":"2021-06-09 09:24:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d334708fdfa88af1bc413bf3f0bbb911ed241170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-21 10:22:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e60c0c6a4cd939404f7f27b5440dce96fdd36501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-21 10:26:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"874daed200853e00cebe7ad8a177a13f2d7d1cf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 12:12:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d1057456a07046f0b1716404aca7f7bc : SUCCESS in 3m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/feab47f324264a5eb83d2a48be40a2d6 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/264c36a416284bf88f9e5b8b733a8ad8 : SUCCESS in 8m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c1926da8dbd44bd99e7fc4b2d12bd7e : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/824b82ebeba1492381e9a4373b8959dd : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08ea1656e04744ccaa979c30e50b59b6 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b654d633a52b4fd08cb443a0f58a7733 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9c1a613976c45a48e0f4c2d8185ce92 : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/65c4c66c71b64b95a83981a694ce8016 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0414c8eaa15443f08df23f266414843e : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75db94f362fc4af88e4b890d8f3bfe47 : SUCCESS in 19m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e70250339f184c40a1eb68414153efde : SUCCESS in 58m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa4ad1cdd8454f23831ef93e7398fff1 : SUCCESS in 42m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7b401b2e490b4822b91a5185ef0950b7 : SUCCESS in 1h 14m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a8c7ddb71c5445dba2d0cff9c5dbf4fd : SUCCESS in 19m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0318c1b40d04ddc969d66db2884e993 : SUCCESS in 14m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e3928188a1b54604a89af1358b3a6eef : SUCCESS in 1h 43m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/18edead317d14d068556b2975fc567f5 : SUCCESS in 33m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/553a91cdf84e4c0d8b0a3e01f7e9c4aa : SUCCESS in 25m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ad50f4f481d342a4a24b8ea464e3ac1f : SUCCESS in 25m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e2101a206d9c4b2e93182c0e9d9ea9a4 : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5063a70403a14f7e8878b36cba121514 : SUCCESS in 10m 39s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"04334360a56d0463a13832d86a34ef0c3d57ce56","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-21 13:50:49.000000000","message":"Patch Set 2: Code-Review+2\n\nNot tested it myself, but LGTM","accounts_in_message":[],"_revision_number":2},{"id":"de93b6d6cdb8bb99b191992e765e432a02f8f2ca","author":{"_account_id":29543,"name":"Scott Solkhon","email":"scott.solkhon@gresearch.co.uk","username":"scott.solkhon"},"date":"2021-05-22 09:48:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"04c2278a7309579cddbbd0e8652cae5716ab6eec","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-09 09:24:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"97e1b6cbf3a34c4c2d8f49dfdda31a11a20ee1a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 09:24:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"eb4e7838ed8acebfcd66dca75ba3d755cbdfefc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:23:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7c3bb72f1def4d5386f3910719b05d56 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3a6a15100f742cab69fb01556928d6b : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55226b80da914ab4a6739987326b9c6c : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ef022b3473840dfa2cde23cfe2665a1 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b7bddd6b5b04014952156b40e3b0575 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6c4c0cf80c24324980a64dbae2b9fbe : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc300d95e1e545a0bd1439373f0fe966 : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68ca6fd53b3f4cb38a7bee5e908c8c84 : SUCCESS in 13m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7135687bb21846248be2577439189008 : SUCCESS in 18m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/62d56ca8a69746a7a4c398bbf64b9ca5 : SUCCESS in 1h 08m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb8f52037f154b9cbadd6212d79e60da : SUCCESS in 41m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/71808a95b446466d8c9b75e745730fe6 : SUCCESS in 1h 11m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8f32f166c2b1495397d7b6d7262cec39 : SUCCESS in 18m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f15823a644944f9db70ac6f51b79c671 : SUCCESS in 16m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e2e0b0d639ad418189f740df15e454b7 : SUCCESS in 1h 51m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/815819cab9ac4fbc94fed2573b4b5b8b : SUCCESS in 27m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7acbdcdaf4b418b9f5a6ca7941d10d5 : SUCCESS in 24m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4f759f77bbcb4014b76325ec3c1ec65c : SUCCESS in 25m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6b186fc5ab614597a37ed9951dd0a9ac : SUCCESS in 14m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46b008550a7141de8e343c7126a5812c : SUCCESS in 12m 51s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"be9265b8b7f2b9c255f0fd847094b5b77edb0989","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:23:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a2bb3f6cfac87edf914aab98052c9ab5e6ec231c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:25:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9fe50c2c0474e50907bd13c4a782077 : SUCCESS in 1m 20s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa444a22b0c04ffa819fb9f28790de7b : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b398e28dc57f546dc62c1eb27a306aecd134125e","revisions":{"7b99c0884b018f8db2a3d08e03c2a820c8ba6219":{"kind":"REWORK","_number":1,"created":"2021-05-21 10:22:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/792571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/792571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/792571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/792571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/792571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/792571/1"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-21 10:17:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-21 10:17:39.000000000","tz":60},"subject":"Add kayobe overcloud service prechecks command","message":"Add kayobe overcloud service prechecks command\n\nRuns \u0027kolla-ansible prechecks\u0027 without deploying services.\n\nChange-Id: I3806b707593fc6ac9d7125e1b12d716ad40a2808\nStory: 2008912\nTask: 42503\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b99c0884b018f8db2a3d08e03c2a820c8ba6219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7b99c0884b018f8db2a3d08e03c2a820c8ba6219"}]},"branch":"refs/heads/master"},"b398e28dc57f546dc62c1eb27a306aecd134125e":{"kind":"REWORK","_number":2,"created":"2021-05-21 10:26:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/792571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/792571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/792571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/792571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/792571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/792571/2"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-21 10:17:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-21 10:26:49.000000000","tz":60},"subject":"Add kayobe overcloud service prechecks command","message":"Add kayobe overcloud service prechecks command\n\nRuns \u0027kolla-ansible prechecks\u0027 without deploying services.\n\nChange-Id: I3806b707593fc6ac9d7125e1b12d716ad40a2808\nStory: 2008912\nTask: 42503\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b398e28dc57f546dc62c1eb27a306aecd134125e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b398e28dc57f546dc62c1eb27a306aecd134125e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
