)]}'
{"id":"openstack%2Fkayobe~792701","triplet_id":"openstack%2Fkayobe~master~Iaba35f237c75392586a4d86baef9bc7836c3fe1b","project":"openstack/kayobe","branch":"master","topic":"clock-sync","hashtags":[],"change_id":"Iaba35f237c75392586a4d86baef9bc7836c3fe1b","subject":"Force sync before checking clock status","status":"ABANDONED","created":"2021-05-21 21:02:58.000000000","updated":"2021-07-22 12:29:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f761b4c6a1f8d021f2b49119da8edeb812fb65fd","_number":792701,"virtual_id_number":792701,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-21 22:39:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-21 22:39:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8836411e17ed5e1a98f900a21c2459220065af4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-21 21:02:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37f140c425fec9ea20b928276a7721eb981a9b41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-21 22:39:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ee614ece5d6433ba6a922a3d4f420bc : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ffdae1eadce437395f1e0131c23acd1 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f991859898446b4b906d6a79d6a811c : SUCCESS in 8m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b094682b22b146d7ba43a7e228b0bdef : SUCCESS in 9m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3e1e88a85264786bc4ddfb46d826a13 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255a84ea64af4911a47471daeae838a2 : SUCCESS in 12m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3bc66d635c3548929efcf6cb93f6b4d8 : SUCCESS in 9m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6436ee7882d64a6f9709c5ee96cae667 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d32f7da707c45cda913b2baca24fb20 : SUCCESS in 15m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0c2a1cbffb2b410aace5944e0cc3c99d : SUCCESS in 52m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/323cb21b9a6b478080325430e2393db7 : SUCCESS in 46m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f3b50bf5c2a24e65a8d7f0d9c5f9ac2d : SUCCESS in 1h 13m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/25e234e554554c888d9835d25975bdc5 : FAILURE in 16m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75de0d4304364fcb8bafe92de14667be : FAILURE in 14m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e0596581ce9741d09f884e83058204bf : SUCCESS in 1h 35m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e73b5be4c2bf42af8a71540a0c67e404 : SUCCESS in 27m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f3c30e940654d0bb91654b1c18b98d7 : SUCCESS in 36m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b0cf93f3dbcf4c63b5b6d1c974bf94b2 : SUCCESS in 27m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/181f1185fa054c8bab8c2e904d002d82 : SUCCESS in 12m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5611258d99694beaaad66a6b57b19c36 : SUCCESS in 10m 40s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"f761b4c6a1f8d021f2b49119da8edeb812fb65fd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-22 12:29:14.000000000","message":"Abandoned\n\nThis issue is not happening as often as before","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1395ead09c7309fec1ae2a368cd3a386e2c65a24","revisions":{"1395ead09c7309fec1ae2a368cd3a386e2c65a24":{"kind":"REWORK","_number":1,"created":"2021-05-21 21:02:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/01/792701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/792701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/792701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/792701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/792701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/792701/1"}}},"commit":{"parents":[{"commit":"6b65b29395b798eed735661a03146798569639d3","subject":"Merge \"Replace custom serial loop with throttle\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6b65b29395b798eed735661a03146798569639d3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-21 21:01:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-21 21:01:02.000000000","tz":120},"subject":"Force sync before checking clock status","message":"Force sync before checking clock status\n\nWe are seeing occasional failures of the test_ntp_clock_synchronized\ntest in the kayobe-overcloud-host-configure-ubuntu-focal job:\n\n______________________ test_ntp_clock_synchronized[local] ______________________\nhost \u003d \u003ctestinfra.host.Host local\u003e\n    def test_ntp_clock_synchronized(host):\n        # Tests that the clock is synchronized\n        status_output \u003d host.check_output(\"timedatectl status\")\n\u003e       assert \"synchronized: yes\" in status_output\nE       AssertionError: assert \u0027synchronized: yes\u0027 in \u0027               Local time: Fri 2021-05-21 06:41:41 HST  \\n           Universal time: Fri 2021-05-21 16:41:41 UTC  \\n ...  \\n              NTP service: active                       \\n          RTC in local TZ: no                           \u0027\nsrc/opendev.org/openstack/kayobe/playbooks/kayobe-overcloud-host-configure-base/tests/test_overcloud_host_configure.py:171: AssertionError\n\nForcing a clock sync before checking clock status should prevent this\nissue.\n\nChange-Id: Iaba35f237c75392586a4d86baef9bc7836c3fe1b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1395ead09c7309fec1ae2a368cd3a386e2c65a24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1395ead09c7309fec1ae2a368cd3a386e2c65a24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
