)]}'
{"id":"openstack%2Fkayobe~792834","triplet_id":"openstack%2Fkayobe~master~Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7","project":"openstack/kayobe","branch":"master","topic":"merge-configs","hashtags":[],"change_id":"Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7","subject":"Make kolla custom config globs configurable","status":"ABANDONED","created":"2021-05-24 17:26:00.000000000","updated":"2021-07-28 18:43:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c4547c190c69eb275a05e5ddc78b967e93133128","_number":792834,"virtual_id_number":792834,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-07-08 17:53:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 17:43:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-12 08:58:17.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"962b7f0cc4d6de4ad8803e6dd3f795b2bc04341b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-24 17:26:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"993def6d236896a26b3431b310d4fcfba9fb4e96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 17:43:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c7fa43e5a29f44c59d59542eae6eca4d : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/04897a34be974aa1adb4d3997f7bf5b6 : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ecb569579694a9fa20c2ee7ffee8dbd : FAILURE in 7m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9041c08a6d294ccb8fb320a95a2778c5 : FAILURE in 11m 35s","accounts_in_message":[],"_revision_number":1},{"id":"5740576dc880016b852829ed0e19e9fbaeebe1ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-25 08:41:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5d0408d453312de967356bbfb04d7bf9218021b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-25 08:45:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"25fcdafc61631fe96327245294fd1651134f79dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 09:08:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/45a52690d5f04d249b2854cd41af87d5 : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b534899fdec542a8982324506efe1594 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/48222534c97f4f04a0b4a000ed4a8af1 : FAILURE in 7m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ad9ccc12dad34c32a9db65fef44644c1 : FAILURE in 18m 31s","accounts_in_message":[],"_revision_number":3},{"id":"4a7fd34ecc8e8a4295983b8c64a6168666efeb2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-25 10:27:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"012b784e129519368395099896ffb7ee8839469f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 10:45:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2270ef0e46c64577b3fb76a3fac03948 : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fec78a73975a4dc69f06df403cf6fbe0 : SUCCESS in 14m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/225e928a423f491e96fc29215ab56c5b : FAILURE in 9m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/71cbafb63fc8484d94980d9a6294e437 : FAILURE in 17m 05s","accounts_in_message":[],"_revision_number":4},{"id":"4c6af59b3509b674e873af06b72806e533b7580f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-25 10:55:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"510b86044803c91881a733f6f7d92facd622ca37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 11:17:17.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f88b62018c03414d89a6214af7574245 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b44b4b9170414d0d93cb224866e6cbca : SUCCESS in 8m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e623835a3d644c6bae664f4116d61e03 : SUCCESS in 14m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5a391728f8944928999648f7e0a277f3 : FAILURE in 17m 19s","accounts_in_message":[],"_revision_number":5},{"id":"f9714ec150bd43fe97cbdb5c7c9620a06549b8a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-25 12:45:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5924485354c147a1e133c0560e44844398cebf05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 13:11:06.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/471a345d59e8483d9bdd148557dab3fc : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4bed20a5825949e3873b8916e22eb49c : FAILURE in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/81670f1b4e5c40828078dcf1eec00587 : SUCCESS in 20m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fbe4cc0280a442fb9de62a2d4eabc4a6 : FAILURE in 18m 58s","accounts_in_message":[],"_revision_number":6},{"id":"df3b54185dce0d4c219839acc8572ab0cd527d05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-25 13:21:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9f033f527326af4bd7e3ffacdae787cee0df42c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 13:49:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b683c5302f545c2a7359916a04ac314 : SUCCESS in 10m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ada3afe546fc4583b0593d7ebc1dc664 : FAILURE in 9m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe769281485041e68cc37df0bb9bb371 : SUCCESS in 22m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/869960fa3071477e9bcee5fc335a2c2f : POST_FAILURE in 23m 55s","accounts_in_message":[],"_revision_number":7},{"id":"b9a045648deefc681baf54c0e6e32c735cd712c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-25 15:17:46.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f833f65b0dc1e0037ddd87704af3b4706a1d1748","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-25 15:33:48.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"003cba089dcf71eb5f1e941e7cccc761d1906bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:03:52.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2535a5a084814573a3b0bb3051d5e3e6 : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9fd81acdacc349f4a1166c3eef3b3480 : FAILURE in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c3cbb880a304b73aded2ec1144c60d3 : SUCCESS in 23m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/126965151434465caacafaf1142b6f49 : FAILURE in 24m 08s","accounts_in_message":[],"_revision_number":9},{"id":"88f359a867ff1a1e6708b4c1e43d9702f2e28dc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-26 08:56:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"eeb714eb624593458f92b96ccd4dfb635f425e25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 09:27:48.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7df2a47782b747aa9c45592d83f4df9c : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a25be0b5fe44ff2bd12c9df5801f4f8 : FAILURE in 10m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dfad0d494bba4dd4a3ce2baf5ef91c58 : SUCCESS in 19m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9a11d686ac2c4cb4a44d4242f01aca7e : FAILURE in 18m 40s","accounts_in_message":[],"_revision_number":10},{"id":"7d85346c2649ba1088c2daa2badd1ae776b0ee90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-26 10:19:40.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"a7aa434b18bd0f9560ced0f0b87392384fc7349b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 11:02:18.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d7408cb4ec1452d89abb64e301ee6d2 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cacf993af7d84480aa70e3fd7b49db36 : FAILURE in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5957f042157849cf8ec68eabe07b57be : SUCCESS in 19m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7d1da91ed1094536a04a59744060f38e : FAILURE in 39m 11s","accounts_in_message":[],"_revision_number":11},{"id":"af7cfe738d235045a15dd787ba7a10b0c67cbef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-26 13:45:08.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"f14703d4277c832c18e5750d027a7295f521a9fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 14:39:22.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f9ad473558ea4d56acea5b77b5e28e8f : SUCCESS in 12m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3aa10a26ca4f4b1faa24c4e5bc6bf67c : FAILURE in 10m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0695e964508a4a4facddaea12029a5e8 : SUCCESS in 14m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3928576c3e504340a2b99c257e29260f : FAILURE in 53m 26s","accounts_in_message":[],"_revision_number":12},{"id":"71a375229cb77afd52e36a86ca0a1bbc95d06270","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-07-08 16:53:09.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"6de0636ab2c7473d1407c6526db766ef0d14af04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-07-08 16:56:06.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"1f791821275f26102408b0d253cba4de73ce8b2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 17:53:31.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d154b7acd4694f0d9859d8f9a4c86801 : SUCCESS in 7m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1edcfe777dce4c65972a4e4a450c21df : FAILURE in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0867a320f76425aabeaa8f3aff60a29 : SUCCESS in 13m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8fa2eb906f914876b4668ed8acc8b653 : FAILURE in 52m 13s","accounts_in_message":[],"_revision_number":14},{"id":"cd8e904a6f4804d019838856bb726da50852136c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-07-23 14:44:31.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"bd734574e885f13399ee963bb3e7ccf08eb1718b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 15:54:25.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/26922f05524241cdb2af2ae53aab9dfc : SUCCESS in 7m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/194d7866f8754dfdba46765f72955b4d : FAILURE in 12m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/32705e835ea64680be6792d53d23cfe3 : SUCCESS in 13m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/dc94f698477b40fab2bfa3c0fe86cfe1 : SUCCESS in 1h 07m 49s","accounts_in_message":[],"_revision_number":14},{"id":"c4547c190c69eb275a05e5ddc78b967e93133128","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-07-28 18:43:58.000000000","message":"Abandoned\n\nAbandoning in favour of rolling in changes to https://review.opendev.org/c/openstack/kayobe/+/782749/6","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"989c8ee4d00ce82bd64caacd6edb0dac24163423","revisions":{"ded0e363586eb831c8e43af4cdc7c58cb01c5632":{"kind":"REWORK","_number":1,"created":"2021-05-24 17:26:00.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/1"}}},"commit":{"parents":[{"commit":"f2f2fcb9a3212604d564d94c9130cd47bb287800","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2f2fcb9a3212604d564d94c9130cd47bb287800"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-24 17:25:10.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ded0e363586eb831c8e43af4cdc7c58cb01c5632"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ded0e363586eb831c8e43af4cdc7c58cb01c5632"}]},"branch":"refs/heads/master"},"3d7e154c329f7f622e38d27e9138018419b67d79":{"kind":"REWORK","_number":2,"created":"2021-05-25 08:41:42.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/2"}}},"commit":{"parents":[{"commit":"f2f2fcb9a3212604d564d94c9130cd47bb287800","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2f2fcb9a3212604d564d94c9130cd47bb287800"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-25 08:04:47.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d7e154c329f7f622e38d27e9138018419b67d79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d7e154c329f7f622e38d27e9138018419b67d79"}]},"branch":"refs/heads/master"},"74f5a8de1b176573425c35c3f77dfa3b607c67ed":{"kind":"REWORK","_number":3,"created":"2021-05-25 08:45:19.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/3"}}},"commit":{"parents":[{"commit":"f2f2fcb9a3212604d564d94c9130cd47bb287800","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2f2fcb9a3212604d564d94c9130cd47bb287800"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-25 08:45:01.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74f5a8de1b176573425c35c3f77dfa3b607c67ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74f5a8de1b176573425c35c3f77dfa3b607c67ed"}]},"branch":"refs/heads/master"},"5ea651215224b799ea6b65556b20c01c9f8bf94e":{"kind":"REWORK","_number":4,"created":"2021-05-25 10:27:13.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/4"}}},"commit":{"parents":[{"commit":"f2f2fcb9a3212604d564d94c9130cd47bb287800","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2f2fcb9a3212604d564d94c9130cd47bb287800"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-25 10:26:36.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ea651215224b799ea6b65556b20c01c9f8bf94e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ea651215224b799ea6b65556b20c01c9f8bf94e"}]},"branch":"refs/heads/master"},"fb1e0c9b63fbec41ac3ca893ceabc7a5675697f1":{"kind":"REWORK","_number":5,"created":"2021-05-25 10:55:53.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/5"}}},"commit":{"parents":[{"commit":"f2f2fcb9a3212604d564d94c9130cd47bb287800","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2f2fcb9a3212604d564d94c9130cd47bb287800"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-25 10:55:35.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb1e0c9b63fbec41ac3ca893ceabc7a5675697f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb1e0c9b63fbec41ac3ca893ceabc7a5675697f1"}]},"branch":"refs/heads/master"},"ca784ea1f130b7af5bba944b86bc75e2fddf0ff7":{"kind":"REWORK","_number":6,"created":"2021-05-25 12:45:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/6"}}},"commit":{"parents":[{"commit":"f2f2fcb9a3212604d564d94c9130cd47bb287800","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2f2fcb9a3212604d564d94c9130cd47bb287800"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-25 12:44:20.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca784ea1f130b7af5bba944b86bc75e2fddf0ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca784ea1f130b7af5bba944b86bc75e2fddf0ff7"}]},"branch":"refs/heads/master"},"32e153cd025a174aef0174632f6c0140174e6a7e":{"kind":"REWORK","_number":7,"created":"2021-05-25 13:21:28.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/7"}}},"commit":{"parents":[{"commit":"f2f2fcb9a3212604d564d94c9130cd47bb287800","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2f2fcb9a3212604d564d94c9130cd47bb287800"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-25 13:19:33.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32e153cd025a174aef0174632f6c0140174e6a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32e153cd025a174aef0174632f6c0140174e6a7e"}]},"branch":"refs/heads/master"},"19d164bc4c132505b2570130d562b2dabca0a7de":{"kind":"REWORK","_number":8,"created":"2021-05-25 15:17:46.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/8"}}},"commit":{"parents":[{"commit":"f2f2fcb9a3212604d564d94c9130cd47bb287800","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2f2fcb9a3212604d564d94c9130cd47bb287800"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-25 15:17:21.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19d164bc4c132505b2570130d562b2dabca0a7de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/19d164bc4c132505b2570130d562b2dabca0a7de"}]},"branch":"refs/heads/master"},"90cec92e53b8e78045d552e0c676cf1a81449d24":{"kind":"REWORK","_number":9,"created":"2021-05-25 15:33:48.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/9"}}},"commit":{"parents":[{"commit":"50b94fa903f604a43b1f75da43a4519aefef097a","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50b94fa903f604a43b1f75da43a4519aefef097a"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-25 15:33:32.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90cec92e53b8e78045d552e0c676cf1a81449d24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/90cec92e53b8e78045d552e0c676cf1a81449d24"}]},"branch":"refs/heads/master"},"162124367e9558252a154c4a540c45a1a49e1734":{"kind":"REWORK","_number":10,"created":"2021-05-26 08:56:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/10"}}},"commit":{"parents":[{"commit":"901b426a88aa130b6aa013204fcd58018d4e306d","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/901b426a88aa130b6aa013204fcd58018d4e306d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-26 08:55:42.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/162124367e9558252a154c4a540c45a1a49e1734"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/162124367e9558252a154c4a540c45a1a49e1734"}]},"branch":"refs/heads/master"},"1edfa18608c8c460763f6fbbe8ac1a24e5a0f908":{"kind":"REWORK","_number":11,"created":"2021-05-26 10:19:40.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/11"}}},"commit":{"parents":[{"commit":"901b426a88aa130b6aa013204fcd58018d4e306d","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/901b426a88aa130b6aa013204fcd58018d4e306d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-26 10:18:35.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1edfa18608c8c460763f6fbbe8ac1a24e5a0f908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1edfa18608c8c460763f6fbbe8ac1a24e5a0f908"}]},"branch":"refs/heads/master"},"b668a595409c015f93d993282128ccb1451e08fc":{"kind":"REWORK","_number":12,"created":"2021-05-26 13:45:08.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/12"}}},"commit":{"parents":[{"commit":"901b426a88aa130b6aa013204fcd58018d4e306d","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/901b426a88aa130b6aa013204fcd58018d4e306d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-26 13:44:47.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b668a595409c015f93d993282128ccb1451e08fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b668a595409c015f93d993282128ccb1451e08fc"}]},"branch":"refs/heads/master"},"6e20ffb6f814d63a4d543b2a3e37bcc6d1188131":{"kind":"REWORK","_number":13,"created":"2021-07-08 16:53:09.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/13"}}},"commit":{"parents":[{"commit":"901b426a88aa130b6aa013204fcd58018d4e306d","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/901b426a88aa130b6aa013204fcd58018d4e306d"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-08 16:52:50.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e20ffb6f814d63a4d543b2a3e37bcc6d1188131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e20ffb6f814d63a4d543b2a3e37bcc6d1188131"}]},"branch":"refs/heads/master"},"989c8ee4d00ce82bd64caacd6edb0dac24163423":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-07-08 16:56:06.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/792834/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/792834/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/792834/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/792834/14"}}},"commit":{"parents":[{"commit":"40968f40e936d5d2fa459526714c05e697b5a7b7","subject":"WIP: Use merge_configs and merge_yaml to generate custom config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40968f40e936d5d2fa459526714c05e697b5a7b7"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-20 14:12:58.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-08 16:55:52.000000000","tz":60},"subject":"Make kolla custom config globs configurable","message":"Make kolla custom config globs configurable\n\nThis also seems quicker than using the find module in a loop.\n\nChange-Id: Ie5a65b6461e3693e47c4b7811b69ef89cf7f74a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/989c8ee4d00ce82bd64caacd6edb0dac24163423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/989c8ee4d00ce82bd64caacd6edb0dac24163423"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
