)]}'
{"id":"openstack%2Fkayobe~793059","triplet_id":"openstack%2Fkayobe~master~I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9","subject":"docs: improve automated seed development info","status":"MERGED","created":"2021-05-25 20:28:01.000000000","updated":"2021-06-09 11:21:33.000000000","submitted":"2021-06-09 11:19:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793059","meta_rev_id":"416be549100e1a315212cb3177dd6cad22464942","_number":793059,"virtual_id_number":793059,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 11:19:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-09 09:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-28 09:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-09 09:23:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-25 22:18:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-26 08:35:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-09 09:23:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"12c2481b4015bd7e01c27b9ff01d69c5b8944626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-25 20:28:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"390ced76d795966728a16667c06b15558d7ab869","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-25 20:29:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cdbe540903a47af7d9ce2a77b5cac88d0be32330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 22:18:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd4abc81d8c24e798a381b65036d45c0 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d763713d6e234578809ffdb0d68069c3 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/440a6a215c494ab8b7f7c98dbec859f8 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2aac419ce4f040dba69e8fd9c6bff037 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b268c930f944f8a9181a3f01197dea3 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f312dfd554e4f418e6cb43f72e497fe : FAILURE in 6m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca9ee8e22c5e4a0a8e0c4d82e64292b8 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/38778de61c3a400ebf1411857baf0298 : SUCCESS in 10m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9356f51cde594db282dabf5a31e31f09 : SUCCESS in 31m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/57726d227a674ce59da86fd2894e6358 : FAILURE in 42m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb304a1ecf044886a04450db5aa20544 : SUCCESS in 53m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c4ffc9abee86461aafd5f2915816cd20 : FAILURE in 41m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f464fae4e9474b45b985a1c913eae007 : SUCCESS in 15m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a91c875ee0ee49a5a039ab8d9a419c1f : SUCCESS in 9m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3261bf755a0145bda5e005ac2da50ad0 : FAILURE in 1h 43m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/84cb15c1fad849ac913c2abaf65f61bb : SUCCESS in 45m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47e418d2318c4321b97144681286e2e1 : SUCCESS in 23m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f967ecacd8f441eba9d6150a701d1f73 : SUCCESS in 23m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e211ab2730e24fbe9d7b257504b1bd33 : FAILURE in 12m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08539cef782045b1ba75bd86e9867664 : SUCCESS in 8m 04s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"46be58ba51912b8f4c2ecd9a515932c692961216","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-26 08:35:27.000000000","message":"Patch Set 2: Code-Review-1\n\ndoc/source/contributor/automated.rst:188: WARNING: undefined label: contributor-automated-seed-vm","accounts_in_message":[],"_revision_number":2},{"id":"639dbe49753ac2643e8231e0799dc579bd028d55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-27 14:10:42.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a1d0ebe84cf1d7bc4a40816c69a9b9bc03870f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:32:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6307ae34e7354df1b9f7b71a3b3a97b6 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6a85b1d2b94bf0aaf44a4f1e623f08 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50a7e95e1add47688e2a986ba493d7db : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16ac33e920ff497382b11bbe49687134 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00d021cbe1784988bf629b4405665c1a : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9515d17115c4aa1be9fecb71eb3c51f : SUCCESS in 9m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a9dc8aadb63f43888e72014b00c58a6d : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d78edcd10b504dba8a6eb270641009b0 : SUCCESS in 10m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f052970d37844109d3a9dfd12230a8d : SUCCESS in 17m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ff3cbaff257a4fb98b42f6e750167c23 : SUCCESS in 54m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c3b5706adba47ea9f9a0a2d1c105d0c : SUCCESS in 41m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e0e26eb08b00447d8e7b5b0f4ef01092 : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0ae20e805b654558b9eedcfc83aed89d : SUCCESS in 15m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fffad7d2e104cacab77f29c4af2b4a4 : FAILURE in 15m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5c0970dfad16436ea7ffc99455bca199 : FAILURE in 1h 15m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b2428a4c2d0e4dc79ae133802f724f10 : SUCCESS in 53m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9b6896465454e3699981b893de8ffd2 : SUCCESS in 26m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cf68750cf7534d74a333bfaac8baff2f : SUCCESS in 27m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7a27ea5b837747e69b71c0dc89230567 : SUCCESS in 14m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d24893ca1b354778a24d12b596947819 : SUCCESS in 12m 29s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":3},{"id":"74575016c2c49068d71806e0fc4983d55406dca5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-28 07:30:59.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"11cb92db7d4acd73b12c972695d7939931c76ba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:07:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f1f7510cfe64750abdac4955fce5e29 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/976032a3bd6a4e29a8362ecb9816c2c2 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91961d55b000445392905d7f87ebd1e3 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84032b1398f441cb9ebc39af920fb786 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5c14a3a308342f698530eeb18de20d6 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92a603e5f5b244cc97f93dd29165efde : SUCCESS in 11m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c807d30ba454d4687b4fadc1a98507a : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ff19707e771481784ca3e86b86277e6 : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86588c4a270a4271b72fd602124481cd : SUCCESS in 19m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9df64c5bdec64e028fed6e0d1c508725 : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6f23cb273924847bff61befc2842ab8 : SUCCESS in 35m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4be809669fdf48d8b6f223d1f3818b74 : SUCCESS in 1h 07m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cad4bf2bb3b44afbbfbfbffbdf9446e6 : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/877faee3d058478ea5c2ebc83a254728 : SUCCESS in 13m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8e49c3009b4648db9ccc3b5841a0fc1c : SUCCESS in 1h 35m 06s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e1dcd62807e1467d8d6243a3a3dc8fbe : SUCCESS in 28m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/852430d53e884e2a85460057a0554080 : SUCCESS in 26m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cc9b1f3eb84747409f870868db184e7c : SUCCESS in 24m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d307d57523364cf1acd52936b583204a : SUCCESS in 11m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f746eb3d84f0441c9d93511d84f6e55e : SUCCESS in 10m 20s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":3},{"id":"8ac8333922820c098425f0ee334ce98788b8434f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-28 09:33:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8bc971154b15d16a836dab7979a2079d225164b5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-09 09:23:18.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4b0194ec118e251d1c264280b75be4cbbc514814","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 09:23:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"347afa9bcc7ae0f2d0bf9195241f547d84fb848d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:19:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9002ef96852f44cb953533fe7d8e492e : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/415699985ee6452caa44bf83404a4cfe : SUCCESS in 9m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ee98d590ad942969b104d66255f9281 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab3a21ffe924023a3280a98ae38348c : SUCCESS in 15m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/df1f9076c3a24629953ad43401a999ca : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/841a9266f22340ee8cdb0f17e07b5ec3 : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fad5e3ca82af49908683a3bd050cf367 : SUCCESS in 21m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d6cf10a79e4c4ba8a7767d7ab41b271d : SUCCESS in 1h 11m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0102bf1792c4443abdd7000cfa8cdc97 : SUCCESS in 50m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/319004eb7d74464d9c881e4583278fa2 : SUCCESS in 1h 16m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1faa2fc8aaa546c0be31dc3c3470f8d9 : SUCCESS in 20m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b578005105cb4994b3b50c0bba4b839a : SUCCESS in 15m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/90696f2b7a9a47c0951cf885ecca0f2d : SUCCESS in 1h 53m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/964ce87acb2347ed9b82621f030d2c55 : SUCCESS in 30m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bde1e3b0b41c45259779ff50cda8ce5e : SUCCESS in 25m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/061ab83b413749b3a0c4e791cb90ddfa : SUCCESS in 28m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2326633485e248228f98d5f9de152a09 : SUCCESS in 14m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fc3ff79a60b4472b6b9517bb6ce6146 : SUCCESS in 12m 40s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"1d96c3d780e01b649cb73e45fe7020bdc693beda","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:19:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"416be549100e1a315212cb3177dd6cad22464942","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 11:21:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d94633905fc469cae9b535c66173cae : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d2e3f6f1aec2c1ed301144ed26872394799346fb","revisions":{"d8e008382469aee62f7b2bed981bad8562e2677e":{"kind":"REWORK","_number":1,"created":"2021-05-25 20:28:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/793059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/793059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/793059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/793059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/793059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/793059/1"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-25 16:14:06.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-25 20:27:54.000000000","tz":60},"subject":"docs: improve automated seed development info","message":"docs: improve automated seed development info\n\n* add \u0027bridge_type: linuxbridge\u0027 to tenks configuration to avoid\n  dependency on OVS\n* extend seed development environment testing to include overcloud\n  provisioning and deployment\n* remove seed hypervisor and seed VM environments. These are very\n  stale, and largely replaced by a-universe-from-nothing. Add a link to\n  that workshop on the same page\n\nChange-Id: I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8e008382469aee62f7b2bed981bad8562e2677e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8e008382469aee62f7b2bed981bad8562e2677e"}]},"branch":"refs/heads/master"},"ba64e27e3d1e7d5c243827e14979801fbde4f9f9":{"kind":"REWORK","_number":2,"created":"2021-05-25 20:29:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/793059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/793059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/793059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/793059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/793059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/793059/2"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-25 16:14:06.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-25 20:28:50.000000000","tz":60},"subject":"docs: improve automated seed development info","message":"docs: improve automated seed development info\n\n* add \u0027bridge_type: linuxbridge\u0027 to tenks configuration to avoid\n  dependency on OVS\n* extend seed development environment testing to include overcloud\n  provisioning and deployment\n* remove seed hypervisor and seed VM environments. These are very\n  stale, and largely replaced by a-universe-from-nothing. Add a link to\n  that workshop on the same page\n\nChange-Id: I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba64e27e3d1e7d5c243827e14979801fbde4f9f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ba64e27e3d1e7d5c243827e14979801fbde4f9f9"}]},"branch":"refs/heads/master"},"d2e3f6f1aec2c1ed301144ed26872394799346fb":{"kind":"REWORK","_number":3,"created":"2021-05-27 14:10:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/59/793059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/793059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/793059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/793059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/793059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/793059/3"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-25 16:14:06.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-27 14:10:31.000000000","tz":60},"subject":"docs: improve automated seed development info","message":"docs: improve automated seed development info\n\n* add \u0027bridge_type: linuxbridge\u0027 to tenks configuration to avoid\n  dependency on OVS\n* extend seed development environment testing to include overcloud\n  provisioning and deployment\n* remove seed hypervisor and seed VM environments. These are very\n  stale, and largely replaced by a-universe-from-nothing. Add a link to\n  that workshop on the same page\n\nChange-Id: I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2e3f6f1aec2c1ed301144ed26872394799346fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2e3f6f1aec2c1ed301144ed26872394799346fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
