)]}'
{"id":"openstack%2Fkayobe~793390","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I5bc8197143ce763cdaf8a9ac0b782b524d5df258","project":"openstack/kayobe","branch":"stable/wallaby","topic":"edk2-ovmf-workaround-stable/wallaby","hashtags":[],"change_id":"I5bc8197143ce763cdaf8a9ac0b782b524d5df258","subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","status":"MERGED","created":"2021-05-27 11:57:12.000000000","updated":"2021-05-28 09:24:39.000000000","submitted":"2021-05-28 09:23:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793390-edk2-ovmf-workaround-stable/wallaby","meta_rev_id":"3d61eeaa7ed2f5099d5722b426893e9a375cfdeb","_number":793390,"virtual_id_number":793390,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-28 09:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-05-28 07:32:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-05-27 20:35:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-05-28 07:32:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-28 07:31:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 13:58:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 14:01:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-05-27 20:35:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-05-28 07:32:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e1eb78adcba1efd8c3938639669a0d9f613fd558","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-27 11:57:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d16c648e442dec5196ff546534f15bda1f63414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 13:58:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d15fa54852ea4b5c86a36da3e4ce2448 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29fcb7bb5ff84776a8df46e83e89241b : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/285ff2139d18434fa041e2d924dee34e : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3478c98a964d41739127030564edb395 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b37425bc214742758aaa4eeb7499343e : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ad15cb87fd46edba8a39d22abbd9b0 : SUCCESS in 8m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f0a808d4e4194b2188e412daa265a092 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c83efd57c1554c22be01a44e213534eb : SUCCESS in 11m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1cf3ea19a49c46a9b046bcbc522d7818 : SUCCESS in 20m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6454ad9ce1074ece9c61f9e111239a36 : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a5045f84bb14d4692ff124c343cdaf8 : FAILURE in 42m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b2f97ff863ee44df979c74cdbe19248b : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b9a25c8cdbcf4ecebbc66dcfc1e7056b : SUCCESS in 17m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8757f5b320bd4ac18c2f61f0f7cec815 : SUCCESS in 12m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7825b97c5622435284db9e620554a5ae : SUCCESS in 1h 56m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/acf8fbcd5c24463488166d76212d2ddc : SUCCESS in 27m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f98924231c9f4b859d2f2026a9cac715 : SUCCESS in 28m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/78b06b2aefe14ac1af58721896629aa6 : SUCCESS in 25m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3703d6dccda143bf9890f38ad69a499c : SUCCESS in 13m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a6e554f8c3a4f7c91df049204444207 : SUCCESS in 9m 43s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"f5c6edbde25c44d4ffdfc0c90c3c0bffb87ee8cf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-27 14:01:23.000000000","message":"Patch Set 1:\n\nrecheck\n\nFailed in \u0027Task Ensure qemu-img is available for Tenks\u0027","accounts_in_message":[],"_revision_number":1},{"id":"7c29f3911bc70d812541008449541b7afc37784a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:03:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f872c85b19645fb8691b1ed077e606a : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dc59515f40a46e2b6929af702dfeb9e : SUCCESS in 16m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d9d33bf4be341e9ba6fc56ab3f24b41 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f1ffee5ddf7477e8246629d98b3f4a9 : SUCCESS in 15m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b7d2eb1520a41fb8f0ac5ef035a928c : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9320eaad3668496880a88b423e4bd11f : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c80dd3981ab4ec68ad45d0c6ce2dc96 : SUCCESS in 8m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/95141e627936486fbfbdaa3512c63280 : SUCCESS in 13m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d92e1424b5084a2daa58b557bf5c6fd3 : SUCCESS in 19m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/312ee8d8859142e28f113b27f1aa163c : SUCCESS in 1h 03m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d034369370e4120a918323fdade78cd : SUCCESS in 48m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5f38de5255b545138660dad3480fef0c : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b68a175f09974c94a2b204b82166334f : SUCCESS in 14m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c4dc90ff4d54b2591227cd369538337 : SUCCESS in 14m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0d0e27d936ad4c65ba079a303e0da026 : SUCCESS in 1h 55m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/40e1704406e84900ad69086aedb78195 : SUCCESS in 28m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40a380d68c3048eb9fa752fd76a1c4e9 : SUCCESS in 25m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e97e3b0977e4143849a0488b42e5d55 : SUCCESS in 22m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bcd58cc0765f45fcbc9d517fa4f8bfcc : SUCCESS in 13m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c275de5cad540b2878f49cddbf2c645 : SUCCESS in 11m 14s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"d8d147f166e0df4c88f4de25f4bf894280fb5819","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-27 20:35:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dfc066f40be169436cb4518cdcb1a458cf78426f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-28 07:31:46.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"2b3283670e2d9ed0a93a1dc31552e7ce1963faac","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-28 07:32:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f80b2fbd5579e405cd160b81c8a500eb4e2e5994","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 07:33:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3ac5bc31c7da519754e8278ac579ba76258edca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:23:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d7d1dd1de94363811188ec104c9d55 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00a0425020e444768eedf660ab1ed6c4 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fab1b0504c04ccb972f2c3a04181940 : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ed1f31bc544215a63d29c5ddb72810 : SUCCESS in 18m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/143754c24c684b35b66a7f54d3fc0a3d : SUCCESS in 7m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3a829185b11b4e2482e2d289a545da4a : SUCCESS in 9m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8fae5bf7b9134eddbfc93fcb9716cfd1 : SUCCESS in 16m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8288d6b1bb5c4c768724f0feee2c804a : SUCCESS in 55m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed76e3915406452a96e09e8154d29b0e : SUCCESS in 43m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d472455aae6a4e0caae155195a111e74 : SUCCESS in 56m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f40ca452c84d4120b670e6cc25b11e6b : SUCCESS in 20m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5b240934e5f420eb55a54817e4d3656 : SUCCESS in 12m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/430321c8b777490db0376b6c25220569 : SUCCESS in 1h 49m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6f4f5ff79de84cd6b5351d75acf85e2c : SUCCESS in 27m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d22e14c3cc314b9d8dd434f4efae6e44 : SUCCESS in 24m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cf80dd8cd660449299c7ec4176f4218b : SUCCESS in 24m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ec0420215a3e4c5bad3b5bf30df9019c : SUCCESS in 9m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c74e46da7164b41b173be64854aa576 : SUCCESS in 10m 10s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"cc6164360a4ae372b71344d3101ec18e3943654c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:23:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3d61eeaa7ed2f5099d5722b426893e9a375cfdeb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:24:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb5ad59e3604f3d812b7a1294627855 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c5e5d697fed97168772fd8f98f3d780726785a44","revisions":{"c5e5d697fed97168772fd8f98f3d780726785a44":{"kind":"REWORK","_number":1,"created":"2021-05-27 11:57:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/90/793390/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/793390/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/793390/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/793390/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/793390/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/793390/1"}}},"commit":{"parents":[{"commit":"a9a2f382002b8e202590ccb761a144ffdcb20fc7","subject":"Replace custom serial loop with throttle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9a2f382002b8e202590ccb761a144ffdcb20fc7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-27 08:35:52.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-05-27 11:57:01.000000000","tz":120},"subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","message":"CI: Fix libvirt errors caused by latest edk2-ovmf\n\nSome CentOS Stream CI jobs started to fail when defining VMs with:\n\n    libvirt.libvirtError: internal error: unknown feature amd-sev-es\n\nThis was introduced by edk2-ovmf-20200602gitca407c7246bf-5.el8.noarch.\nThis change excludes this package version in Zuul jobs, which fixes the\nkayobe-seed-vm-centos8s job.\n\nOvercloud centos8s jobs were also broken by the nova-libvirt Kolla image\ncontaining this package, which is used by Tenks. Updated images using\nthe previous version of edk2-ovmf [1][2] are now available from quay.io.\n\nThis also needs to be fixed for Kayobe and Tenks in general, unless a\nnew package is released soon.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/792999\n[2] https://review.opendev.org/c/openstack/kolla/+/793174\n\nChange-Id: I5bc8197143ce763cdaf8a9ac0b782b524d5df258\n(cherry picked from commit 1e14fa3a24fa64b84cc993c985e696a9cba78dea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5e5d697fed97168772fd8f98f3d780726785a44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5e5d697fed97168772fd8f98f3d780726785a44"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
