)]}'
{"id":"openstack%2Fkayobe~793697","triplet_id":"openstack%2Fkayobe~master~If6c34db3711d465d5c0a2aab225c6dcd3ca067e3","project":"openstack/kayobe","branch":"master","topic":"sync-tls-settings","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2026-06-04 15:22:09.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-06-04 15:22:09.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2026-06-04 15:22:09.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-02-10 12:38:41.000000000","reason":"Change was marked work in progress"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-12-01 11:57:55.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"If6c34db3711d465d5c0a2aab225c6dcd3ca067e3","subject":"Sync TLS configuration with Kolla-Ansible","status":"NEW","created":"2021-05-30 18:31:57.000000000","updated":"2026-06-04 17:26:32.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":79,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c0db75201b22cf7b93bf291c6185a77a46478a65","_number":793697,"virtual_id_number":793697,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-04 17:26:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-06-04 15:31:29.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-30 19:59:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-30 21:02:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2021-05-30 21:02:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-05-30 21:02:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2021-06-03 10:25:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-06-16 08:53:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-08-17 12:16:47.000000000","updated_by":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"reviewer":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"state":"REVIEWER"},{"updated":"2021-10-01 05:36:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-04-05 17:52:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REMOVED"},{"updated":"2022-04-05 17:52:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REMOVED"},{"updated":"2022-04-05 17:52:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"state":"REMOVED"},{"updated":"2022-04-05 17:52:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REMOVED"},{"updated":"2023-08-31 15:42:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-12-08 16:03:39.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-11-06 14:40:42.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-12-01 11:57:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REMOVED"}],"messages":[{"id":"17c3ed9b97a02368487ade7407d3cd180750311b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-30 18:31:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"730ac231f7936543310c22772d085d8f750ee7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-30 19:59:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/625545b9d56e475d89b7545f97a57904 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6904dadcc45d4079ba917cee0a32a82d : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0689a750bb9a46dc91b092891cb0e976 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/686c821ce4164008bb0be0d7dcfa4e69 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32fd6a15b6e14ecb9e190ded9bef55e0 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/099c3d3dec3f43248f18a8a79f818f74 : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f90224075584f6e9a9198d53cadbb5e : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c2d69786a25c41dbbaf6c59c55bfe835 : SUCCESS in 8m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/83b0af9891a14d0c9c4d6b7201865523 : SUCCESS in 15m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/47f461328d85487f942c74fa73cffb40 : SUCCESS in 52m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7012b6b4c404120b85de778d4f47aa9 : SUCCESS in 36m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3885492280434026acc80125bb2a3a98 : SUCCESS in 49m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/230110d93dd84c76851aa48b05eff471 : SUCCESS in 13m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/016cda96593348b88235fac0dc65803e : SUCCESS in 10m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6d7b93fde631451f869489848089b1c5 : SUCCESS in 1h 26m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/efd78c35a7e94f56bab2b3dd3220a929 : SUCCESS in 29m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d4f995f1fe44bbaa340a7803934d8d9 : SUCCESS in 29m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/315269f026f24e0d91a316be06313324 : SUCCESS in 22m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/36aa8aa2e1e54dc09ed78c4cb868c270 : SUCCESS in 12m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46ae92896e934466acd5c4e23c5423b7 : SUCCESS in 10m 26s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"73cec89218c5eca6fe3b8bdf1dd6806cfa95c57e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-30 21:02:00.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"752b587ee4e652c83be46ff3d648e402851dd4aa","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-05-31 08:32:34.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e0ee41568a719df80e77972496c6ce9391dd26d7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-01 08:43:40.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b5978ab9e95ac87c4211dfbbc6913242af0d99dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-01 09:35:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0a2a4e531a09ac253bdce0e025667d548e5d796e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-01 11:22:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d10dcc16ee330277dd5db2b8c3063f54a6054674","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-01 12:17:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2d61ef26adef46d0e2468ce2943aaf07dec0fe97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 14:11:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef970335d6f5447fa7dd7e3478baf64f : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdecaebd926447daaf065d7b04def478 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8aee9cf18d2449c19a1160e5336584ad : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb845776a9fb4d2eb8d1d8567cff08b8 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53fb6dcc99de41d2adb4c1c4316edb29 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e06c6a97b6bc4111bfb5954e63e49eb6 : SUCCESS in 9m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f799fd9871044924850ddf7b8889b826 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/553486dc025e425e8324fd2e19a062a1 : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/63b8765004ca4eb48c4cfe0238292055 : SUCCESS in 21m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8eef90cff0664acdaa40a86d6a407561 : SUCCESS in 1h 14m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d52c08b6e8bb478cbcc513f630f9ec96 : SUCCESS in 44m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/80fae662251140709b0b4be5a3a44f18 : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/87fc78cf3f994bddbf7364ce542b4b6d : SUCCESS in 16m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa8f0d9bb07e412c8dc842712dddf654 : SUCCESS in 14m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3af1b1a916e447869ce1a4dc65406c0d : SUCCESS in 1h 48m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e08732ceeb6e4048a5d76f15fef21cbd : SUCCESS in 30m 50s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f73abfd2c16473f8dd2f0692f635187 : SUCCESS in 23m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ca102fd9909c4ab0b1c7ce2df403f24e : SUCCESS in 30m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/47f8869b073f441fb614c6bba4ff8ce9 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ded55251477a4f80a4bf474bfccf1939 : SUCCESS in 11m 03s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":4},{"id":"f143a075d3ca7fb841bb13bc7f87c2f00db5dd3c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-01 14:21:20.000000000","message":"Patch Set 4: Backport-Candidate+1","accounts_in_message":[],"_revision_number":4},{"id":"bf641243ec423f659f9c2eeb7453fcdcbd966540","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-01 14:22:33.000000000","message":"Patch Set 4: -Backport-Candidate Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"a2f28ec268a6389c9c61a6b44a6ad68e777e552e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-01 14:35:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fe7fd9598277b132a039718482aec3bc25ddd3a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 16:14:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2cba2fe96e2e4021b57ebd4f8deda99b : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e81470dd7e74435eae1626fdbbdcc539 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e0a733b448d4e6a9eb279c7e7aed9a1 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f310a7f3b4d442f884db11e00334aef4 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2245b94ecdec43da8ece1b28b2b1afe1 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e46dd85f5fe04753829e9263f0fd2a08 : SUCCESS in 13m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f81c1992f37b46238db13e5453352dd5 : SUCCESS in 6m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2de0a157bf01454a979b8a2f5977a39c : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2faaa90605ac4f3aabab84707bb7ef5e : SUCCESS in 23m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/43c56e7d9ee24189867468e7c5d2d83a : SUCCESS in 52m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c98f6f1d2e84ee480020c8c779f9373 : SUCCESS in 48m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/677b59aa4090447d989f63ed7df42e5d : SUCCESS in 51m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/108f0c3d8b914038acaa50ab1813be8a : SUCCESS in 18m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61371175c98a4348b1fc77c3449f4e64 : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/99b61dac357640f59693dee1871d603f : SUCCESS in 1h 35m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0a127733c750481b95b0510bef540b92 : SUCCESS in 31m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5021ae5e0f048e38bb5ef710307907c : SUCCESS in 22m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/66a848c1ea6f49a4bc39cb47a8bb3704 : SUCCESS in 28m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/883081936156475eae11a2f32e5b502f : SUCCESS in 12m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dd9d0daaa1c406ab54d4bb9b702cec0 : SUCCESS in 8m 43s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":5},{"id":"435b29db80f22c36714a11f7634d990c6a5827e9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-01 17:19:07.000000000","message":"Patch Set 5: Backport-Candidate+1","accounts_in_message":[],"_revision_number":5},{"id":"9f6c652ceba94c4356ed9b4b8b272a4b1676e36e","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-06-01 17:22:10.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"0530f3a6ec4b77f38eb03aceedd1fb0be3fd2c18","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-03 10:25:14.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nThanks for fixing this.\n\nIn general I prefer not to expose all Kolla Ansible variables in Kayobe, but use Kolla Ansible\u0027s globals.yml instead. These variables may be useful enough to expose though.","accounts_in_message":[],"_revision_number":5},{"id":"49215f8e364986e38508919a46d0c3c529ac18f8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-03 11:16:52.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"26a4c4431e70fdee291889c217f2021ee6be756e","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-06-08 08:55:36.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4e7ee5aac3d9118ea285bc20930497963ab26a25","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 09:22:59.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e37496c13f301fbb5dad4e8ee86e2852048dfb0f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 09:25:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7a3409bfde82910d876545923fc2dd66f9b1c0d5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-09 13:07:44.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0dda454c20ffce31edd92c866f14964757a4d961","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 13:11:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7aaf8a4f06eef0cba1f51ee8432acd72faaa0efa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-09 13:24:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"be675b2a03987e6dcd2054ba4549d3ff955cb249","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-09 13:25:37.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dd3b155116b0bba024766c2d869a10e117d95851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-09 23:28:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bba0ef0ed3a02544c85b349c472e29384b02a158","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-09 23:41:45.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"19ade906a3f7c7538585f9bcc8b9874e0a189469","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 00:34:33.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2d9314ec04a448d94057afc9ffbbf02 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c3b09935b0a4bc8834ec9e3c3c97cbd : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ac1be6a4d9e492e83fbc40ac00cd412 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81dbefd47b0f45b8bf03180677595c9e : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b91b6ae8b3a4488a819c05543ef1ba41 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8fb066f3a31405a9bf07f51f66fea0e : SUCCESS in 8m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c7b655143c5344a6a5a5195221f9cb02 : SUCCESS in 6m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72a389371071454490c611e5abbe1ad1 : SUCCESS in 10m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/587986ce12004e8bb7e93dc3eb5bb145 : SUCCESS in 18m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d95696ac8c024295975a8533de4c7f55 : FAILURE in 10m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8804d702be14fb1904306c305565922 : FAILURE in 8m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/48ec8e5a9d4b4680b93638737c883a2b : FAILURE in 10m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c39aec8fd87345e39311c2db308c92ec : FAILURE in 12m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6684e1af41564deb85cfb6d846709c1b : FAILURE in 10m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/51af7ce03edf4579bba59faf3764bbee : FAILURE in 1h 02m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7d29fb80d9064963b7f3bdf0bfd346bb : FAILURE in 9m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/93f4cb2df56148eb8fc48072ee223da3 : FAILURE in 7m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f885b2d290714965b6d76c5d23babbe6 : FAILURE in 20m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/eab0a68b7a444695ac8533db51464c8c : FAILURE in 12m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8a241064b404e478a52be121d1f0405 : FAILURE in 10m 33s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"bded37b07b56e3c925870fccbe8f866974e76a90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-10 08:25:54.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0ef7df1b9bbd5e876d13166f1bbea8dcf8aecdb4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-10 08:28:55.000000000","message":"Patch Set 7: Workflow-1","accounts_in_message":[],"_revision_number":7},{"id":"9dc9e8396f74a53d015a813f41dfe079bfead1e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 10:25:28.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea332b63f13041a1ad7325691c7d23dd : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a47e6491ed4b42998ca95dde2a283b : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d56a69f298a475cb0f83f245609004b : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/053b46aba02f47198ae8d584532c6451 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f39515e7da5e4586bce8d874bf972066 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d33771ca46bb4a69a3d4e8d5440c53d4 : SUCCESS in 8m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a8b4a2239fa4b449b3db726df7bc45c : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cf267721aae64e83a861c44b344fa455 : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5fae695893e74c11ac48a977772c334e : SUCCESS in 19m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f165c31b607e456c9e09b0f0c81499bb : SUCCESS in 1h 09m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78c7aad7861c4339a46d39a312e9c4d9 : SUCCESS in 50m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c5c37fb0f0d54289904d512d7d8dc8ce : SUCCESS in 1h 03m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d8a5c0bd2635494289f8316a946bb090 : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f3af93c00224ed085209ebe0b8a0251 : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b7440319d2d2494ca90fc1e04525686c : SUCCESS in 1h 54m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/32830169831d46739e3f69da8f37e869 : SUCCESS in 33m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63d65079fb6e44b9846de5f444c308d2 : SUCCESS in 29m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9539082c6fd346249a4591d23d53fe1a : SUCCESS in 29m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/947c0c62407c49e0ab0a13209677a9d8 : SUCCESS in 14m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c700d6af2faa46e0938bcee138df4ff8 : SUCCESS in 10m 37s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":7},{"id":"bf9aa57a3cdb1c78ba42c953ed0e368728165890","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-10 10:33:38.000000000","message":"Patch Set 7: Backport-Candidate+1 -Workflow","accounts_in_message":[],"_revision_number":7},{"id":"3c440e1279a322e9b3a9d7c10b61ee9d2b29ed61","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-06-10 10:38:16.000000000","message":"Patch Set 7: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"5df19e7b75a1238edc484fa05f5055f30cf8e808","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-06-10 10:41:35.000000000","message":"Patch Set 7: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"43bab0881f17b23deea14e26ae361ade7ffd50da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-15 10:24:05.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"ce43229ce8064bb89a9fdb4446112fd9c4834f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 12:30:06.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78dba64facd34fb689661764a789588a : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/020aece5e93645f98190e31bccd2f3c5 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bcc50d605294fab98180a8f4e24b66f : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8302550390a4ce6a90207d07cb166c2 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26f5dcb07f524f3da836fbcd32782273 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4211493da2ac41008f3d688fdd9ef875 : SUCCESS in 8m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/26b10acad4c4489e95689024f0b4225c : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9d6b2d388e614d6da014b1ecbc6db38e : SUCCESS in 10m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e7eab34221884194bde973393cf3544d : SUCCESS in 20m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7e7dd40b2f0641b8bcfba31d35cbe18c : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3d4622b2b8942f397301ce6b80a3b88 : SUCCESS in 55m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ed45f7d5bae74621958d72e68baa4772 : SUCCESS in 1h 10m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a255b0b23fb448ce8e5d996c8012e379 : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01835a65dfc74aeba902e9489c9e5ab8 : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0c8ce40012374ee889b6c6a7f3480e44 : SUCCESS in 1h 53m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/915abdfa8dc648e3b8868cc528ff7e60 : SUCCESS in 35m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ea64ca661e34530b08592c338f1dcc4 : SUCCESS in 26m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/856538bdd5ed4cffb2b7e7d6f62f113d : SUCCESS in 26m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/97901c6a511b4159b023de93560ad3ab : SUCCESS in 14m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e50aece1a9f64de3adb8bba41a7b1e3d : SUCCESS in 10m 46s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":8},{"id":"330da254d76d984b53de97292573941bff5476d9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-15 14:54:53.000000000","message":"Patch Set 8: Workflow-1","accounts_in_message":[],"_revision_number":8},{"id":"d34fb6e63fdd777a173504f2e5830351eeb67bce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-15 15:49:04.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"2880090cf46a7c5ca9700dbc7534fe0d223ddaae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 17:32:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0a1911523a8487b8d39d1af62b5088e : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3bb464837464625ab98985eeec76e50 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4f5691ce87349b7a38de050f1d4155d : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/745bc17ebbba4b109a9412e5344afde2 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fcc34b8bf2149e987d095f609b2ca35 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86778797cb524be5a3de9e8c5fb0ee39 : SUCCESS in 7m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82282463ae094ff2a6825b4451c61b82 : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81b32f7179a6401fa472881b3f516dfc : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/11f6df34980843ee9f1aa56aa2fa3a8d : SUCCESS in 19m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/39bca32a60474b36b4b368360d08f988 : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9e4cd3d7df6443e860f300f55ad8264 : SUCCESS in 45m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1a683876ae1a4ba19bd3eea573812430 : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/657b245068aa464fbda0db18ad8e514c : SUCCESS in 18m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad18ec6fc65343f39146e121996d0884 : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2731904701784746845908f18958ac94 : SUCCESS in 1h 41m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/82682dd2da314417afcdf0ff3bfa6076 : SUCCESS in 32m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7d3055be2d84b3dbf9e4419d6a39b09 : SUCCESS in 25m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9caff02e312c47df907bf879879552d4 : SUCCESS in 26m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/727b6333ff3c42f69f43badcbda708f8 : SUCCESS in 12m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8491471933f040eeb6c7f4b17673c1b8 : SUCCESS in 9m 25s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":9},{"id":"9a9eaae87ff374a4885cb8ac5898d293180b5462","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-16 08:21:30.000000000","message":"Patch Set 9: Workflow-1","accounts_in_message":[],"_revision_number":9},{"id":"be102b1d3281ba5826adde1c9b751d73ff8f58db","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-16 08:53:14.000000000","message":"Patch Set 9: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bf4cc59756b8f6ba612a15ef98bec569138d8d17","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-16 09:18:56.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c3b6cd9489bbe2b6727551db9c90a9b82c56a4da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-16 09:19:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ddcbbd685b1a10a4f5ae544af54239eee90bb74d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-16 09:20:46.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"e12fd46e62093e3d7850c1846a60bc9bff500a79","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-16 09:21:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bd9f2ea89adac51f11e50803bf5b1ddea9637549","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 11:17:00.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d1c9c1ee8b0479f9f287fa5b32bb920 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3885331e519e482fb11f3c4d53d07baf : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54a4b15e281c424bba8648f7056dde44 : SUCCESS in 10m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6395f4ed89734ff9b0113c036c4a96fb : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a81c7d9b60342929a39b067ea1baa0c : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/583d867df2b64601ad7980264e75954b : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/63d74ab4b5f34e2294a7fd61fb4acc5e : SUCCESS in 7m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98a19291240d49a6a98ae8ff52447544 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0de8c7f12760449a8833a28abe0bd076 : SUCCESS in 21m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ef905fb24c6d4577965f20e9371d9872 : SUCCESS in 59m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/523a5c1b5e5241b2a2f7103fee7a2857 : SUCCESS in 43m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1a31bf93f10c441c9ed679b5e8998c42 : SUCCESS in 1h 03m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/63df4e8e524c47fdbb4e25a42ad27b5a : SUCCESS in 16m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad1a13d5b608437fba97f676a4e376b6 : SUCCESS in 14m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e184d1dabf4c43fb82815560f8578a17 : SUCCESS in 1h 47m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/79c7b8c547314c16807d6e1adf58678d : SUCCESS in 34m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95d1959ca2254d1facdfd6ca1fa8821d : SUCCESS in 27m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a6f0f82f06f145048cd672a01f6fe180 : FAILURE in 19m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9074d185d6f94a68baadcc0fb71c064c : SUCCESS in 12m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebb5cb82c09e426eb645c29affea1b85 : SUCCESS in 9m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":10},{"id":"39c967cd89479acea9af4d5b713977def669cfbf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-16 11:44:44.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"db8de39bcde20fa19e0f549f8cf20f569d6c1247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 13:37:43.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2305fda9076e4886b1feed817f6027e9 : SUCCESS in 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c54268dfc6554186821debcc4dba40c0 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a30f6b8f587f456fbac90208a20c5009 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c5668e1650f47ddbba5325bdf14d945 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a292adaacdb4434b8c88aa3c82232cf0 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63068343ec0140e0b69ea224b4d848b6 : SUCCESS in 8m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b9735a75cd04f80993a140408fa1642 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee2a7043923b426199934c1808643495 : SUCCESS in 13m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/64cfbdf418a146e8b19b59a61c1f1b98 : SUCCESS in 18m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/efb5a41aa1224b37b31399c3e4b5c4d0 : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f2f4a97c0684550b38b1f64e732b0fe : SUCCESS in 51m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/909ba9b82b0b4daf91e02bb78832ad12 : SUCCESS in 1h 12m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7ec90329344341dbac9e479cc88c20f6 : SUCCESS in 19m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e87bbe50a8a4d07a9c9771ed5614e21 : SUCCESS in 16m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/af118f9633354728a7edfeb216926f74 : SUCCESS in 1h 50m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/80ed59eaba9d40aa9478780217c53462 : SUCCESS in 37m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7770068f4b3649faa487743d18227623 : SUCCESS in 30m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a1d5d3a3da424184b790f554a936f748 : SUCCESS in 27m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c51697f50abd4337804c0854d73fc13b : SUCCESS in 13m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f13151cbb48e4e289f2ecd3a04dde117 : SUCCESS in 10m 44s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":10},{"id":"f35cd4485b3850cfcde329b2acbf59ac5aa110e6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-19 22:09:36.000000000","message":"Patch Set 10: -Workflow","accounts_in_message":[],"_revision_number":10},{"id":"5560ac10c515e42da5b06538179c8a964dd6ae3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-30 09:24:17.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"21fa80096730369cc166c96e258f20eb1b66e5c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 10:31:02.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/447a9844eab94ca0b70b11ee608de196 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e90866d45474a3e9817cd19abbbbde5 : SUCCESS in 8m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d727f202b5c4ffaa64ef24a85e7d4c9 : SUCCESS in 7m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e420099fe77543cc847fb01b267b8c94 : SUCCESS in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8303b0def5b4e2f824370ccbf0b7652 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6ae2107010c4545b394e8210de844c7 : SUCCESS in 11m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9076d8328a7648d9b72d8310947ba9b3 : SUCCESS in 7m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e3df4de05a254b4b904056b92c13c371 : FAILURE in 13m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4f0ae36e61e448794de6c5408ed8cee : SUCCESS in 20m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e53bbd5d99c946b99ff91981ab9209f3 : FAILURE in 11m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19212b0894e54c6bafca30a39aeb5605 : FAILURE in 9m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/618d1ac58aee44d39c6ab938b048ddf8 : FAILURE in 11m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e964380cb59e4880aace71e6e4614653 : FAILURE in 19m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faeed26c13394e6c9730ea871f38996d : FAILURE in 14m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a1219b899a3e4529a5356be236a88c4a : FAILURE in 1h 02m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f8d765af9b7b47ebaa4cb37e53845999 : FAILURE in 15m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fca271c98d0047659cf71b15f4f12d60 : FAILURE in 11m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7c63f7ca0d54437b9e05116aac691d68 : FAILURE in 21m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ea7802f2160b473c886dc75ced3a7f20 : FAILURE in 13m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59412969b75048f2b5de92c57af0e470 : FAILURE in 12m 14s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":11},{"id":"b5031e9dcc4980e1654e167ee4530ee10dea615a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-30 11:50:10.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"953b474ce6377d3866c7a1c824f2bc7a9da8696a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-30 12:15:57.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ab0b2a809412372cbe12452a6a6435cad5f53d55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 14:04:37.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e859bd6025f54c759599137fb345e174 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf9e6ace4a04e9bbedef67066e69323 : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a634455d55a41c5a56c18cb4970a791 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/667bde7ffdd94e3f8c94cbaa40a6a026 : SUCCESS in 7m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d49f5a764864bb2960ace5a38f1b893 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e756ab95654c22b0a4e1b0d2019b30 : SUCCESS in 10m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/db8cc7b24dce4a19b93d02f421360408 : SUCCESS in 7m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6012ce0ba439461892287d87b3010128 : SUCCESS in 13m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b6df6c30f43454799363f162f543723 : SUCCESS in 21m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8abf177a0cff44ceb752bd4eb11d6a36 : SUCCESS in 1h 08m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70336d8f77fc48a1889e107d276e6b7a : SUCCESS in 47m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bf731ab3ba7a4d78b0415491586a8195 : SUCCESS in 1h 17m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/30de05bad8f541febeb351018cb05d82 : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c24929c3a071407e89b5beb90180b5a8 : SUCCESS in 16m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/88641aefda644fc9a111f001752660ad : SUCCESS in 1h 45m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/35f384eb53a54576b4a3ce3d580317b6 : SUCCESS in 39m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb307e7d767940a7b230df88d697a418 : FAILURE in 21m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2881bd010a6a467ca024ec08d8b3e227 : SUCCESS in 29m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9f1a57e41094421e882d72216162ea16 : SUCCESS in 15m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e05337a01d7e45528241e0749c19bc3f : SUCCESS in 12m 32s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":13},{"id":"b1c8b72dbcbf42e9f50d23967840e1aec9c27b67","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-30 14:49:27.000000000","message":"Patch Set 13:\n\nrecheck seed-deploy: Failed to find required executable ufw","accounts_in_message":[],"_revision_number":13},{"id":"5304894a9643b06f9414e204535dd446b2b3f9d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:43:51.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0bc68ec6aa5410ba2dd520617ebd2a2 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a49a996233459ab59acbf7b44aadd7 : SUCCESS in 8m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6991332ba8674026972e11c3168a1a38 : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1816dd664e7b4bb1b832451a5adeee87 : SUCCESS in 8m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/605ffaceb41541fe9938eaa4bc6b8bf1 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544c1c92334c498c92419a3fe0bbfdda : SUCCESS in 12m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/49a2180cabfd4df2ab3097e9bdea3f79 : SUCCESS in 7m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aee253591e00464cbfff248c01ed040c : SUCCESS in 13m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b637a3ecc79a422dbca5f6d767c2549d : SUCCESS in 22m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c030aa6e53f649dbb495bc34ef119f93 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d189772c01664eaf890593c4ff8a6ae0 : SUCCESS in 52m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/13ed4f6d63964a17a755f2c7df9b4b82 : SUCCESS in 1h 08m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a073cc31e9404f9b8cc1b4383509d06e : SUCCESS in 22m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca52d3bf5408460d89d04e8cdd187ae0 : SUCCESS in 17m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d84fb9d9ecdf42688b1e85cdbcd2bc28 : SUCCESS in 1h 53m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f4f8be645a564f27a0d600c9e985ff1b : SUCCESS in 36m 58s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7108d88e9484b2e9d421ed346d46a41 : SUCCESS in 30m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/552136c4fb49490f9574f1e2248adf77 : SUCCESS in 29m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ad4cb583d8da4c69bd2b9e36a4c8d210 : SUCCESS in 14m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a228255ebf984ac7aa9a31dabd66a048 : SUCCESS in 12m 55s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":13},{"id":"1ed3bbd9390238930d32097a1847d870b9eeeaf3","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-07-02 06:33:40.000000000","message":"Patch Set 13: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"989bfa311a8963054fa6512eb56bff4dbdaef4c1","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-07-02 07:34:22.000000000","message":"Patch Set 13: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"91d920326bb8d7cfec521ccb0db207a7767471cf","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-07-05 21:02:40.000000000","message":"Patch Set 13: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"c5fca6bfdd5c2546de0987c51aaac363477cbb86","author":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"date":"2021-08-17 12:16:47.000000000","message":"Patch Set 13: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"c7c989dac65cdd2dbc101090f653a994bebd3624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-20 08:55:58.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"d74480fdace89afd2ab70e65272172b538612ab1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 10:40:12.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b35e7d30a6174c259bc71e580080194f : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac3c8d40731f452f9127de55c588a926 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c26b2cfb8f7f4439868c24fc6ff7fa54 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df1d2e58b35348f6855a9a09a40b23f4 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9044cf34f5f498eb9a8b1f896806a16 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d96d4cfc2f4a88b2e7b18643dd7c43 : SUCCESS in 8m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/392a3f2c0f9548afb278758091a142db : SUCCESS in 6m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc8c5fbe98d141c7a840553c75582329 : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75ecf1835f8c4e12aedd4479e9dc8929 : SUCCESS in 18m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7af08dd1df134a1b83b46471c313aa79 : SUCCESS in 59m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4247b147384e461eb98d8a4318540ac0 : SUCCESS in 46m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4f664470d8534c6cb4c8ab7f71661e4b : SUCCESS in 1h 02m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/80912ad1fb4c4e06b14540ba3373f9b3 : SUCCESS in 17m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33db704988b749c88c450d31e5f91330 : SUCCESS in 13m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/128e8a1dde6741e69208eefc284de731 : SUCCESS in 1h 40m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2f0e08988aa043048a58ae5fddd65363 : SUCCESS in 32m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/240a23b9c16742999d8f9d3273f27f27 : SUCCESS in 26m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b21ba973253e471dafdeba7c90feecd1 : SUCCESS in 25m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a4977df2af7e48988a4e0c8975fdba67 : SUCCESS in 12m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/618dc1b8f1cd42539596ea568e0f4eeb : SUCCESS in 11m 40s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":14},{"id":"5abe5aaecaabcb6007f98d6e57f17a142d712c4b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-01 05:36:27.000000000","message":"Patch Set 14: Review-Priority-1","accounts_in_message":[],"_revision_number":14},{"id":"73e76041fd77e3ef3e57eba99001b884e83fd839","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-15 09:33:54.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"4e87cd9794185bfd43f59855d21d500e364859bd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-15 09:45:26.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e129bf3e95bcd0fb707a2641e340225d3ac5de9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 11:15:51.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fc767ca08034d56bfc75ac922555790 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf75ffcb320a46d7b5ff91fe55cf2c67 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c58bc11160c409ab59252d2edcfcad2 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f81f87bf11ed4f5da3bece8e3666b086 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99994c15aa2b4f89b17b8e7cd6c94802 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c739b2891db84da1988ac462e1f786ad : SUCCESS in 9m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/97ef6c38d23d4f41b81cfa0805fc6ea0 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a6b57f3d456c47629357d8034025adfe : SUCCESS in 13m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3c1c23c5d0b84944bf4700217a2ce6db : SUCCESS in 32m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/878eb6c0e6eb4463b023f270acbd5d5a : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aec8c89093354e75b5578b9a816d75b1 : FAILURE in 45m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e877da3ccbbb44f1b699420a493b30d9 : SUCCESS in 1h 07m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c98162344abf4e8ea0cba864c7dc6fe8 : SUCCESS in 20m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4974bbe939d94d43824853bc6b83ee60 : SUCCESS in 17m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fc6963e825a941e1a545fe424ac4f75b : SUCCESS in 1h 40m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cc72122e06543cfb5d94ddf6321ceaa : SUCCESS in 1h 12m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/595584a5b6694936bf59543b0e7b1517 : SUCCESS in 35m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2afd16783dad40298b6dfa88fe3518cd : SUCCESS in 27m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/70277905033a4437abb77d479feb5257 : SUCCESS in 32m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2deaa1fd490b480885e713f5cd4c2e29 : SUCCESS in 22m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b864e40cd3cf48b29d6926a00b2d8f94 : SUCCESS in 15m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c71d89c862e740aaa37a32a45a3d6044 : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":15},{"id":"798adc7aad29d07994bcbc902929c23cb0284d40","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-15 11:16:21.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"0b69872d33ece4fbc8b4a0af32674a2e38c649a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 13:22:17.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b858cdd497404fb38a876ce89b0d0346 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ede2fd60c3d46819e5573970552b296 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1289f7a438d436e9ddb9d31f8c1ef80 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/274f5403c66c4a15ad99f70d2ccda354 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ce4a534475a417bb9c6e42abe66943d : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27adbfe68572435bbf7d59cceabae1b1 : SUCCESS in 8m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d60fde83e16843689b2387c489cebab8 : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aec7ca953b70474a859b537c2f3a6440 : SUCCESS in 13m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/325292b2e61f4ad38b3399570ad1b14d : SUCCESS in 29m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3bea2f1d17724f64aa0d5cd6c11de977 : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c872d64829784ce4aae0c11ad7cff346 : SUCCESS in 51m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dd62ce9737f34e83b9343c188e99f797 : SUCCESS in 1h 16m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d9453c6f52c440d4b033a7fb5f182939 : SUCCESS in 23m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91a029b5117f4984b5f088b87ffaaabf : SUCCESS in 15m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/828dcf7362bf4c68895a9f7340a4ca43 : SUCCESS in 2h 00m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c808295681b4babbd1e1a0af6de2210 : SUCCESS in 1h 25m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c170cfc3743c4815b5d6f7c16b18e52c : SUCCESS in 37m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b6a330198c04aa9b234f03338de351b : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/46210b0c22ad4f5d94ce16a82ea0d0f3 : SUCCESS in 31m 59s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02e1415a75c4456abe21a9b68107b87c : SUCCESS in 27m 53s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/216b4c51101a409fb8e4b0b2ccd34707 : SUCCESS in 15m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16265ea7b8dc499e9e84d648c35311be : SUCCESS in 13m 09s","accounts_in_message":[],"_revision_number":15},{"id":"8c4876a6239c6a93a983bf65e1ace7fb99019b15","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-09 13:51:52.000000000","message":"Removed Review-Priority-1 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":15},{"id":"b05c840553b1cbe98176210d2714aac6f00e8365","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-01-12 13:11:21.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"5721512b0a65ff8c9b6d3bbbd91c4d519ae47d30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 16:23:15.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d660bc16b8764f6286ff55d65fe1daed : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b99ce177bb4b39af1fe0eec5cde81c : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b45e9a7ba66a406384fcf1005f2a450b : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/600ac1517b4b4a94a5286d27009edb28 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a410033709d2450d9659eb7ef1516160 : SUCCESS in 10m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c87ed76ad364bb29287a4584195bcdf : SUCCESS in 7m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2667ea5cb7ee46db934a8c27c26a3674 : SUCCESS in 15m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef3fc7e68fe248988f5aadc765eb0693 : SUCCESS in 36m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cf00eeec99654ec5adbe2294099c24ac : SUCCESS in 1h 14m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71cbcfd0f69c491fade0aacf9ee2822c : SUCCESS in 50m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cf6724a5fe414cabb49b49c4f8bae19c : SUCCESS in 1h 19m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cf687c9a99254990abe2894eee895a4a : SUCCESS in 23m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66b376c86db1424abd98fb2d77028341 : SUCCESS in 17m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f65978d8098f4e58bdd286f1b4aadbd5 : SUCCESS in 1h 47m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d5f3b5935e34af7a3dc3b3b7797251d : SUCCESS in 1h 03m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ed2a974abfee4334958f9f8b719c2123 : SUCCESS in 1h 07m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3784b64db3ab42618be37c40636458c3 : SUCCESS in 45m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0e87f0f3b416439fb16cc56dd99f32cd : SUCCESS in 34m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22c5fc9d431749b6a45618f7e478a34c : SUCCESS in 27m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/66e3a9e595184e4598383c422cee4ac3 : SUCCESS in 18m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/215320a35fd243a197d38a9404632f6a : SUCCESS in 13m 52s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e4e2d5d40bb54581899be72c84435b6a : SUCCESS in 23m 05s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55b7c62cd6b442d3a27fa48f10d913af : SUCCESS in 15m 16s","accounts_in_message":[],"_revision_number":16},{"id":"6eae85e7780d4d9c762ad096dd0d343f1c101b8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-09 22:08:58.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"39bb7669497231a42575450af87ecc441a8cc90c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-09 22:57:23.000000000","message":"Removed Backport-Candidate+1 by Andrei Yachmenev \u003cweanew@yandex.ru\u003e\n","accounts_in_message":[],"_revision_number":17},{"id":"0888901effd5fe5474fd921c08da9cc9181fcf7f","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-09 22:57:25.000000000","message":"Removed Backport-Candidate+1 by MargaritaShakhova \u003cshakhova.margarita@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":17},{"id":"28d167981ba598dca6c219a8d02047c4b95ac99c","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-09 22:57:30.000000000","message":"Removed Backport-Candidate+1 by Sergei Raiskii \u003cconnoreika@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":17},{"id":"8fd1ae05b83a4cf03cb718eb1e26a5006fb94e1d","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-09 22:57:33.000000000","message":"Removed Backport-Candidate+1 by Ilya Popov \u003chebulrih@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":17},{"id":"c6e88f7bbcd09fefa6416ac48753ce243b0a3773","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 23:59:05.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea7865235d174b46a3808e4dd5918456 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70ae2049c6514edd9358467dce4c3144 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a600275d75284b66baa2fd3ec2f02c69 : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53779fe161c34f73820dfe9093e324e2 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be904d373d4495f8f72398e33f6ec10 : SUCCESS in 9m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1eae7f0bf9c9425dbf518f5d70205803 : SUCCESS in 7m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c18557b350c64d8fa7368c9b45188a5e : SUCCESS in 13m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/88c9769b273f4adda0abb5638ccd2339 : SUCCESS in 40m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/41f5671fce27494dbd161d7cd7e91913 : SUCCESS in 1h 08m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/361dc27d521d481fad4f8a37817ccb00 : SUCCESS in 45m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c08b3028f1ca457bb6ddae5f626d8bfb : FAILURE in 1h 08m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/167eceef022645dca848ab2254ff4a8c : SUCCESS in 23m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84190e7f94554965acb165d9fa5eb5b9 : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/79ebaf066ead4be9bdff7aba271c0db0 : SUCCESS in 1h 44m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1481258109f47efb6bcfb6a696f125e : SUCCESS in 1h 13m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c75de0dfeb6a4d6f966c129458ef32c2 : SUCCESS in 45m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a9dbacbf6ee407385a01e1dffdcf48f : SUCCESS in 35m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7826b6d9d26442d096fa0b74783e6810 : SUCCESS in 34m 16s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2da1febbca2e492090083b9e7a61d031 : SUCCESS in 24m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/96b532622e5e4832b67289f0a37b43c0 : SUCCESS in 16m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed27ceb199ef4a55a3f4fe9740c3f4d2 : SUCCESS in 12m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/80a8b6014aaf4e049a9a5ad76b95df41 : SUCCESS in 21m 50s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59b5dfb7ed924a5e9baa551f0bf11e56 : SUCCESS in 16m 17s","accounts_in_message":[],"_revision_number":17},{"id":"f73173f33662c35d933a77f889bc37d5d8598898","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-02-10 00:09:54.000000000","message":"Patch Set 17: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"70960ef7481eef4bddfceee31372ef0610c03f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 02:58:28.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8728395d734b421b845267a547c70bb5 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88f380e30f224aa9a1dd422c8147c0ee : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b893a2bf92d4e7b8ade1e2aabce674e : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c25f40ac0ce409195cae187b343af65 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ec01fb9d284646b01328b93c5ff1ae : SUCCESS in 8m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7275341456b34dfe8ede54d483e703d8 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e1f852a4f2a4269ae345cd3e0784077 : SUCCESS in 12m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08e584261338414eb58fc10e94b51bbf : SUCCESS in 27m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/40614ea9e1ab4bb587158b766a407e10 : SUCCESS in 57m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6109e5bd3d41465da41674435e49f393 : SUCCESS in 45m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/210978dac82f4941afcfd89ecbc17448 : SUCCESS in 1h 07m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f8a1400165a54175949f3c07541b6768 : SUCCESS in 21m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27fcc962ea344512b4a36d75677a3053 : SUCCESS in 17m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/81b5d1c255e54bb5b107d877b3f17641 : SUCCESS in 1h 37m 58s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2025e99f2f024565aa3b5cb82053592f : SUCCESS in 1h 10m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2fa360ae27ff4204b7ab21de1de2fdbb : SUCCESS in 37m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd61e82dcf8c4dd383f7f86894e7af37 : SUCCESS in 32m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e8079331d19940b9bb311e6c037090e5 : SUCCESS in 28m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aab4860963ea4f708e31976119237611 : SUCCESS in 22m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8cc0aa00477b463bb7604a8496e87991 : SUCCESS in 16m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89fdfedce1ae4406af121accb7aba1d5 : SUCCESS in 12m 28s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6e4a385d2c4844d19a4b5a78367dd279 : SUCCESS in 17m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21300e12a95240b79ef5dd1f317f5e09 : SUCCESS in 12m 37s","accounts_in_message":[],"_revision_number":17},{"id":"c033daec594a0df9925e79aefbcb8aad94e288bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-04 22:43:03.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"36cfc8e08ea69997fe9d9c9f04c95f7c2efb4a9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-04 22:56:36.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"0092730fa2599e5fe091f588655a24957bb1391d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 00:24:44.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed37a5511ea84bf2bc592d2ffe2f7907 : FAILURE in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b767b53eb318448f80abec609dac5716 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f194173a46549bd9e5a66572f191879 : FAILURE in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43be0dfc814e434aba9e86485946c565 : FAILURE in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6c1654f010c44d09bf023932213c073 : SUCCESS in 10m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e87abeb26ce64178812cd870f63f135a : SUCCESS in 6m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/97516789517c46edadc7b3895ebfd43d : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b84d94dbe6cb4bf6863168deb81cf53f : SUCCESS in 28m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/99666597620b4ef7b3e63e3c190d240b : SUCCESS in 51m 40s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/5ef1a31be74740feaf42cc92fc2e7df8 : SUCCESS in 1h 06m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8aa947f9f0134952bd3cb1a134420e41 : SUCCESS in 43m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/54a81882607c4543bb6dc4b1cf3d12a3 : SUCCESS in 54m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/466fb9f35981481f83c687cc352fcb26 : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/52b5c514e25740478ffaa4e823aa9b50 : SUCCESS in 19m 30s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21e1371673944f10b907775ffb4e1780 : SUCCESS in 12m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c13bf1ac3f804b5d93c861d34360bebc : SUCCESS in 1h 20m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faa4b0b344804519880d23d20286e3b7 : SUCCESS in 59m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/89edd61725234b608cf54ae38d7f5014 : SUCCESS in 19m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/fe5482dddb064a1a811451f799ef844c : SUCCESS in 24m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0e0f150065b4649adc5bf77693fdaac : SUCCESS in 13m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cf987173e4cf49508070d27c31f9ea9c : SUCCESS in 31m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d046dc2816b40f0bf644ad4f4e3bf1f : SUCCESS in 24m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9a1aa4e30db84b5ea99d1bf96b4d555f : SUCCESS in 13m 26s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b7607f27bfd6425d9a69630e8c6311aa : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9002b775cc1c4187b42d7ac3ee39ef2d : SUCCESS in 13m 40s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9a528965b30c42a7a8f419834fc68a1b : SUCCESS in 15m 05s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ae5e3b01471f441c9e6ffe86c7133eea : SUCCESS in 17m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/071d43c214e74439b102d9a904671382 : SUCCESS in 11m 10s","accounts_in_message":[],"_revision_number":19},{"id":"3fa01e28d318799ac8ca983c9dccb0ec695765c8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-05 15:16:45.000000000","message":"Patch Set 19: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"d0a744e22551e36583e813ebb4e110740e354dac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 16:27:10.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02496ef1efa14308bbcd44900f3baff9 : FAILURE in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19dcc99602e748b1b9d0663e83fd95a7 : FAILURE in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28f94b5ad2764ec5ad2ae9000c8bca06 : FAILURE in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf4e1c714eae40a98a3345701f4a3843 : FAILURE in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09a44c1c63c94eb2af078f380de78b10 : FAILURE in 5m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/14ee010988cd4cd19011aa92673f19a5 : FAILURE in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4cb279847a1c4c968d75adaaa43c36e0 : FAILURE in 3m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/53428ccc64524f5087d0117bd79c56a5 : SUCCESS in 33m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5bb459e1edd84421864346f639082cee : FAILURE in 15m 47s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/6b99ea046dac4090b42f2ffbbd3ea9d6 : FAILURE in 14m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a207e3d316624cfe8500609865984048 : FAILURE in 12m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/658ab1c57c304b9e9307ea2342331b8b : FAILURE in 17m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4daae373ae9a40ef8715a908ca22a00a : SUCCESS in 22m 07s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9fedec2193de4bd095b403cf09bb16ec : FAILURE in 14m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/deedb47f96d74dd597a72b87b8d1c753 : FAILURE in 13m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1acd2dbdc4854e1d816e8e1bb0614c60 : FAILURE in 51m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f58fed9a32184377ace512339ac0be78 : FAILURE in 1h 02m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8b8c67ecbeea44e2862f0cdacaf67498 : FAILURE in 18m 37s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/04d1173a678b4fffa75c5d647c905d1c : FAILURE in 14m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7983870dd44c477da821290faa29a458 : FAILURE in 13m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c010902217664ceba2468320694ff75a : FAILURE in 48m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12cf3b99378e47a4856ead84b32bbf93 : FAILURE in 23m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7db18e4d5ca14b2d984a9f0f5eb0c5fd : FAILURE in 15m 32s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/19ae8a187ef240e8ac9f895fdcd2bf1a : FAILURE in 14m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1c514e6ee744a379f372f0774bbb2da : FAILURE in 12m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f5c263748e7d4065aaeae819e76db56c : FAILURE in 15m 49s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e78f662ed96a4f799b49a47d23968ee9 : FAILURE in 15m 15s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/012acc9e0c844033b13e976e7b8ca4cd : FAILURE in 12m 31s","accounts_in_message":[],"_revision_number":19},{"id":"279cb301fc0e28f1abb282f734d21c287d738095","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-05 17:52:15.000000000","message":"Patch Set 19: -Backport-Candidate\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"7b137a004bd3222a4431f167b237d30ee38fb4f3","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-05 17:52:16.000000000","message":"Removed reviewer MargaritaShakhova.","accounts_in_message":[],"_revision_number":19},{"id":"870faf28d580a8abf537d30a0c5f788a1d07d053","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-05 17:52:16.000000000","message":"Removed reviewer Sergei Raiskii.","accounts_in_message":[],"_revision_number":19},{"id":"cc3c12247d51bfd727a008f98143f04f7d54b68b","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-05 17:52:16.000000000","message":"Removed reviewer Andrei Yachmenev.","accounts_in_message":[],"_revision_number":19},{"id":"dc0f2c6395b7a4043f4a33cd1cf96475977f3616","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-05 17:52:16.000000000","message":"Removed reviewer Ilya Popov.","accounts_in_message":[],"_revision_number":19},{"id":"1ea320fd7553a2a2e9f7795b690437d4f5b1d26c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 19:42:34.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75a44e8304d145c880ef5bddf07e345f : FAILURE in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0f9e2121ea4537bcbec9dc4d58a2f7 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2e748cebe5a45679a4c6d895e605372 : FAILURE in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f46861adcbb4b05a114f28e9f2079fb : FAILURE in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dacec76d2cf49d0bfd8b4d616934229 : SUCCESS in 8m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6d4bfc7a6ed04bc6b0238a2fb761fd68 : SUCCESS in 7m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bb1c82af86af48a8852a6f0f7c1c7dc9 : SUCCESS in 12m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ea9193aaba7747b58b85fbc23eb9f660 : SUCCESS in 31m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e3652dd93be94974b5047172009aa7b1 : SUCCESS in 1h 02m 02s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/102bbe9eac484df9b87c139010ddcad5 : SUCCESS in 1h 04m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5eb20333ee448dfbbb31f5f83d3ba0c : SUCCESS in 49m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ae3f8e47b7554d02ac3e744bd52c2fd7 : SUCCESS in 1h 10m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0fb7cad14367424aa310a90230c7a705 : SUCCESS in 20m 36s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/25668fb34b814d8eb4066c255e5bf51f : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46365ae3d2cc4e688e51b5703560c87f : SUCCESS in 13m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e5c594105194e75a38a7b6e3ef2309e : SUCCESS in 1h 44m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8d386304f4b4d3683c6ca820e50638f : SUCCESS in 1h 08m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8dfe40a890b849848254f6824db57afc : SUCCESS in 23m 32s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4e61237b7f9b4df497b682f6862fe097 : SUCCESS in 21m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec951f615e08453491dfda60f4e85501 : SUCCESS in 15m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e6b38f728e6647cbbc5ef0600a9314ca : FAILURE in 20m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6190fad1e7484e5da9608c96d0e19921 : SUCCESS in 28m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/168b0ebb809f43dd8f7daa7de384e423 : SUCCESS in 14m 38s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8f9c3d01c8ad41fc9caacb6c32cd185b : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8189a59145de4c119b81b0574c678a08 : SUCCESS in 12m 01s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9c0c2b9665cb4edeb277a68e21b2e904 : SUCCESS in 19m 52s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2408681a586e47948654fbc3bf1bb314 : SUCCESS in 15m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a24b5804649480a9fdae584c03c7002 : SUCCESS in 14m 06s","accounts_in_message":[],"_revision_number":19},{"id":"550c7420859038a7bbad54bd88fed93360897580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-04-06 09:36:37.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"86c3add8793e4d94ebf24bafc8c08d91e685af89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-06 11:09:47.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2ba3112a1264c53b305297401b8a426 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8998ed03f645dfb3b82f17bd1de155 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94eead50d47f4e6387a637be9e6b65b5 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/356c552e72f3485d9b806a635d1c17de : SUCCESS in 7m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f432465533f24d559656bf3590e54144 : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b87b459c7ab74a00a3d4df3a189419c4 : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b915d367e23425d88a84073c05c2437 : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aea69d156b0a4987bf48ccf332b1f6de : SUCCESS in 35m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6757edba2bcb473a8fa62ad5931fe91d : SUCCESS in 1h 06m 35s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/33d1d18e9b22417099d9a013164d9d7d : SUCCESS in 1h 10m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca546f8a0dd84c5daf68b6d330588cf8 : SUCCESS in 52m 30s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/87e7b1653df84185b67a0546ee8ffbb2 : SUCCESS in 59m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/538e935f0a76465da179517a32e77128 : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c51e4303c6e748c3934c90bb490a1831 : SUCCESS in 21m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2503e3f9d6a74a25b9da5335dd394407 : SUCCESS in 13m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2134fe002aab4b8e80bba8497c1881e1 : SUCCESS in 1h 31m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7be58547c82248aa898308781d48b025 : SUCCESS in 1h 02m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4d2659357341459881a58e3aa5ccf7a4 : SUCCESS in 21m 26s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5273715506ab447a8f32d1dcfa3704be : SUCCESS in 24m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/918e06213c864dea9bdab69bdcb3c565 : SUCCESS in 14m 47s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ee45b67aa89145d6881cc30f7d715e40 : SUCCESS in 34m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78454954ca4f4570a6228618060c7198 : SUCCESS in 25m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/484e85eec7da4975b024cae645dca61d : SUCCESS in 14m 07s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/20577e7b7e944d58aeae2d71692d654e : SUCCESS in 16m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b64dde7779354fa6b9b6f6b339bdff43 : SUCCESS in 10m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3ea52389424844ab9c948343a5da857c : SUCCESS in 18m 33s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/81687138b53c4e84b970c99d71ee9fbc : SUCCESS in 19m 51s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4576644c6b9445aaef6d440e05a86ec : SUCCESS in 14m 43s","accounts_in_message":[],"_revision_number":20},{"id":"3826caa0a668e4001e708f07df12fdac8aa72dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-09-29 15:34:53.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"87054b67a7d7bd04766fdd8dd215b3c9c6443cb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 17:05:03.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a6f45a2b77d4af28398a091696bcfce\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d97d4cf1fa8c495bb6eb0a1fcd41897e : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec2c168ee76e4ac7835a3c6577e70045 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dea48d043052482eb8ae197476d6108f : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edefffb49a48468f86c037ce02e816b6 : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a73224ba7580423db1772b6b07bf1626 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e123d20f894990b1b9b0f6236c713f : SUCCESS in 10m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61d6f9c681e54d6d93fa2a1f9c24b8b9 : SUCCESS in 7m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/96a6b1b0abdd4bd4be77678c4e347610 : SUCCESS in 15m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2d57cf3107074012b82f16789d6940e5 : SUCCESS in 16m 59s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae7e81fcef3847beada26eefaa3c5c59 : SUCCESS in 53m 37s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fd44f9a75af4a68bc26d17e496fce2e : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e2cd01113a8465687d4e40dbc7eaf6a : FAILURE in 1h 17m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0829048f06ad4bb48a491619d0a48d4b : SUCCESS in 20m 17s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92936f8593124b808246d5d6ab8f0911 : FAILURE in 31m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da4925dfda264ffb99f7efe9b0db1451 : SUCCESS in 15m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16246bfc35a24c0abfa13370621fc60f : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":21},{"id":"40e3f5498297cba5e5b821bc1fef0a40ca3b34d4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-09-29 18:27:30.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"ebf7f79ff49ad6383259596508e8b6c23cf150a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-29 19:29:14.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13b3d28e74024ae2927f659182d74687\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcd763a450d84deba1c52183efc655ff : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/834692527e514e25bd5c693e2cf8e406 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7234297e3d0c4f85bdb254f79e815b72 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c904a612f04455b8135dcc8dd7c2ad0 : SUCCESS in 7m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43bb6b2724594a61882527a8e1e9e17e : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda86dbc526d48b3b5de175904bebe63 : SUCCESS in 9m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/050fcfbd9d8447b8a14fcd1a4866a7b4 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/040f14d0962748ddbf01d433b1b206f8 : SUCCESS in 15m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e9a5916079c94de580a5a31acba359eb : SUCCESS in 15m 24s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb04cae4bddb4ba7b79f3e1b9ff0982e : SUCCESS in 1h 00m 28s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50d30b7c527845c4ba39561f6b17ae85 : SUCCESS in 15m 53s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e4d1cad23c744cca82159d828ac36af : SUCCESS in 56m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63d3a0f8120e4db1911f3b54b99e660b : SUCCESS in 18m 12s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a726c33d1e1445ebfbb1b3526b67d0e : SUCCESS in 21m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4063dbddeb4a486595394ef92d40e050 : SUCCESS in 11m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa029f5bb35d418ab3f4adbaf650bfcc : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":21},{"id":"3613d0cb7584eb7dea04332bb007fa6a12949b2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-12-19 08:53:30.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"0963d8f042b56df96d75afc5faca89511016fa7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-19 10:07:24.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4ef12493457409eadc14a0f3b46fd40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76699470b84e447783fc63f3e14481f3 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4322bbed3c8540cdacbc715c7f8f2fad : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f907bcd7f33a4848b539b6ce7ee8ebdd : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd61614fe6b84a81869ed80443405734 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f4cfdc7485e4d44803913f5af8da730 : SUCCESS in 6m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/088ff73daab64f448397d21da1ead692 : SUCCESS in 4m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/513f1a90005749138a7fb988c6153d0f : SUCCESS in 9m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77446b3aa7934d6ba2dcf0b9dad3caaa : SUCCESS in 14m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2dcd742283a4481ca9c66dade8e941d2 : SUCCESS in 49m 04s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/99ec3140ae94419190ace0ab8bcc8519 : SUCCESS in 1h 01m 44s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce453cbfe30248d88424a0eb9f741bf9 : SUCCESS in 44m 10s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6796f59b1d41432a8855a972645b52f5 : SUCCESS in 58m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b1d1046093e9420e946db8176980a5c0 : SUCCESS in 15m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/74bf33f575ae4be786c172e00a77e687 : SUCCESS in 17m 25s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad3e5c6dfcda4d36a142771299fcf20e : SUCCESS in 11m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e65a6e319eb44ddab02864430a9ab11e : SUCCESS in 1h 09m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/bef9d98608004271ab523536e457e0f3 : SUCCESS in 16m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fd0f85aa6b8480d80ff2a26a6b5c85e : SUCCESS in 17m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a3f14731a1324c5fbb56ffc0e9c5f01a : SUCCESS in 35m 03s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5162ac86a964f91923ebf602a7163fd : SUCCESS in 32m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dd35b39d757b4a6a8cf2c881f9f617e7 : SUCCESS in 13m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d18de7867e10447cb75ced0170758cdb : SUCCESS in 13m 08s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8b07a92a6f384d6bb901a515af6229a2 : SUCCESS in 16m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18a2c510d2da4c72b357d655c93fec3b : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":22},{"id":"2502a3f14d676b84aa78a0f7ab105dc47d5befb3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-20 11:12:31.000000000","message":"Patch Set 22: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":22},{"id":"e69d9c5d3451208391a8953ca9bedea3f2c9056f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-20 11:13:02.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"0d44490d383059ded0c359ae1a1be1cc5b95657f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-12-20 11:54:00.000000000","message":"Patch Set 22:\n\n(6 comments)","accounts_in_message":[],"_revision_number":22},{"id":"7fa6ef605a7a3a76b1551ba9e208e7e6d8044413","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-12-20 13:38:47.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"3debc10d090199f66d3e1b851f0140f13ca75a9c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-12-20 14:50:55.000000000","message":"Patch Set 22:\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"a3dc8c5c4c1b820d7aec9b99bd2fbb40b902a5a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-01 22:15:09.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"4dd5cc165f947730dbbdb942417e93abaf3f6a5b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-01 22:15:48.000000000","message":"Patch Set 23: Backport-Candidate+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"231c05cbc8d6cd9151e0937e9a44764383b1f48a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-01 22:18:31.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"dc2c159a85ff7cc39a23abf42de47193112d1ac5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-02 00:10:57.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17460f40454c4e4786d26f8279100760\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/208b53cb52fb4631a5a626787228e407 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2069e99c04947499e7dc4019634e227 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb7ac07c9f174f1da36150eb3926fa32 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdc8e1ccb2d0474d8442e34d11f2b9c4 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a4d5c4b9028454c99e607349775f2b2 : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a1352efb880c4e2a81609210b263f48b : SUCCESS in 3m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a7feb3248fae430d88fcd595fbd31c3b : SUCCESS in 7m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/954828387c22469ab2e4df26fc816ef5 : SUCCESS in 14m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/24e69edd5a2b4ade98a95c6cf0d3b4db : SUCCESS in 55m 21s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/405589e1ce2841409fd826546fd9ff4c : SUCCESS in 50m 05s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/107466a17d0a44609e24a57993c2df52 : SUCCESS in 36m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7f92b1d52b264fc7a0863d424375a959 : FAILURE in 51m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c99c57af0126408098df5d74aae2b4c4 : SUCCESS in 13m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d89348e75bfe4ae195291008c123058d : SUCCESS in 14m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3b08cd4dd3249a399eda3cd16fa7e8a : SUCCESS in 11m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b8eb0635337649f189e80793afde124f : SUCCESS in 1h 50m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b97cb38e35ff4b88a819afcfb3dfc8d2 : SUCCESS in 52m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/00d5a030c324486da308b42ac7dd14b3 : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6a245aee8cde49598fbca100c03eb540 : SUCCESS in 13m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/1aea8c60a61e49a1b9eba76f50b17435 : SUCCESS in 24m 40s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/16adfdc4182e417fabf3b643f0f6b947 : SUCCESS in 27m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be3755a6ec58460c99e144e7f8125624 : SUCCESS in 19m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5e9ef967f44a4e0487e7a563a2ce1562 : SUCCESS in 10m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aee1fbba87c4448803965aac2d99047 : SUCCESS in 8m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3fca9c47094741e381331c80c390c0ef : SUCCESS in 16m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1e66d739217419d94ceb92de3b91677 : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":23},{"id":"4d1e616433c6c9937246adb60134d76c9b3aacdb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-02 00:29:02.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"25048c36cea9749bf35dc3491f090470e49fe649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-02 02:17:52.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/484dc4b2cbd14e7aa46111475d70a77a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5299b32cb68478786c9ce9a93a7bf49 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/254547e8576340d2beaae16ed81ff2e1 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8a2fa8b84a640d1b4b97bc707c3d3a8 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d404e0f4d7a14b8183381d1d496ac446 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/454b322da4024a5989a2bfd730f0a5d6 : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9ad9a47ff30f4e2c9ae654d9faaa8ef6 : SUCCESS in 3m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68e172beefdc441bbfb9462c79302828 : SUCCESS in 9m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2984c6c3b61341de81fd7ac0e97a3185 : SUCCESS in 11m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d588bd456dad4ded993ae3be047a4fa5 : SUCCESS in 54m 38s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a187becf120443ec92c582f61587233b : SUCCESS in 44m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/761b2906caa74beebeb87bdb544498a0 : SUCCESS in 38m 00s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0976116d6ab3421e95b2b93c92654c53 : SUCCESS in 1h 01m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3c68c09d7eba4774a5c4666fdd9608e9 : SUCCESS in 16m 41s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ee9dd45dd62b464ab01321b0cfd6c208 : SUCCESS in 13m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b79178b85d824dfda3520efa20d60d05 : SUCCESS in 11m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f279c24ec2f74a8597001a617bb35b6d : SUCCESS in 1h 42m 19s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eec28a721363450c864a4a4de41197f2 : SUCCESS in 1h 19m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4b6c7668bac14f3a98a4956ef9bf69d8 : SUCCESS in 16m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f45fe5f0b9a4940a9ec75033ab22059 : SUCCESS in 12m 07s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/edceb8a2fc234bbca31635d00ce8b389 : SUCCESS in 32m 31s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/865e73d423c9485985665071588b143e : SUCCESS in 25m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/039a39ca998842499e298e1b8d907161 : SUCCESS in 17m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/851315ffed8d4a6f911e46b1b2a7d3ee : SUCCESS in 12m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ff900b5b95240b2ae0138fe82b9b7ce : SUCCESS in 9m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a17ad04edd844caab887962795d6415a : SUCCESS in 17m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc0b0e7a654c4fc7bc108e7c5cb03c26 : SUCCESS in 10m 54s","accounts_in_message":[],"_revision_number":23},{"id":"68272c1a8e90d395b4ab92d85120a46280ac6d3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-20 23:06:03.000000000","message":"Uploaded patch set 24.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"81b3d75b8dd59929b291345fa91be681b039bb03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 00:57:46.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73fd137e7054a73899fb256168b4793\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ef50e1cb5b84cec97f928d6122f6604 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1655c938157249f087f520ab130148e4 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/568ea91aed3844b491f61eeee50df3d7 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/842e2cd7fd6a4286a65b39fbf3e4d857 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb2a67a5a05c4fe69e9ec159444e16fe : SUCCESS in 7m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f948e4297714494a08809cbe6d34d73 : SUCCESS in 3m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ba0c2e4adc0d4d7baea85770fccbc27f : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01d66defbf2e412c9b800dc05f66bccd : SUCCESS in 11m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3d7a875c940644588d627191db4e00f5 : SUCCESS in 57m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/14bb326d34c147e6a6b14f0e5deb2028 : SUCCESS in 52m 10s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f708c46c2ee34eb4bd5620867b33d6bc : SUCCESS in 40m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a45e776773f243b4b65cf7ec284ae185 : SUCCESS in 54m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0463f0d3e04d4066890786a160f765cb : SUCCESS in 16m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/26f4254df8f0418498b557438225de27 : SUCCESS in 13m 19s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb8fc9accdc745eb850a9f5e984e9449 : SUCCESS in 12m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/50de9e3af8cb413ca382afbfd3ccb470 : SUCCESS in 1h 40m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f8a10c1ef0c4a048de254c137c22a78 : SUCCESS in 1h 00m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4774b9ee36cb42e3b4334d5340ed6cb0 : SUCCESS in 16m 21s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92e60692899c46ecabb5a38f0316bc73 : SUCCESS in 12m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/72c2e19f16674b82958535c5fee55a20 : FAILURE in 30m 22s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/255b4bad66e54b9a9f79ff41f6a6b9c8 : SUCCESS in 28m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a41f0c6c7084441a86d30db6cf2fef2c : SUCCESS in 18m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b09139b5c35446f3b190de836b5e496f : SUCCESS in 13m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a718d58e37c4476f93ceb4cc2e0d119b : SUCCESS in 10m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7e2b94c1c305433e8596e61dcadddf78 : SUCCESS in 12m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38d75e3ed1fe4dd8b504924e7d722f74 : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":24},{"id":"ea1c68d15d9443af5d98045cefea359d4c48959b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-07 14:12:24.000000000","message":"Uploaded patch set 25.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"96db7ec49f9904803b6fcec316b6e5ebf756731f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 15:44:55.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a27b9b407664e6fa71961e4c9f065ed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12cb2ca622c445f4b43df9d18ce6e14d : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0359d66705046659158527887c99905 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cce71ecc747f4f1f9d4df70305ce654c : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6ee1ec9373b4653b015080d87d1b473 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7800703d92dc4ff9a545f927c6ac62c2 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/524e2a18729c41ef94ae63406819da8a : SUCCESS in 6m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d5afae6f44b459499897f2838432dc5 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/40039d040bbe4d89b84705cc0546a8b6 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fdbb68a5963748e68daac27f5e47a4f5 : SUCCESS in 15m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/76e7c26a41b5428b9a79dbf892d8029b : SUCCESS in 46m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3472385deaa84592b6212f32e5432cd8 : SUCCESS in 1h 00m 12s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0455bdb170d54f14bbc33123f024b015 : SUCCESS in 40m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/08181fabe9ce421bbbf26a90914ebe07 : SUCCESS in 59m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4769cd4eee234aafb6169ba6bb646081 : SUCCESS in 14m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/816ff7af4ce74bc69f83be265ddc256d : SUCCESS in 15m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb339980de0c4b5ba4f19c821a4c6d02 : SUCCESS in 15m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e2dd5e4f67db411fb4af8d113577b193 : SUCCESS in 1h 13m 59s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90768d9c7d80454d9433fe692016867e : SUCCESS in 1h 30m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9d6742d9d0e844df98196ff3120c0844 : SUCCESS in 16m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3b1c2642fe34b4ca239f2ae6b968862 : SUCCESS in 12m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f4878c40d10e402f8744aa5b664fc731 : SUCCESS in 32m 02s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6894b5bcd6c84a24aec6acfe0f402a0f : SUCCESS in 23m 07s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/473e8014a63241f093d30aafe6d047dd : SUCCESS in 25m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/af9d049053354e3e8339ce8c13df4839 : SUCCESS in 12m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a39f292f7f5435a82358324f86ee709 : SUCCESS in 10m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/afa67d0a58e5449db095c4d2687fae45 : SUCCESS in 38m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bbd036432e44c8e8a48c9beee13c7b9 : SUCCESS in 31m 45s","accounts_in_message":[],"_revision_number":25},{"id":"13920ea3903181c46466774755480abf347462c3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-31 15:42:20.000000000","message":"Patch Set 25: Backport-Candidate+1","accounts_in_message":[],"_revision_number":25},{"id":"b7405bf2c5a5ad73590d2ba9b2b226ffdfdd28e7","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-01 11:55:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":25},{"id":"9e40e6b8a69aa46a4ea5b1662c72398e34eae634","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-01 11:56:37.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"cc1b95852279cdc0bb7db109ae644a6107e3cc70","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 13:49:57.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"8c9d42a1e2af7b1ed595f6e5c0d4328673231992","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-28 00:21:50.000000000","message":"Uploaded patch set 26.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"f14745118886d411e6e7d553cb71680235b0b4f7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-28 00:22:39.000000000","message":"Topic vtb/master-tls-sync-kolla-ansible removed","accounts_in_message":[],"_revision_number":26},{"id":"8915978ba4eb350aea6ef5c6ab5ea483fcf69f4b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-28 00:26:16.000000000","message":"Patch Set 26: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"84b3facedaac23d0aff11a48b8bfa78717e47d77","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-28 00:31:40.000000000","message":"Uploaded patch set 27.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":27},{"id":"59678c6922503cad735c17f15ab73926561b7946","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-28 00:40:11.000000000","message":"Uploaded patch set 28.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":28},{"id":"1f009082a21a5d6c0a2633a486511174702ef8cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 03:31:06.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab49a557609f47419ef186d171308085\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cfd7bb611df4e6cb6bc315060ffd500 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556005c64daa448884d0118d4205c07d : SUCCESS in 7m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6318aa2d7ff84486be0508df3eda6241 : SUCCESS in 3m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c6a91222bcbf42d198c4ac00f3eb6ff8 : FAILURE in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b6f1dbb25eea47159306c2eb7d0ee38d : SUCCESS in 9m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/02a74e5f394d421b809a61423133c45d : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cf107d8e9dd2401cad12cce4a5b744eb : SUCCESS in 48m 59s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39d336cbc38548ceab0b211c21e85f98 : SUCCESS in 57m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1b07c4307ca44df699ae35401848e104 : FAILURE in 1h 06m 41s\n- kayobe-overcloud-tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91a249fdc90d411c9e4b4f40b1fb32a3 : FAILURE in 46m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0e7d307c5c79422bbc81e88a00e8c93e : SUCCESS in 15m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/93b2b8a165ee4c0196d2d3e4ee01d2e8 : SUCCESS in 17m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60124f96fcf64999bd568b47436ddb2f : SUCCESS in 13m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c2884d3ef383474197facec4acf0eb84 : SUCCESS in 2h 43m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b3a51dee547477ebe1b07c43754fd08 : FAILURE in 1h 18m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b607a0a5c2bc4a59a2e8af1401dc8733 : SUCCESS in 20m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab35f6a8554947b3847c7282862382f1 : SUCCESS in 13m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2796a90b26cc426d9f88a6ca0e19d858 : SUCCESS in 26m 49s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e678c730ca5045c8b5d3b97afb3d1356 : SUCCESS in 26m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/228605d7161d4788b8fe523b56bb333e : SUCCESS in 21m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/45fa7041499b4f0e9716dbe5c25ffb37 : SUCCESS in 12m 41s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89bfa388b25f40eba5184a0f055d808c : SUCCESS in 10m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/312e6c0e62494942a268e202524f0300 : SUCCESS in 17m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3e9b1ea9d0843d5b0d169fd9df36f1f : SUCCESS in 14m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6425c1b1384428cb7b05a8d5f6bfa13 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c623ad240df3462c9b4c39e5dc62b325 : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41be2d2b2db24892a62ae1de42f38807 : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":28},{"id":"31911e46993e85f688f5c691c82fa95120a94711","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-28 09:20:46.000000000","message":"Uploaded patch set 29.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"5c9b506a2fe989334a81c523ee46a33fa2c9d2be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 11:53:29.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9236efdbc1ec480ba73ab8acdbd9eeb8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/817bb0aeb312479f87f6414290c60b2d : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbaa5f31d39441e080ac108e44e90806 : SUCCESS in 8m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec702b8345a3404abc784f544a35eaf6 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79cfc7591b664f7c9b7eb1a675b2b08d : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70a3c00ec6f14dd4925b854c8dca3601 : SUCCESS in 12m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/187da00ed38d4de2a0b13da90b097b5c : SUCCESS in 1h 01m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2af190de683840be9247fb8497afe533 : SUCCESS in 1h 03m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3067d26966b427a81f75f2527d5951a : FAILURE in 16m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/dd6e5d3120134841969592ec10c416cf : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fce778ecd6f4f879e2d4abb3bc6c984 : SUCCESS in 58m 32s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1481e1500027427285b68673a4699b53 : SUCCESS in 14m 36s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0fd51dcc37604be5b2bf798507532c81 : SUCCESS in 19m 02s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/946273f4c1d345449d9e5e3f6c06176f : SUCCESS in 14m 09s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0afad9f7215a4d499bad542deffdf8e0 : SUCCESS in 2h 25m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba02f5ac774946a087825bf22369abc2 : SUCCESS in 2h 26m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f0f082fcee8341d985d4d9f84c1913ba : SUCCESS in 15m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf8a1f71b63f4844a302196834a7b1f3 : SUCCESS in 12m 02s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/93c73aead6d84eac8f02706747fd1856 : SUCCESS in 35m 35s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/608fdbf5fc85499dbb0df973b425f00b : SUCCESS in 33m 24s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b5de168279e4fa8b581815230b0c112 : SUCCESS in 26m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/34fe42d8616a493c9ebca6fe48f09c3e : SUCCESS in 12m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f751e290653442cda6ae97c657f651f7 : SUCCESS in 10m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d8bf11221ce843d3b103c84f6163cde6 : SUCCESS in 17m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daedad4cb2514dc8bb4fa84a032a2727 : SUCCESS in 13m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/836983f3b2a24bb98cdf0be78eb4ce04 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/faa7ca1a84504d9bb860e335ea000f53 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac9c0efcb20245ee9c4aba1232f983d2 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":29},{"id":"87bebdb796e00ddd3cd81e0679dc79a08eca7f51","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-28 12:51:48.000000000","message":"Patch Set 29: Backport-Candidate+1\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":29},{"id":"18c2ab4cc0dfa2f57d898c676c8bd104c637fc8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 15:36:25.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e52b9c98482740b4aa5cc797a0ec3aa3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e564f60d560d4dadb96aa963ace031b8 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caab5233456c40f7bbf8248ec3942613 : SUCCESS in 7m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d5fdb01d8eb45c8aa726cd24e7f08a7 : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9799811401b3481a89f43a93e807f4c2 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cddb556caf6443babb46d913da3973ae : SUCCESS in 14m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8d62178e620640109bebad3f0e3af69c : SUCCESS in 1h 00m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e16c28eb77a9464097b8389254087b05 : SUCCESS in 1h 04m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abe6b5703cde465081f2802218da5b21 : SUCCESS in 48m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9ed07df1956f4a3ab731d2bdc4d4c536 : SUCCESS in 1h 15m 32s\n- kayobe-overcloud-tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d04072e52e504be2a0f0958ec21c2542 : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c4c188e5ae04417fbd31aa84b8f51627 : SUCCESS in 16m 37s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8b0afea63cf24c68bfe12ec34827b426 : SUCCESS in 16m 41s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f885ab651a64f0e81b6d6cd9a79decd : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/346dc8ea205e400baf906b0cb431375f : SUCCESS in 2h 39m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/461bef9dc0bd4e74b0214a482103521a : SUCCESS in 2h 14m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9ffc05d29f23403a81aaff78ced4237a : SUCCESS in 16m 57s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/235c053d99ab4e10a63356fbca2eeb8c : SUCCESS in 15m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/eefbe3f6f54d4e6da60a9e87287fcd7d : SUCCESS in 31m 54s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5122d0c6c00147cd9494517aaf14b960 : SUCCESS in 26m 18s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02908b67a0f94b14aeedfea1c15622a4 : SUCCESS in 26m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a74dcc5ecdb7426081ed8f1a58ef70c5 : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7c5f7c1448741dd80fda5c818395493 : SUCCESS in 10m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cd9756f383dd4fcd8803760999a35b9b : SUCCESS in 17m 38s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/346417a51a894b29b6ec49eaa86af72f : SUCCESS in 14m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/952036e67b6741d083c70a8eb4add19c : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9436cb6a37f745888af3bb73ee57a970 : SUCCESS in 4m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/482b873aa1d64acf84ebf7b90544a653 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":29},{"id":"10c0d4e9ab72a71672c96b40b4373db9a8929bd8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-12-08 16:03:39.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"8da220aa3b526b3fe008afb4ab2a17e37fda50c1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-01-03 10:26:22.000000000","message":"Patch Set 29:\n\n(8 comments)","accounts_in_message":[],"_revision_number":29},{"id":"a63ac9c3c76fb562eb368ad541cdcd5404b864f2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-09 15:33:07.000000000","message":"Patch Set 29:\n\n(9 comments)","accounts_in_message":[],"_revision_number":29},{"id":"4a63f4fdca2aef31c7803519b6a8b64d6a221cf9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-04-09 15:34:05.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"bdb6061515e78892d26175f01706ff19093435a7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-11-06 14:11:55.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"f36b751c5cc2559f6e53bd7e3ef0ca79d611ddb5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-06 14:40:33.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"a870c4ed7d4062046d944ba1c601cc41fb4028ad","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-06 14:40:42.000000000","message":"Patch Set 29: Code-Review-1","accounts_in_message":[],"_revision_number":29},{"id":"b4b1ed221184bf49b4cf2b9da3dbe07b80ea8c63","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-11-06 14:51:29.000000000","message":"Patch Set 29:\n\n(2 comments)","accounts_in_message":[],"_revision_number":29},{"id":"dddf688aa85715b2932a8124eafb1b7b09eb55f4","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-12-01 11:57:55.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_14826\u003e.","accounts_in_message":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"_revision_number":29},{"id":"bc31cf45720fbfb01d944ad83e5c10c6c63afdec","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-10 12:38:41.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":29},{"id":"5b4cd561cbe6cb17fe78e3c7020f0f3ecdf55174","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-04 15:22:09.000000000","message":"Patch Set 29:\n\n(3 comments)","accounts_in_message":[],"_revision_number":29},{"id":"650781f5a443f37db15653ecd1a9ee043e7dc793","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-04 15:25:23.000000000","message":"Uploaded patch set 30.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":30},{"id":"39094652b5b92d8e2c52cabc86976fdd5488030a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-04 15:31:29.000000000","message":"Uploaded patch set 31.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":31},{"id":"7debfe0a21644b3e4e413858d8a68f53cb073f6f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-06-04 17:24:55.000000000","message":"Patch Set 31:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":31},{"id":"c0db75201b22cf7b93bf291c6185a77a46478a65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-04 17:26:32.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fa400c4ac85451ba43cb54f2f3a2351\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52127ca974a44c1dbd4ad1f0b3fab776 : SUCCESS in 5m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2bdff637b7be4f0a962f6758328ce820 : SUCCESS in 6m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c84bbe71abb148358af171dd933c684d : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0467049d28554787a9e62b241f718722 : SUCCESS in 6m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04cc8048f33d40c8a3c6b8455938ef49 : SUCCESS in 3m 23s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/79c1d21e511e482599f2a1e0a898622a : SUCCESS in 4m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c90e9e2d35341beb0f1ba8128b210d5 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3bffb462c58e4b0ca00fcedeedd77391 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d162e2309ad04250bb561705d2481dfb : SUCCESS in 8m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ed480346835e4011bdb2e568957d8431 : SUCCESS in 6m 56s\n- kayobe-ansible-control-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/6d476818872d478bb71b98ad35ae8919 : SUCCESS in 22m 43s\n- kayobe-ansible-control-host-configure-rocky10-podman https://zuul.opendev.org/t/openstack/build/7e937d37a3c6496f9cbb02edd8b744ef : SUCCESS in 23m 43s\n- kayobe-ansible-control-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e585b8ff32374addab5f4764ec5e5e94 : SUCCESS in 17m 21s\n- kayobe-ansible-control-host-configure-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ba2f37e2b1bb46019c7bb032a6c00708 : SUCCESS in 17m 52s\n- kayobe-infra-vm-rocky10 https://zuul.opendev.org/t/openstack/build/5a300c82879b4f578aae6e8fb47ccc71 : SUCCESS in 19m 50s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a553f613701c433e810bf392b3852ba9 : SUCCESS in 20m 00s\n- kayobe-overcloud-host-configure-rocky10 https://zuul.opendev.org/t/openstack/build/05ff08a2b1ce48e88706bdac95fe6428 : SUCCESS in 13m 06s\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfee1f13d64f4c0bb3bb15603aa24f1c : SUCCESS in 18m 03s\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/c8f6387d07a54fddab9c6d51debda96f : SUCCESS in 57m 47s\n- kayobe-overcloud-rocky10-podman https://zuul.opendev.org/t/openstack/build/6e68f27c4b6f4fd6ae6a4a03b31e1797 : SUCCESS in 1h 01m 08s\n- kayobe-overcloud-tls-rocky10 https://zuul.opendev.org/t/openstack/build/167ba6a7e2f541698b3bb3b89d6b1860 : SUCCESS in 1h 01m 25s\n- kayobe-overcloud-tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f21c8376a1ad45c1ab741f776a570264 : SUCCESS in 1h 01m 10s\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9d3a0ed8f2a74416a2652462738b69b6 : SUCCESS in 1h 01m 05s\n- kayobe-overcloud-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/8f76f6870ec74e3e8df3cfd3d743d446 : SUCCESS in 1h 00m 26s\n- kayobe-overcloud-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/ada5bd7910f34bd1811c7ea4ddd9818f : SUCCESS in 1h 43m 29s\n- kayobe-overcloud-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d6fa5d1118ef426e830774382f892094 : SUCCESS in 1h 51m 54s\n- kayobe-seed-rocky10 https://zuul.opendev.org/t/openstack/build/f689adab4bd347d5bc4591c8ce500936 : SUCCESS in 19m 11s\n- kayobe-seed-rocky10-podman https://zuul.opendev.org/t/openstack/build/ea7313fbbbe14e9eba221562d392841d : SUCCESS in 15m 42s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53e63fc2e47241a998fb57188e16d3bf : SUCCESS in 28m 36s\n- kayobe-seed-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/3e35f6b818154669b563967f92f1b4af : SUCCESS in 20m 41s\n- kayobe-seed-upgrade-rocky10 https://zuul.opendev.org/t/openstack/build/9e555ad96c4f4f339d24761c9e6eb9bc : SUCCESS in 42m 44s\n- kayobe-seed-upgrade-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5482704234b24640b948ccee965dd473 : SUCCESS in 20m 15s\n- kayobe-seed-vm-rocky10 https://zuul.opendev.org/t/openstack/build/ee52fb62db364b348539e936216b9a1f : SUCCESS in 25m 13s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f7f56d7494204ac29c2ba37529b81ec5 : SUCCESS in 19m 37s","accounts_in_message":[],"_revision_number":31}],"current_revision_number":31,"current_revision":"fabfc3d6dc3c77f1430bbf19242b253f1892e437","revisions":{"bb99adf5b619047f47db9d8f32e4e6b32bfb573b":{"kind":"REWORK","_number":1,"created":"2021-05-30 18:31:57.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/1"}}},"commit":{"parents":[{"commit":"1e14fa3a24fa64b84cc993c985e696a9cba78dea","subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e14fa3a24fa64b84cc993c985e696a9cba78dea"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"subject":"TlS certificates management sync with Kolla-Ansible","message":"TlS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf.\n\nReplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set.\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb99adf5b619047f47db9d8f32e4e6b32bfb573b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bb99adf5b619047f47db9d8f32e4e6b32bfb573b"}]},"branch":"refs/heads/master"},"adee001c424362ec49d2c39387704c0c797624a4":{"kind":"REWORK","_number":2,"created":"2021-06-01 09:35:39.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/2"}}},"commit":{"parents":[{"commit":"1e14fa3a24fa64b84cc993c985e696a9cba78dea","subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e14fa3a24fa64b84cc993c985e696a9cba78dea"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-01 09:26:35.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adee001c424362ec49d2c39387704c0c797624a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/adee001c424362ec49d2c39387704c0c797624a4"}]},"branch":"refs/heads/master"},"01d43c2a31953969fb41f8c8bff222f6d3e2c5ac":{"kind":"REWORK","_number":3,"created":"2021-06-01 11:22:05.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/3"}}},"commit":{"parents":[{"commit":"1e14fa3a24fa64b84cc993c985e696a9cba78dea","subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e14fa3a24fa64b84cc993c985e696a9cba78dea"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-01 11:21:43.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01d43c2a31953969fb41f8c8bff222f6d3e2c5ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01d43c2a31953969fb41f8c8bff222f6d3e2c5ac"}]},"branch":"refs/heads/master"},"78fd46a23445f7dfd0bd55382083bea51ff5583b":{"kind":"REWORK","_number":4,"created":"2021-06-01 12:17:49.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/4"}}},"commit":{"parents":[{"commit":"1e14fa3a24fa64b84cc993c985e696a9cba78dea","subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e14fa3a24fa64b84cc993c985e696a9cba78dea"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-01 12:17:35.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78fd46a23445f7dfd0bd55382083bea51ff5583b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78fd46a23445f7dfd0bd55382083bea51ff5583b"}]},"branch":"refs/heads/master"},"723e62702275259d3dbb4701e1d5d91e35d8f512":{"kind":"REWORK","_number":5,"created":"2021-06-01 14:35:32.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/5"}}},"commit":{"parents":[{"commit":"1e14fa3a24fa64b84cc993c985e696a9cba78dea","subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e14fa3a24fa64b84cc993c985e696a9cba78dea"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-01 14:34:27.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/723e62702275259d3dbb4701e1d5d91e35d8f512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/723e62702275259d3dbb4701e1d5d91e35d8f512"}]},"branch":"refs/heads/master"},"015129f16fa4cabbf75168d3b580c0a51ab04cb1":{"kind":"REWORK","_number":6,"created":"2021-06-09 23:28:41.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/6"}}},"commit":{"parents":[{"commit":"1e14fa3a24fa64b84cc993c985e696a9cba78dea","subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e14fa3a24fa64b84cc993c985e696a9cba78dea"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-09 23:28:17.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and overcloud hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/015129f16fa4cabbf75168d3b580c0a51ab04cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/015129f16fa4cabbf75168d3b580c0a51ab04cb1"}]},"branch":"refs/heads/master"},"568a53b19266151fd7e3012d71b9d90aa3391f09":{"kind":"REWORK","_number":7,"created":"2021-06-10 08:25:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/7"}}},"commit":{"parents":[{"commit":"1e14fa3a24fa64b84cc993c985e696a9cba78dea","subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e14fa3a24fa64b84cc993c985e696a9cba78dea"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-10 07:51:51.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/568a53b19266151fd7e3012d71b9d90aa3391f09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/568a53b19266151fd7e3012d71b9d90aa3391f09"}]},"branch":"refs/heads/master"},"95ac0534a265795d1cb4049cc8a219566b78b1ce":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-06-15 10:24:05.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/8"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-15 10:23:39.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95ac0534a265795d1cb4049cc8a219566b78b1ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95ac0534a265795d1cb4049cc8a219566b78b1ce"}]},"branch":"refs/heads/master"},"d06a03c9e2c3cb1b2eca34ddc2b77caf11635869":{"kind":"REWORK","_number":9,"created":"2021-06-15 15:49:04.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/9"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-15 15:47:33.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d06a03c9e2c3cb1b2eca34ddc2b77caf11635869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d06a03c9e2c3cb1b2eca34ddc2b77caf11635869"}]},"branch":"refs/heads/master"},"72ed4e93ac198d22bb3f3f7e25d6b64d9a1f79b6":{"kind":"REWORK","_number":10,"created":"2021-06-16 09:19:12.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/10"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-16 09:18:42.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72ed4e93ac198d22bb3f3f7e25d6b64d9a1f79b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72ed4e93ac198d22bb3f3f7e25d6b64d9a1f79b6"}]},"branch":"refs/heads/master"},"9a711509926a5994c1decd61b9871ea6c13d33a8":{"kind":"REWORK","_number":11,"created":"2021-06-30 09:24:17.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/11"}}},"commit":{"parents":[{"commit":"22b0feb2594929fc4a1604c0ebc3c921d7e41c6b","subject":"Set defaults for openstack_cacert","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22b0feb2594929fc4a1604c0ebc3c921d7e41c6b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-30 09:21:29.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nDepends-On: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a711509926a5994c1decd61b9871ea6c13d33a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a711509926a5994c1decd61b9871ea6c13d33a8"}]},"branch":"refs/heads/master"},"9f1a0ea66659fcbf30f14c9979e8fa118fb91540":{"kind":"REWORK","_number":12,"created":"2021-06-30 11:50:10.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/12"}}},"commit":{"parents":[{"commit":"22b0feb2594929fc4a1604c0ebc3c921d7e41c6b","subject":"Set defaults for openstack_cacert","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22b0feb2594929fc4a1604c0ebc3c921d7e41c6b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-30 11:48:46.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nDepends-On: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f1a0ea66659fcbf30f14c9979e8fa118fb91540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f1a0ea66659fcbf30f14c9979e8fa118fb91540"}]},"branch":"refs/heads/master"},"513d407b9a7680218b958cb34f1e150ac2c9acf0":{"kind":"REWORK","_number":13,"created":"2021-06-30 12:15:57.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/13"}}},"commit":{"parents":[{"commit":"22b0feb2594929fc4a1604c0ebc3c921d7e41c6b","subject":"Set defaults for openstack_cacert","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22b0feb2594929fc4a1604c0ebc3c921d7e41c6b"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-30 12:14:54.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nDepends-On: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/513d407b9a7680218b958cb34f1e150ac2c9acf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/513d407b9a7680218b958cb34f1e150ac2c9acf0"}]},"branch":"refs/heads/master"},"e7a56c7b8e8ddc88655a88a13cfced6fd0cf1458":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-08-20 08:55:58.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/14"}}},"commit":{"parents":[{"commit":"0b05eb95b8f24050514ab44d3ee4d4353f2810ea","subject":"Set defaults for openstack_cacert","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b05eb95b8f24050514ab44d3ee4d4353f2810ea"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-08-20 08:55:11.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nDepends-On: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7a56c7b8e8ddc88655a88a13cfced6fd0cf1458"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e7a56c7b8e8ddc88655a88a13cfced6fd0cf1458"}]},"branch":"refs/heads/master"},"f0682266b486790a84a7add635637bcc8dce9a38":{"kind":"REWORK","_number":15,"created":"2021-10-15 09:33:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/15"}}},"commit":{"parents":[{"commit":"d8360ea357ae62a135b7c94cc19ead72cfb5b521","subject":"Set defaults for openstack_cacert","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8360ea357ae62a135b7c94cc19ead72cfb5b521"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 18:22:51.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-15 09:33:26.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nDepends-On: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0682266b486790a84a7add635637bcc8dce9a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0682266b486790a84a7add635637bcc8dce9a38"}]},"branch":"refs/heads/master"},"a678af75f53413fd38e518fd8fa0d4ca637775a1":{"kind":"REWORK","_number":16,"created":"2022-01-12 13:11:21.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/16"}}},"commit":{"parents":[{"commit":"561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7","subject":"Fix Sphinx syntax typo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/561ee09c03223b63d0f6d6ce8ae56fee6ddb80d7"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:09:17.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nAlso this change set defaults for openstack_cacert.\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a678af75f53413fd38e518fd8fa0d4ca637775a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a678af75f53413fd38e518fd8fa0d4ca637775a1"}]},"branch":"refs/heads/master"},"c218ab6faba958d168919bec66b78014f529bea6":{"kind":"REWORK","_number":17,"created":"2022-02-09 22:08:58.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/17"}}},"commit":{"parents":[{"commit":"112a08a37d8820db4bb4f50e69719ea483189d64","subject":"Merge \"dnf: support exclude and includepkgs arguments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/112a08a37d8820db4bb4f50e69719ea483189d64"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-09 22:04:47.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nAlso this change set defaults for openstack_cacert.\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nAlso this change partially reverts\nI9e99c04173acb7adb0d0e367b6db388b86baf5f2 to restore kolla-host.yml and\ndrop only iSCSI related task.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c218ab6faba958d168919bec66b78014f529bea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c218ab6faba958d168919bec66b78014f529bea6"}]},"branch":"refs/heads/master"},"61f9afd162e1e63b73e027b82c9de52211b05ca2":{"kind":"REWORK","_number":18,"created":"2022-04-04 22:43:03.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/18"}}},"commit":{"parents":[{"commit":"60fc11fa265a59840918a47d2cca110c02889f76","subject":"[release] Sync Kolla Ansible inventory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60fc11fa265a59840918a47d2cca110c02889f76"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-04 22:34:15.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nAlso this change set defaults for openstack_cacert.\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nAlso this change partially reverts\nI9e99c04173acb7adb0d0e367b6db388b86baf5f2 to restore kolla-host.yml and\ndrop only iSCSI related task.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61f9afd162e1e63b73e027b82c9de52211b05ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61f9afd162e1e63b73e027b82c9de52211b05ca2"}]},"branch":"refs/heads/master"},"fa911b116cfc1c5b59d98deb9138d7e46eec27ea":{"kind":"REWORK","_number":19,"created":"2022-04-04 22:56:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/19"}}},"commit":{"parents":[{"commit":"60fc11fa265a59840918a47d2cca110c02889f76","subject":"[release] Sync Kolla Ansible inventory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60fc11fa265a59840918a47d2cca110c02889f76"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-04 22:56:23.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nAlso this change set defaults for openstack_cacert.\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nAlso this change partially reverts\nI9e99c04173acb7adb0d0e367b6db388b86baf5f2 to restore kolla-host.yml and\ndrop only iSCSI related task.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa911b116cfc1c5b59d98deb9138d7e46eec27ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa911b116cfc1c5b59d98deb9138d7e46eec27ea"}]},"branch":"refs/heads/master"},"6f897b25a23dde2b83bb34630f32914e5a15bfec":{"kind":"REWORK","_number":20,"created":"2022-04-06 09:36:37.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/20"}}},"commit":{"parents":[{"commit":"60fc11fa265a59840918a47d2cca110c02889f76","subject":"[release] Sync Kolla Ansible inventory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60fc11fa265a59840918a47d2cca110c02889f76"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-04-06 09:35:58.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nAlso this change set defaults for openstack_cacert.\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nAlso this change partially reverts\nI9e99c04173acb7adb0d0e367b6db388b86baf5f2 to restore kolla-host.yml and\ndrop only iSCSI related task.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f897b25a23dde2b83bb34630f32914e5a15bfec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f897b25a23dde2b83bb34630f32914e5a15bfec"}]},"branch":"refs/heads/master"},"b83d42bc1a8d32bb85efedc2bace9c189ef4845f":{"kind":"REWORK","_number":21,"created":"2022-09-29 15:34:53.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/21"}}},"commit":{"parents":[{"commit":"608951fcce38428b155030badcff15a5a6a0445c","subject":"Merge \"Add new DIB upper constraints variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/608951fcce38428b155030badcff15a5a6a0445c"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-09-29 15:34:16.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nAlso this change set defaults for openstack_cacert.\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nAlso this change partially reverts\nI9e99c04173acb7adb0d0e367b6db388b86baf5f2 to restore kolla-host.yml and\ndrop only iSCSI related task.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b83d42bc1a8d32bb85efedc2bace9c189ef4845f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b83d42bc1a8d32bb85efedc2bace9c189ef4845f"}]},"branch":"refs/heads/master"},"5247cef39efe9da86eacbe93b090f8644c4a722a":{"kind":"REWORK","_number":22,"created":"2022-12-19 08:53:30.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/22"}}},"commit":{"parents":[{"commit":"fbf000689559c434157d49b147da60725ba9562a","subject":"Run handlers before configuring veth interfaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbf000689559c434157d49b147da60725ba9562a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-12-19 08:53:04.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nAlso this change set defaults for openstack_cacert.\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nAlso this change partially reverts\nI9e99c04173acb7adb0d0e367b6db388b86baf5f2 to restore kolla-host.yml and\ndrop only iSCSI related task.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5247cef39efe9da86eacbe93b090f8644c4a722a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5247cef39efe9da86eacbe93b090f8644c4a722a"}]},"branch":"refs/heads/master"},"638a3866087e408a9dbb3387b2f4394dbe654cea":{"kind":"REWORK","_number":23,"created":"2023-04-01 22:15:09.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/23"}}},"commit":{"parents":[{"commit":"fbf000689559c434157d49b147da60725ba9562a","subject":"Run handlers before configuring veth interfaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbf000689559c434157d49b147da60725ba9562a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-01 22:14:40.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nAlso this change set defaults for openstack_cacert.\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nAlso this change partially reverts\nI9e99c04173acb7adb0d0e367b6db388b86baf5f2 to restore kolla-host.yml and\ndrop only iSCSI related task.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/638a3866087e408a9dbb3387b2f4394dbe654cea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/638a3866087e408a9dbb3387b2f4394dbe654cea"}]},"branch":"refs/heads/master"},"cf27f45ece55b7d290d4a9ed01efec1c0b4c73d1":{"kind":"REWORK","_number":24,"created":"2023-04-20 23:06:03.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/24"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-20 23:05:41.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nAlso this change set defaults for openstack_cacert.\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nAlso this change partially reverts\nI9e99c04173acb7adb0d0e367b6db388b86baf5f2 to restore kolla-host.yml and\ndrop only iSCSI related task.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf27f45ece55b7d290d4a9ed01efec1c0b4c73d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf27f45ece55b7d290d4a9ed01efec1c0b4c73d1"}]},"branch":"refs/heads/master"},"181d62d06e4de1ef4bbd8d9c45736f0cd80fccb5":{"kind":"REWORK","_number":25,"created":"2023-06-07 14:12:24.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/25"}}},"commit":{"parents":[{"commit":"cde32faa4e79228a90b35b7f5cbc0d872e0c7567","subject":"Merge \"CI: Fix kayobe-seed-images-* jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cde32faa4e79228a90b35b7f5cbc0d872e0c7567"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-07 14:11:53.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nThis change is a part of If195d5402579cee9a14b91f63f5fde84eb84cccf fixes\nreplaced \u0027kolla_external_fqdn_cacert\u0027 and \u0027kolla_internal_fqdn_cacert\u0027\nwith \u0027kolla_admin_openrc_cacert\u0027, which by default is not set. Also\n\u0027OS_CACERT\u0027 is now set to the value of \u0027kolla_admin_openrc_cacert\u0027 in\nthe both \u0027admin-openrc.sh\u0027 and \u0027public-openrc.sh\u0027 files.\n\nAlso this change sync the usage of the other TLS related variables like\n\u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Adds tests\nfor these variables in the \u0027kolla-ansible\u0027 role and updates\ndocumentation.\n\nThe \u0027kolla-ansible\u0027 tasks are also updates CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled.\n\nAlso this change set defaults for openstack_cacert.\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nAlso this change partially reverts\nI9e99c04173acb7adb0d0e367b6db388b86baf5f2 to restore kolla-host.yml and\ndrop only iSCSI related task.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/181d62d06e4de1ef4bbd8d9c45736f0cd80fccb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/181d62d06e4de1ef4bbd8d9c45736f0cd80fccb5"}]},"branch":"refs/heads/master"},"f7ce439924dc636d068e7725660832f07a6b93bd":{"kind":"REWORK","_number":26,"created":"2023-11-28 00:21:50.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/26"}}},"commit":{"parents":[{"commit":"14bcaba0a360537f0575ae6545d91d860d4c6399","subject":"Merge \"Add support for setting Docker daemon debug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14bcaba0a360537f0575ae6545d91d860d4c6399"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-28 00:15:23.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nAs the follow-up on If195d5402579cee9a14b91f63f5fde84eb84cccf and\nI9e1cc20579cf80525d6ef732a1aac99a65bc171b. This change adds some\nneeded system defaults and common settings to manage TLS configuration.\n\nThis change syncs the usage of the forgotten TLS-related variables\nlike \u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Make the\ndefaults for the \u0027kolla_public_openrc_cacert\u0027 variable the same way as\n\u0027kolla_admin_openrc_cacert\u0027 for the \u0027globals.yml\u0027 and\n\u0027public-openrc.sh\u0027 files. Adds tests for these variables in the\n\u0027kolla-ansible\u0027 role and updates documentation.\n\nThe \u0027kolla-ansible\u0027 tasks now update CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled, to achieve this we restore\nkolla-host.yml removed in I9e99c04173acb7adb0d0e367b6db388b86baf5f2\nwith a new content.\n\nAlso, this change set the system defaults for \u0027openstack_cacert\u0027.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f7ce439924dc636d068e7725660832f07a6b93bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f7ce439924dc636d068e7725660832f07a6b93bd"}]},"branch":"refs/heads/master"},"eba10669dd2bca1a9196ee2430d5698a330efbaa":{"kind":"REWORK","_number":27,"created":"2023-11-28 00:31:40.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/27"}}},"commit":{"parents":[{"commit":"14bcaba0a360537f0575ae6545d91d860d4c6399","subject":"Merge \"Add support for setting Docker daemon debug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14bcaba0a360537f0575ae6545d91d860d4c6399"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-28 00:31:23.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nAs the follow-up on If195d5402579cee9a14b91f63f5fde84eb84cccf and\nI9e1cc20579cf80525d6ef732a1aac99a65bc171b. This change adds some\nneeded system defaults and common settings to manage TLS configuration.\n\nThis change syncs the usage of the forgotten TLS-related variables\nlike \u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Make the\ndefaults for the \u0027kolla_public_openrc_cacert\u0027 variable the same way as\n\u0027kolla_admin_openrc_cacert\u0027 for the \u0027globals.yml\u0027 and\n\u0027public-openrc.sh\u0027 files. Adds tests for these variables in the\n\u0027kolla-ansible\u0027 role and updates documentation.\n\nThe \u0027kolla-ansible\u0027 tasks now update CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled, to achieve this we restore\nkolla-host.yml removed in I9e99c04173acb7adb0d0e367b6db388b86baf5f2\nwith a new content.\n\nAlso, this change set the system defaults for \u0027openstack_cacert\u0027.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eba10669dd2bca1a9196ee2430d5698a330efbaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eba10669dd2bca1a9196ee2430d5698a330efbaa"}]},"branch":"refs/heads/master"},"20d387ae84d84f198f4306309ec46dec602a8145":{"kind":"REWORK","_number":28,"created":"2023-11-28 00:40:11.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/28"}}},"commit":{"parents":[{"commit":"14bcaba0a360537f0575ae6545d91d860d4c6399","subject":"Merge \"Add support for setting Docker daemon debug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14bcaba0a360537f0575ae6545d91d860d4c6399"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-28 00:39:55.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nAs the follow-up on If195d5402579cee9a14b91f63f5fde84eb84cccf and\nI9e1cc20579cf80525d6ef732a1aac99a65bc171b. This change adds some\nneeded system defaults and common settings to manage TLS configuration.\n\nThis change syncs the usage of the forgotten TLS-related variables\nlike \u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Make the\ndefaults for the \u0027kolla_public_openrc_cacert\u0027 variable the same way as\n\u0027kolla_admin_openrc_cacert\u0027 for the \u0027globals.yml\u0027 and\n\u0027public-openrc.sh\u0027 files. Adds tests for these variables in the\n\u0027kolla-ansible\u0027 role and updates documentation.\n\nThe \u0027kolla-ansible\u0027 tasks now update CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled, to achieve this we restore\nkolla-host.yml removed in I9e99c04173acb7adb0d0e367b6db388b86baf5f2\nwith a new content.\n\nAlso, this change set the system defaults for \u0027openstack_cacert\u0027.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20d387ae84d84f198f4306309ec46dec602a8145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20d387ae84d84f198f4306309ec46dec602a8145"}]},"branch":"refs/heads/master"},"548760482b0d2f07a59ebe35b8a9328a8c45c37d":{"kind":"REWORK","_number":29,"created":"2023-11-28 09:20:46.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/29"}}},"commit":{"parents":[{"commit":"14bcaba0a360537f0575ae6545d91d860d4c6399","subject":"Merge \"Add support for setting Docker daemon debug\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/14bcaba0a360537f0575ae6545d91d860d4c6399"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-01-12 13:01:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-11-28 09:20:37.000000000","tz":180},"subject":"TLS certificates management sync with Kolla-Ansible","message":"TLS certificates management sync with Kolla-Ansible\n\nAs the follow-up on If195d5402579cee9a14b91f63f5fde84eb84cccf and\nI9e1cc20579cf80525d6ef732a1aac99a65bc171b. This change adds some\nneeded system defaults and common settings to manage TLS configuration.\n\nThis change syncs the usage of the forgotten TLS-related variables\nlike \u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Make the\ndefaults for the \u0027kolla_public_openrc_cacert\u0027 variable the same way as\n\u0027kolla_admin_openrc_cacert\u0027 for the \u0027globals.yml\u0027 and\n\u0027public-openrc.sh\u0027 files. Adds tests for these variables in the\n\u0027kolla-ansible\u0027 role and updates documentation.\n\nThe \u0027kolla-ansible\u0027 tasks now update CA certificates bundle with\nprovided CA files on the seed and controller hosts when the\n\u0027kolla_copy_ca_into_containers\u0027 is enabled, to achieve this we restore\nkolla-host.yml removed in I9e99c04173acb7adb0d0e367b6db388b86baf5f2\nwith a new content.\n\nAlso, this change set the system defaults for \u0027openstack_cacert\u0027.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/548760482b0d2f07a59ebe35b8a9328a8c45c37d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/548760482b0d2f07a59ebe35b8a9328a8c45c37d"}]},"branch":"refs/heads/master"},"92fb716a3cf1df459e79ecae919a9048302c64e4":{"kind":"REWORK","_number":30,"created":"2026-06-04 15:25:23.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/30"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-06-04 15:17:09.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-06-04 15:24:49.000000000","tz":180},"subject":"Sync TLS configuration with Kolla-Ansible","message":"Sync TLS configuration with Kolla-Ansible\n\nThis change syncs the usage of the forgotten TLS-related variables\nlike \u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Make the\ndefaults for the \u0027kolla_public_openrc_cacert\u0027 variable the same way as\n\u0027kolla_admin_openrc_cacert\u0027 for the \u0027globals.yml\u0027 and\n\u0027public-openrc.sh\u0027 files. Adds tests for these variables in the\n\u0027kolla-ansible\u0027 role and updates documentation.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92fb716a3cf1df459e79ecae919a9048302c64e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/92fb716a3cf1df459e79ecae919a9048302c64e4"}]},"branch":"refs/heads/master"},"fabfc3d6dc3c77f1430bbf19242b253f1892e437":{"kind":"REWORK","_number":31,"created":"2026-06-04 15:31:29.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/97/793697/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/793697/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/793697/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/793697/31"}}},"commit":{"parents":[{"commit":"95e47148f1a2aa14429fe72c3879d7e6157789e9","subject":"Merge \"CI: Stop using image from Docker Hub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/95e47148f1a2aa14429fe72c3879d7e6157789e9"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-06-04 15:31:21.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2026-06-04 15:31:21.000000000","tz":180},"subject":"Sync TLS configuration with Kolla-Ansible","message":"Sync TLS configuration with Kolla-Ansible\n\nThis change syncs the usage of the forgotten TLS-related variables\nlike \u0027kolla_enable_tls_backend\u0027, \u0027kolla_verify_tls_backend\u0027 and\n\u0027kolla_copy_ca_into_containers\u0027 from Kolla-Ansible to Kayobe. Make the\ndefaults for the \u0027kolla_public_openrc_cacert\u0027 variable the same way as\n\u0027kolla_admin_openrc_cacert\u0027 for the \u0027globals.yml\u0027 and\n\u0027public-openrc.sh\u0027 files. Adds tests for these variables in the\n\u0027kolla-ansible\u0027 role and updates documentation.\n\nChange-Id: If6c34db3711d465d5c0a2aab225c6dcd3ca067e3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fabfc3d6dc3c77f1430bbf19242b253f1892e437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fabfc3d6dc3c77f1430bbf19242b253f1892e437"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
