)]}'
{"id":"openstack%2Fkayobe~793703","triplet_id":"openstack%2Fkayobe~master~I916de4232155f36052ee8ffbd1947a750ffe9778","project":"openstack/kayobe","branch":"master","topic":"tls-sync-kolla-ansible","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-01-12 13:10:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I916de4232155f36052ee8ffbd1947a750ffe9778","subject":"Set defaults for openstack_cacert","status":"ABANDONED","created":"2021-05-30 23:39:32.000000000","updated":"2022-01-12 13:10:45.000000000","total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"48c5c9b05b1dac66408885177dd7bacff0d78602","_number":793703,"virtual_id_number":793703,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-15 11:36:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-01-12 13:10:45.000000000","_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2022-01-12 13:10:45.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2022-01-12 13:10:45.000000000","_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-12 13:10:45.000000000","_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},{"value":1,"date":"2022-01-12 13:10:45.000000000","_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2021-12-09 13:51:39.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 01:30:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-31 08:11:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2021-05-31 08:11:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-05-31 08:11:44.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2021-06-03 10:14:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-08-17 12:17:10.000000000","updated_by":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"reviewer":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"state":"REVIEWER"},{"updated":"2021-10-01 05:36:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e3dc15c3db6b902c6a0ad8c4816a9405be6ef399","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-30 23:39:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e42f6673c9ad6c270273a5f1433c72f73f3c2c26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 01:30:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/230e2096385b455cb678be6a3d90013f : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a68a88ee726c48f2b8ee0102576bd835 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7619baa2ad2d4806b96ec4c920e2504d : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e24303a020d74ed7b837c2faa76b02cd : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a38d65f01d046f3ac785603a29a1b88 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd0bb20a85d945c9a50359b1257f8a89 : SUCCESS in 8m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5bbc9465c8a344fb975ddd06200e4a55 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/66feafe00faa4df0adb86e8016004c9f : SUCCESS in 10m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6dd8b820af434039b042ba052185a577 : SUCCESS in 18m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b49eaf7059da47578ee378b41767d794 : SUCCESS in 58m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3996ddacfcdf465aaa46957d941501f4 : SUCCESS in 44m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4ad7bcc525bb4ccc9da930058c1799a1 : SUCCESS in 59m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4f0369e0904744179e450f46e7fa6f70 : FAILURE in 15m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d777707c1ce42e9a0fd914aa75c90df : FAILURE in 11m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c5b5640293a4d0a95ecd84bd09af3d8 : SUCCESS in 1h 46m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/351d2ce4bc2d4bfe8abfaa4b8d935dc1 : SUCCESS in 28m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cdceb7c1cf844c4b12becf0b7332de1 : SUCCESS in 23m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bc5a7331475d4e07a2a9ea34c0b19f06 : SUCCESS in 25m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/221c8fa00e2045fcad323eab621e0062 : SUCCESS in 12m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f0af6c5ac454f0cb1ef8b6eb2b63680 : SUCCESS in 9m 55s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"10c9edf2dd27435d364254c95f6275eb9256eed3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-31 05:50:02.000000000","message":"Patch Set 1:\n\nrecheck ntp fail","accounts_in_message":[],"_revision_number":1},{"id":"832b64295857cc39fb08522c49263a688f37171c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 07:57:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eef80cb216e64ee8a83f6e311823dfb9 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfa0eeac9bcd4a47bc75fa73726029a6 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f31fc8286c64facb552b3bf1025655d : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2703ab645474db1a1442bc5bc1fc6f4 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/defb3745113142c1b5cf8006522c2ee1 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c665e377e0f84858b493f36d32742ad6 : SUCCESS in 9m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4eb562838b44802874571dd66f38f13 : SUCCESS in 6m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ebb5f8ac0b2a435f81c396763df16443 : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d74aedeb51c0403f831183bd99dbf040 : SUCCESS in 20m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/687362798a764551b2a8af6b62bfe4a0 : SUCCESS in 1h 13m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5329977bc77488a91d37c4a09b678d9 : SUCCESS in 47m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9ece872987704a47a88e850276688a54 : SUCCESS in 1h 08m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cc200c3e9bf04bd7a9b9df2ef2329cb2 : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30bd1410fb624855909c6804f4e8f705 : SUCCESS in 13m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cbcaa7114b7745979cbe3f04cdf62801 : SUCCESS in 2h 02m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/66d6c28fafd94ba2ad2448df93b21398 : SUCCESS in 32m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39fde4f0b93d4096acf07d2466e8ced9 : SUCCESS in 29m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/33cdb6f13909404e8ac31c057f267b71 : SUCCESS in 25m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/84362de639e2405e870c6faf7718ffb2 : SUCCESS in 12m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3cc2c4ddd364ecd8848cad1688a714a : SUCCESS in 10m 11s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"de82656b66edc9d28a5bf433f24453058d2b25c7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-05-31 08:11:44.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"97f6f79a1034b63af6808fbaf860735714bae23f","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-05-31 08:30:33.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1e8af2cf4a2e3e36e5dfca79ab9db64ae81ba72f","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-05-31 09:10:26.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1cf7f4709214ef307fe73ffdc7f331e501f6090f","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-06-01 07:06:03.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c7616964fe3589c4449c71025ab4e92c36a861ac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-03 10:14:22.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"23f608fa848e4c75d533cf5137dae9b14222be17","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-03 11:10:17.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4e9d869705d44b4e24ad003444f8a4706769dacd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-03 17:46:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9dec09e27fdf87e0b9f244c650b4a414d94fed86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 19:35:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3bba9edf78cc48d4aefe47c581cc6daa : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a3c599a850471f9b103bc27686d25d : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ba14de6fc674fa3bf3b3d17bc7d5500 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4084e63b746144fcb07608d7a804c490 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83e1b1bf800f4f8a91e613557e74c559 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcedb036f84c4e93b6d6ac968fec123d : SUCCESS in 9m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c661ea57a4f94286bc5845fe9c1b9529 : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a2758b930dd449793d7ee856e8e8994 : SUCCESS in 13m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/14208c571ff54f4d845c67e3675345a0 : SUCCESS in 19m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/1b23675991fd492397783bd7fb771b6f : SUCCESS in 58m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5138bc89621d4ded94484cb8b1fe6170 : SUCCESS in 43m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c6c2b8daa5ac4fb99329592f0728c730 : SUCCESS in 1h 08m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bda7b5dfae524325ae49de8f5777f817 : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/557d3f3817a844c683a7ec4100f106fb : SUCCESS in 12m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e7ab6cd0a8d2419ca85e98c1cbeef534 : SUCCESS in 1h 46m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/62678eedf55e49daa7ce5f4620a730bb : SUCCESS in 28m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b62967c565a1485daff60731e0ea99d6 : SUCCESS in 27m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4479b3b23bb947a3bb6917f60c35bf0b : SUCCESS in 23m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c7c6050997314f97b2a0609eda79eea0 : SUCCESS in 12m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cc198d092c84d2eb9594f1d4e6a4751 : SUCCESS in 11m 07s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"23e3bd1cda20bd265a0b1a4297a530792e045d34","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-06-03 20:23:45.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b2b6b0b41337c5f7c08447167b4590ca0073c6d5","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-06-08 08:55:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a89fd2b8794418d52078fccc0bd00cf118c1ade0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 09:26:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b20e153fa59c5cdbe9d79f4a0c1ae59ffe58ec3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 09:27:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3bc1f2779e26df22324b378de85c1a2358f0b9ed","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-09 12:26:05.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bf8ddc1eefbc0b10b4cd881d250fe88c35a11387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-15 10:22:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"37ddedfdfec269d92ae45e6b5d641c5889e32f26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 12:28:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dbd100e4219e4f50acc0448ab6926ed8 : POST_FAILURE in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50dbade5f0934683a9a1b826e13f146c : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c06f790b3aa46b0bff19e58d1f38381 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c22a12dcdf543938d035a9df23227e6 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/747d3db8e3d74178ab271988088fdd5e : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba52055903dd4550b83fff350a974060 : SUCCESS in 8m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/364586d7f5fc44969a2ce7390a26a7e0 : SUCCESS in 6m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b1d1cdb2bc3843a4a0334b17f87eadb4 : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a25769c36f94190af808b64539b3026 : SUCCESS in 20m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c6ba63d89a83443ba5aeb312ca59b538 : SUCCESS in 1h 10m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/245b95a51d7b41e5bb50c1087bb9ce54 : SUCCESS in 42m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c631caac09bd44ddaced32867b4dd938 : SUCCESS in 1h 14m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5a0471a137224fbba4723df0bcc417d0 : SUCCESS in 20m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb7e2753a96d403793b3230abc763338 : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/694d8fbc2c404ef0a32169ba1489bbfd : SUCCESS in 2h 01m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a839fb8db02d45b9a1b4fbbf8692d94b : SUCCESS in 35m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6143bcd3e2aa42b3b7b738a8f8adf3b8 : SUCCESS in 25m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1d71c3cf315e438e9cbe0024ff367655 : SUCCESS in 21m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c252d74b558c4f2db4d893265566f640 : SUCCESS in 15m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/063b9a438e1c4e17b9b364234828fa49 : SUCCESS in 11m 52s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"6b53dc429057ac2b1dfb9ab1a7df16936a1122cb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-15 15:50:26.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ac459325c39ca447a502f9aa895c7201e73c75b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 17:35:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec099dd4b26e466f85f29c209f35c6ac : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f7ac344896f45a0be14a82259aaa103 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b130fb2a41f4488690a3b10aca0835af : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f1de7c59a5640a392e24e795b4681a5 : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e48caf734e2445d2ade21b5b87388fe0 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21056868f5364be4982d6fb413051f5f : SUCCESS in 9m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d9e28b0d5c0426ebc4b4ee0037dc897 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c02b6aa425274050b374a04441e31feb : SUCCESS in 12m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5447d1a00db640bbbcdcb29a22130fb6 : SUCCESS in 20m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e1598c8d856b4e60b5879d07757e0406 : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5707bae14aa74d32b0c890124d08231f : SUCCESS in 49m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8a8b1681ff9f4ea9bd8284b05cea6db1 : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4edb1da689aa4c1681ab69b42e32b2a9 : SUCCESS in 19m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d800142f8424bc8bc4fc30c8c4f1780 : SUCCESS in 15m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/658e2f7148474e01aaa7a1dc9a46a999 : SUCCESS in 1h 43m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/25422d12b5ee48feb35eedf90661853c : SUCCESS in 31m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d1be71b07ae414c8d90ea9f027d57e3 : SUCCESS in 31m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b9883810a879445eb822459da42b22bf : SUCCESS in 28m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0a16112be8ac4eb79a8a706b2383ecf2 : SUCCESS in 12m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eae1a95011c4136bec00b5f840a3888 : SUCCESS in 11m 22s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"223fcaaee8610602a0495a603a4fd50c44e9da5e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-16 08:55:11.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4a42c947e70eef40d3971efc022be269858ef78e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-16 09:28:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d89df2abe90a3a35b5a1653bc182caa844928e63","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-16 13:25:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a9df292f94a18d9975727c43cdb3871762bab7fc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-16 14:52:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4fd3d9a7af2b4ffd4a4afe82f59e9088a66b0070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-06-30 09:17:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0d91d971bd13afd3c123c4de3071bfc2fdb24ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 11:12:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1efc060511748dc85a450b3040cd847 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51dfef5ca753442cab58b31812dd88fe : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6e206dbceb445b387faad870fa3e4d2 : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a39000ef31e047b4b3a064580ce7148d : SUCCESS in 10m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3713db648e142baa98e8895025f2f4f : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2220d6f24b654e9ebe8076026e7b1dae : SUCCESS in 10m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5943d92c276d4c39a1dfa925a234722c : SUCCESS in 7m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/afc7fe57c58846d6bb422d7fc0559339 : SUCCESS in 15m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5b99b139b06464ba0075e6af02c70cc : SUCCESS in 21m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e08d5984e48e48e1b522113ab601f66b : SUCCESS in 1h 12m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7c8fa7472bf4596b9a044f1757bea54 : SUCCESS in 46m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/df3f16a65b1a4bf69bc333c023e6a330 : SUCCESS in 1h 10m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c1ece61e50e3499d9c4798861a7804ea : SUCCESS in 20m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b87dc140360741bb9e3774927b341fed : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c9ecc8e2b4f344d5bfccb98606e84a25 : SUCCESS in 1h 50m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3ec12cf5b89b4159b35e1e4e61d4aa90 : SUCCESS in 36m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec07fbc8cd26484fbdd340813e5e83ff : SUCCESS in 27m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bb0a88888f50408f805754fb47d60c29 : SUCCESS in 27m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f0b571eb0f11493bb72a23c1b7785fd7 : SUCCESS in 13m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09a0d6aed06b4cc1a37d1c79257b4bb1 : SUCCESS in 13m 15s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":4},{"id":"b4afee29594cc7689ce1d54c5fe367f64981af2f","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-07-02 06:33:24.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"74a9b0c54b88c0337624d0db69d97f1edea751a1","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-07-02 07:33:44.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d9fb503906c420447d222f9ce846de4efa36c7ee","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-07-05 21:02:23.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"0c5af6cd7c99dfb9058bc4ac7b7565e9424825d3","author":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"date":"2021-08-17 12:17:10.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"276db023c52d28ccacd96d8b1e6a1beb52954de5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-20 08:55:58.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"49bb16384e916c346e4e339170a87aedfbdb53f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 10:11:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a443ab0e7276478a94cc9104e12db397 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5fe2f75658431db1bce644b860847e : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b46aa151390e4406b2ef08843b6da047 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61ce1548a1924fc3b909a4482dc2f377 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/993e55ec792a498d9e332042c79a0646 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e723d3193440f89a28b30b50000e1f : SUCCESS in 8m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e77406093f18474c97233340c024c722 : SUCCESS in 7m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf1bfedb73aa4f7c81ab827dab6be746 : SUCCESS in 11m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1fa78a2d5ce8426784d03491e2135883 : SUCCESS in 18m 35s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4a5bbd74eb724c938d50db36416b056a : SUCCESS in 1h 05m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/791a6b9999634e698ddcadb99e3650af : SUCCESS in 56m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0eaac11c318c42e38fb870e9830de878 : SUCCESS in 1h 14m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/36622674946745e9b2cfdd9bf77acd9c : SUCCESS in 17m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d72fc722e3084e8bb392f772f60f6b63 : SUCCESS in 15m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3dcb2c19cf254e5d9a044a9233471420 : FAILURE in 1h 14m 17s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a0eb2a736c904448a55d6dd3453387f3 : SUCCESS in 38m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a5b87976feb40eba7a3689b05bc73bb : SUCCESS in 34m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/929f0d80fed24c0bbe88eacd1e06f0cc : SUCCESS in 31m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d7e27f1f558f4c15aaebbd553e085d35 : SUCCESS in 12m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5eff5ba49d5b4afd83b2971ee3016782 : SUCCESS in 12m 17s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"0dae6690cd8b3dd0edbc3380401e2d19b1ca679e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-20 13:00:23.000000000","message":"Patch Set 5:\n\nrecheck Ansible Galaxy urlopen error","accounts_in_message":[],"_revision_number":5},{"id":"38018b11ad35a47c6ba17500ddf8c18d39b1d3de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-20 14:57:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b14535b8cda54042b105795775d9fe18 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9546ca928552453591f246fcec33ede1 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df256cc067c04be49a040046068e030a : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f4370546f6d4974a644af9f3825f9c4 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf1b8f249a544338ac0296359589501a : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b802758d51684c5096988207e650526a : SUCCESS in 9m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c8b9a0956c904cfb9edbda76333b1efa : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d8be59e1844840f2a409755fcf7a8ef0 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f850192117a14774a3bdcaea97c3cb45 : SUCCESS in 19m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ad6cf386484f42bdbe6e6a8a118ff91e : SUCCESS in 1h 07m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f915c34e1b14865b1ee27b26dbb0726 : SUCCESS in 48m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f76584130ae94f3582983f11ca4dcaba : SUCCESS in 1h 11m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f4f1260792de46558f9a05979234003b : SUCCESS in 20m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e14323b02b344184a0165ba1d5bfc921 : SUCCESS in 15m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/88848af1c1d74fed8b02f9a96c73fe11 : SUCCESS in 1h 51m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4652d955d7304236b0b4ea2535f3eace : SUCCESS in 39m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/def1c339f99b442abf713fff6325ecc1 : SUCCESS in 27m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e9064b590001455783c821be76ed8f5c : SUCCESS in 26m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/90274d7d78494d35a0aa7e2bfeb80ad5 : SUCCESS in 15m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a80add8bd6574ef69febe1428315c6a1 : SUCCESS in 12m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":5},{"id":"45bc001370d622a546cd199ed0d99ac2f547e16d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-01 05:36:34.000000000","message":"Patch Set 5: Review-Priority-1","accounts_in_message":[],"_revision_number":5},{"id":"102bcac749779f249d5f2d51761d97d466a6bc48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-15 09:33:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"47624245aa6ef974785fefa434eb4bdeb8c3de35","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-10-15 09:45:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"af8f2a6be488fbee2b951867ef299e5a6cfd8e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 11:36:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/282ae868f5aa4f26bfce8f134084c852 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c91034b3d3894c2e93a95964bcb47d4d : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e80bb7f3fc604077874ac2664fe0867a : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9965c159c39e4034b6e7c92e725e8420 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1859d6c1402942a58d6c4dd4ed9d71ce : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c51c12908da4d759464c8b37fd35c55 : SUCCESS in 9m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b5f5d562eb2c484d949fd7d7afa68b3a : SUCCESS in 7m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1785972dd2bf474e920d8fc47703b949 : SUCCESS in 12m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b08c6842697048c79e30b60461bec896 : SUCCESS in 32m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/23615326dcd84de1b816a75407a4d1c6 : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22cc462c34f34811b011ff081c5e0cc8 : SUCCESS in 51m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b157e89e923947f0ad8adf21cd6c3019 : SUCCESS in 1h 08m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7f223c6835f842e88b6c24136ed8ae32 : SUCCESS in 23m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b8694027dfc460f96462b46b85998f9 : SUCCESS in 17m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ed982cabb9a44fb3941e345c9624962a : SUCCESS in 2h 01m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5263beaddfa433c941cf1fc97c2698f : SUCCESS in 1h 29m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/68e27d1bd10c4c72886ce95744723250 : SUCCESS in 41m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87a6c3b7c5144876902dbafafc87ccfa : SUCCESS in 32m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8fa34ae142b64cffaffbe975e1ee009c : SUCCESS in 29m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b36d11ec99d347ef9505a629ce759740 : SUCCESS in 27m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7f350368549c49b1b444bfddfd9fe6a5 : SUCCESS in 21m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0799573e2746475a9e91ee6319ad6c81 : SUCCESS in 14m 15s","accounts_in_message":[],"_revision_number":6},{"id":"27a4ddb6fcfc9eebd0390901c5bdbd4e5c49d897","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-12-09 13:51:39.000000000","message":"Removed Review-Priority-1 by Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":6},{"id":"d6574f452b77fb93e0e09f1d76c30ad3bf440357","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2022-01-12 13:10:45.000000000","message":"Abandoned\n\nMerged with https://review.opendev.org/c/openstack/kayobe/+/793697","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d8360ea357ae62a135b7c94cc19ead72cfb5b521","revisions":{"5551c6dfaa55a0598a242a98c788b5e3a0767802":{"kind":"REWORK","_number":1,"created":"2021-05-30 23:39:32.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/03/793703/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/793703/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/793703/1"}}},"commit":{"parents":[{"commit":"1e14fa3a24fa64b84cc993c985e696a9cba78dea","subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e14fa3a24fa64b84cc993c985e696a9cba78dea"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 22:53:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 23:27:06.000000000","tz":180},"subject":"Set defaults for openstack_cacert","message":"Set defaults for openstack_cacert\n\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nChange-Id: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5551c6dfaa55a0598a242a98c788b5e3a0767802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5551c6dfaa55a0598a242a98c788b5e3a0767802"}]},"branch":"refs/heads/master"},"430a567aa40d5cd33ab88e917612c7877d3ed3c7":{"kind":"REWORK","_number":2,"created":"2021-06-03 17:46:37.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/03/793703/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/793703/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/793703/2"}}},"commit":{"parents":[{"commit":"1e14fa3a24fa64b84cc993c985e696a9cba78dea","subject":"CI: Fix libvirt errors caused by latest edk2-ovmf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e14fa3a24fa64b84cc993c985e696a9cba78dea"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 22:53:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-03 17:46:08.000000000","tz":180},"subject":"Set defaults for openstack_cacert","message":"Set defaults for openstack_cacert\n\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nChange-Id: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/430a567aa40d5cd33ab88e917612c7877d3ed3c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/430a567aa40d5cd33ab88e917612c7877d3ed3c7"}]},"branch":"refs/heads/master"},"9ec5776024393e41c5ac1d8146bb44ad196c4a05":{"kind":"REWORK","_number":3,"created":"2021-06-15 10:22:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/03/793703/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/793703/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/793703/3"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 22:53:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-15 10:21:16.000000000","tz":180},"subject":"Set defaults for openstack_cacert","message":"Set defaults for openstack_cacert\n\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nChange-Id: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ec5776024393e41c5ac1d8146bb44ad196c4a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ec5776024393e41c5ac1d8146bb44ad196c4a05"}]},"branch":"refs/heads/master"},"22b0feb2594929fc4a1604c0ebc3c921d7e41c6b":{"kind":"REWORK","_number":4,"created":"2021-06-30 09:17:37.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/03/793703/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/793703/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/793703/4"}}},"commit":{"parents":[{"commit":"e9a3733e24c5280dda241aec839097d005014c99","subject":"Merge \"Prevent matching unrelated images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9a3733e24c5280dda241aec839097d005014c99"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 22:53:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-30 09:16:26.000000000","tz":180},"subject":"Set defaults for openstack_cacert","message":"Set defaults for openstack_cacert\n\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nChange-Id: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22b0feb2594929fc4a1604c0ebc3c921d7e41c6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22b0feb2594929fc4a1604c0ebc3c921d7e41c6b"}]},"branch":"refs/heads/master"},"0b05eb95b8f24050514ab44d3ee4d4353f2810ea":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-08-20 08:55:58.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/03/793703/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/793703/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/793703/5"}}},"commit":{"parents":[{"commit":"a48cc2498859f5949d304266d32ce3e620ad7f36","subject":"dnf: add metalink argument to custom dnf repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a48cc2498859f5949d304266d32ce3e620ad7f36"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 22:53:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-08-20 08:54:38.000000000","tz":180},"subject":"Set defaults for openstack_cacert","message":"Set defaults for openstack_cacert\n\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nChange-Id: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b05eb95b8f24050514ab44d3ee4d4353f2810ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b05eb95b8f24050514ab44d3ee4d4353f2810ea"}]},"branch":"refs/heads/master"},"d8360ea357ae62a135b7c94cc19ead72cfb5b521":{"kind":"REWORK","_number":6,"created":"2021-10-15 09:33:16.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/03/793703/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/793703/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/793703/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/793703/6"}}},"commit":{"parents":[{"commit":"e97a8fe63d2cd57579ba4288234e6edad495e37a","subject":"Merge \"Use merge_configs and merge_yaml to generate Kolla configs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e97a8fe63d2cd57579ba4288234e6edad495e37a"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-05-30 22:53:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-10-15 09:32:26.000000000","tz":180},"subject":"Set defaults for openstack_cacert","message":"Set defaults for openstack_cacert\n\nSince I0fd596d93a0e575a391250d4bed261ad370a3664 we always have the\n\u0027ca-certificates\u0027 package installed in the all containers so its fine\nto set defaults but leave an ability to override the \u0027openstack_cacert\u0027\nvia environment variable too. All supported systems already install\nthe \u0027ca-certificates\u0027 package by default but for sure we install it\non the Ansible control host in the kolla-ansible role itself.\n\nChange-Id: I916de4232155f36052ee8ffbd1947a750ffe9778\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8360ea357ae62a135b7c94cc19ead72cfb5b521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8360ea357ae62a135b7c94cc19ead72cfb5b521"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
