)]}'
{"id":"openstack%2Fkayobe~794451","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Ic0ff71bb11692eb6c5699a1673df2d16b8f894ec","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"Ic0ff71bb11692eb6c5699a1673df2d16b8f894ec","subject":"ntp: check for ntp group in inventory","status":"MERGED","created":"2021-06-03 14:05:02.000000000","updated":"2021-06-04 11:13:49.000000000","submitted":"2021-06-04 11:12:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794451","cherry_pick_of_change":793414,"cherry_pick_of_patch_set":1,"meta_rev_id":"9c3c40ae9c911f311a9e8d0d4dc9a5937890e98b","_number":794451,"virtual_id_number":794451,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-04 11:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-04 09:34:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-03 14:16:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-06-04 09:34:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 14:16:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-03 16:11:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d69199b17eefaf004040def146c2ac2eae64e595","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-03 14:05:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2a3271e807e8366c3063a3161a2911030538471e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-03 14:16:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"70e7604ccb2e7344f2d5ae021ef39d7556f337e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 16:11:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24712dab64ee4bc79fd0f4c78d8d3500 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f33a2ded2d74ff3b2228a4a6d129ad7 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b009e02103e4cf49ab9ea12c1d73ff1 : SUCCESS in 7m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bf95871cf1549ed9e12aef88d5742ad : SUCCESS in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea1901b5d2f44536a414bca1c96c03e1 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41a1c75ab10a4d9dbd90e9b41794740c : SUCCESS in 9m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e5e3400991074a778e84294b47be8450 : SUCCESS in 8m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a02e989e86764f019a1fa2d30aeebb6d : SUCCESS in 12m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3dfd224045e54ec2a2b2cef0c5db3eab : SUCCESS in 21m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e4bcf9d47ab54976a8bec89f5cfef03d : SUCCESS in 1h 05m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8eac2479d5ac44ba82db1c2e43e83542 : FAILURE in 44m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/afdde71b89fd4154be2ec8fd169d5091 : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0fd4bedc8d904009904dd3d8d4225155 : SUCCESS in 22m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b8dcde10e32454b851ad7c46d5014d5 : SUCCESS in 15m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6bc5d9f0b2de4f23bf2a911714009423 : SUCCESS in 1h 59m 52s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c71df1ab04e848fb9de55e143d2759ff : SUCCESS in 32m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/172577e2294146aab6c44ff41d8967b8 : SUCCESS in 30m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/facc713d225f41c9bc9fbec218d0bed2 : SUCCESS in 30m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b92fb38bca94448fa1d56af55035f605 : SUCCESS in 14m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6631f5cf8dbb49448635623bd348a21a : SUCCESS in 10m 14s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"8a48ea74792b93bbed18b76450e75bd789c8543d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-03 19:50:44.000000000","message":"Patch Set 1:\n\nrecheck\n\nubuntu mirror issue","accounts_in_message":[],"_revision_number":1},{"id":"88e8d72ecc6a103c04ddf2a5c48b7678803043c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 21:40:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb6d8d946e7f4e5bbab5a71882526981 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2030885c0e3e4f78b7b876f707a95754 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75407bc03182437ca34ae7f52e919a72 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53dd1138025e40c5bc65a5cdbc36b792 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43438df83e164a88bd9fc10aedbe14cb : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cf41ddd5ea84ae89578c9516b7539d7 : SUCCESS in 9m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c34ff0546f494e4f9b5af53e918412dd : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf5d1e2e91744938939e63709af20fd1 : SUCCESS in 13m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/85c678c969e241e785ed6b67ba55c4dc : SUCCESS in 20m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0f80c6be8505408b9763e5d625ea7827 : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f1f91e4add7406d9891db9713e1677c : SUCCESS in 47m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dd880472172145958564c203737f9e53 : SUCCESS in 1h 09m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4d41d52d1028498c96033309a17895a0 : SUCCESS in 19m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/640ab42d834f410bac82c0fac62063f5 : SUCCESS in 14m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d5f23f4b7b9d4386aeb24bf97c842422 : SUCCESS in 1h 45m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6359b569ec8844219e72dd4689797d09 : SUCCESS in 31m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf3534068f6f4da98885097bb965a44a : SUCCESS in 27m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/17d95d07c17f402796af359bd59ce735 : SUCCESS in 27m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/65166024f3154cb890d5d25bcd2084b2 : SUCCESS in 13m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2a866b785034c6e92cf2bd15674412a : SUCCESS in 11m 36s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"ce554c6dbd850641522f49af8af49b78c04dabae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-04 09:34:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3bd1000cadd7ac1ac1e6f2306982ff6e0c49ff4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 09:35:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6bbc010d0b301dc19df4e432373ed99f13151690","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 11:12:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/891a6a562dc34901811bd2a90b0b36b8 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8febb34f8ee4fefaa4f3485f983990b : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad4147ed385e4e1fa56cefad28f1e70d : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea04a1822dc44e8a5c10dcfd6680dba : SUCCESS in 9m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a5194314a7424d39847869414da74bb9 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9bd01888554a4cf28d5df216b43ecf85 : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1207f18168394577b4410b51560581cb : SUCCESS in 20m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/782143bbec4f4f4698c7bba2b53e9992 : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a92071a9e6c486089f30eb1bd9c3fca : SUCCESS in 44m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c665cf232fe848e5a5a1bc920928baf5 : SUCCESS in 1h 20m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/dc750b2ace354946828aad2cf0a2ad3e : SUCCESS in 17m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f12ab63712af43eca6e66c114d8e7d58 : SUCCESS in 12m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/320283c14c9f45bbb363825bdfce7698 : SUCCESS in 1h 36m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/712286e25aa542fe9721b8a243fc24a4 : SUCCESS in 27m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/854102751656416a9a6fb01af1734773 : SUCCESS in 25m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a78cb6acb4a841178af09e22b5ca3bd2 : SUCCESS in 25m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9f9fc5084bfb472b8d10943849d8b2b2 : SUCCESS in 10m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5d713f3b1fe4f2fb7a3590b2bd2291c : SUCCESS in 10m 28s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"603c0aaf8d0c5cfc5d32c4eff1205efdb2c64a4d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 11:12:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9c3c40ae9c911f311a9e8d0d4dc9a5937890e98b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 11:13:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7833ffc7004d8a938c00e0c08596a2 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4648b90bcf31cd6c293653c7500b5f4ae0a2eddc","revisions":{"4648b90bcf31cd6c293653c7500b5f4ae0a2eddc":{"kind":"REWORK","_number":1,"created":"2021-06-03 14:05:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/51/794451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/51/794451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/794451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/794451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/51/794451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/51/794451/1"}}},"commit":{"parents":[{"commit":"1d43eda862816b7e71ae2b61151d8d4c4dce396d","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d43eda862816b7e71ae2b61151d8d4c4dce396d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-27 14:16:14.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:05:02.000000000","tz":0},"subject":"ntp: check for ntp group in inventory","message":"ntp: check for ntp group in inventory\n\nThe Wallaby release adds support for configuring a chrony daemon on\nhosts in the ntp group. The kolla chrony container is disabled by\ndefault, meaning that users could be left without an NTP client if they\nforget to add hosts to the ntp group.\n\nThis change adds a check for the existence of an ntp group in the\ninventory. The group may be empty.\n\nChange-Id: Ic0ff71bb11692eb6c5699a1673df2d16b8f894ec\nStory: 2007872\nTask: 42533\n(cherry picked from commit 9d3045dbf0b6a45cd9c2dcb0a5fae8c8d81045f7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4648b90bcf31cd6c293653c7500b5f4ae0a2eddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4648b90bcf31cd6c293653c7500b5f4ae0a2eddc"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
