)]}'
{"id":"openstack%2Fkayobe~794599","triplet_id":"openstack%2Fkayobe~master~I608c0362e6eeea65ffafb32cda5c19ccb6ec8b5c","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I608c0362e6eeea65ffafb32cda5c19ccb6ec8b5c","subject":"Fix container image builds","status":"MERGED","created":"2021-06-03 14:25:32.000000000","updated":"2021-06-08 00:00:22.000000000","submitted":"2021-06-07 23:59:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794599","meta_rev_id":"4bc53c0d8f023a9f0e2a0960f7746e9272109570","_number":794599,"virtual_id_number":794599,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-07 23:59:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-08 00:00:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-08 00:00:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-06-07 15:19:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-08 00:00:22.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 14:32:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-03 16:26:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 09:09:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bb0fae01c8567ddcc6f67118388747de4e3b74dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-03 14:25:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28465727a42accd222ff299c00d59950315a5d01","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-03 14:32:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1caad2f4069ede9b42bb0b6e8f3b6d71c8b446a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 16:26:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b249b62e05a6414aa9341eb89de975e8 : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0f0cf4c0e55480284b7e435225cc983 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11ad6a2960ce4a3396b60cd8fb8db893 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0218bbae66654b208bebc57b4c91bba1 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c139d93fc6a4bcea931bc355efa2d08 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bc280111efa4365a3ddab3c3b037b56 : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ac3d1df4baa4c94927ad05bc3db86ca : SUCCESS in 5m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/400d7b4fe88d492d97ea58a6335cb8b8 : SUCCESS in 8m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c13bc3625a449c588e8f9392db2a8a8 : SUCCESS in 13m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b53aacdb7230483b8461e26c17f82dbc : SUCCESS in 21m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/344e5040dc2843a0bcd774169bf501c9 : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8efe49fbce324d72b1086283cba8ffb4 : SUCCESS in 52m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3b48cc41e59a4831a0a358db7db3a2b0 : SUCCESS in 1h 11m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e6c5bd14ade2429fb35b7c29950de8c7 : SUCCESS in 17m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/215fd9c3a26549d293d989436f7f994d : SUCCESS in 13m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0da2ed29d13a45f29e218af3ef1bcd86 : SUCCESS in 1h 55m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d2ee420eda5541b8aed7e6b92ddfe809 : SUCCESS in 33m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c0de2dc81444960856907cbda96f30b : SUCCESS in 29m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9b7724074e864a53a71b38b3b325078e : SUCCESS in 29m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c49af2ba02aa42259e2ed970563ee215 : SUCCESS in 14m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3382604a2c14c56bb33968898a27db7 : SUCCESS in 11m 04s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"405cce61234ee5fb100f7ab8e3637b896b21041a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-04 12:51:37.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"08e1444e320a79f4836a1c105eca852dcbd37cc1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-04 12:51:46.000000000","message":"Patch Set 1:\n\nBackport to stable/wallaby","accounts_in_message":[],"_revision_number":1},{"id":"64d020b1ecf42e92f4d436f97679e3be30c6d596","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-07 13:41:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f86e436fba378c99b830726dba884e206527d204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 13:41:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"806383e1efff8787c571e5e659ebf356f4a0a7be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 15:12:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e78a6cd42504bc3b0a6b987d587ec7a : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d229007a99d4ac18a0d9fe5f114ab06 : SUCCESS in 7m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d627bbd1d6f41f09282d3bdf69464a7 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8985862041a648e68625d26b393a83dc : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b9585ec4ff245378eb7ccc2f8d6a9e3 : SUCCESS in 4m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/edc165b9c3d34ec6982b578a85a10dc7 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c799041efe8c43489a571830a0d257a4 : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b3f280bc51d4e61a0e4bfaad7a2d6b8 : SUCCESS in 19m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/be916e146f4d4bcb8e5215c2d777d975 : SUCCESS in 1h 06m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a6e3ba63b984a3c8a4409b04e83c74d : SUCCESS in 46m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d65276b847d6413a92388e17f2a3e0fb : SUCCESS in 1h 15m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/207a04655cee4b5fa22231c73edbfcaa : SUCCESS in 17m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb21f68911104261a9e3ff242fecd3db : SUCCESS in 15m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/560bb35c8c654cc5ba1fdcfbbce236e7 : FAILURE in 1h 28m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/18484c2f40db457c81173da48a7a84ca : POST_FAILURE in 58m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7bdb358f5b745e7b2b0ef4a420ba673 : SUCCESS in 23m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/efdf2bbff90a4594850f6861a302325f : SUCCESS in 27m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8a5afcb766984c6a81955b03ff75f88d : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21e8f219ebfd4088882d6d042611e7c7 : SUCCESS in 12m 39s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c5ec2f92a153cbebd10be9de89e0b1587ad0dee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 15:18:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4e389f781bf116896b3231f1137d8a19de52d9ef","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 15:19:36.000000000","message":"Patch Set 2: Workflow+1\n\nBroken by kolla-ansible chrony precheck. Rebased onto https://review.opendev.org/c/openstack/kayobe/+/792243 which should fix it.","accounts_in_message":[],"_revision_number":2},{"id":"382ad5684a336e4273eed6c8a337fffcebfb57e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 17:13:00.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5e4c465ced14aa5b3398e1cc4f13a15 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9109e84ff39d4874925969685de4244e : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/248ab786e4644f44bc3b9e2ec79fa1dd : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd90c5811d53447baa9487d0cdb86da2 : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2db776f460e44d4aaf5f469c8ecacf87 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a130ebe8ff45c09938545e35153780 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fe8a85dbbf445758776e68a1504a3ab : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea5146e0cc0346d5b8d410a98300fb08 : SUCCESS in 8m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7f4e0929a2754208b7cd3007f3cea513 : SUCCESS in 12m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b69c1efce7d4fcb8eecb96073e6aa4a : SUCCESS in 20m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/86e62123bf1c48b382ad4896c4ed9f6d : SUCCESS in 1h 14m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b2a3ea3b1984f099d3a1ad71886e685 : SUCCESS in 45m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5851bc5893dd4933a595337655e3a453 : SUCCESS in 1h 10m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3f2a5a871b544d74a64446feb8b2c4ee : SUCCESS in 16m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef0d889e9edc4fc18cf3798945b348bf : SUCCESS in 13m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d124d33195ab4318b3f0c5afd09b1d53 : FAILURE in 1h 20m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bec42b7397404771afe1bcf809dd4142 : SUCCESS in 27m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c1fe813cba44731a5e24db6a1dd50dc : SUCCESS in 22m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9f7c76853b154671b8739d5167ad9592 : SUCCESS in 30m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/890ea56a666d403dab703475aae1aa7d : SUCCESS in 12m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df690230eb4f4d1c8023492bc88e70cf : SUCCESS in 12m 10s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"d57f1404833a0a8cc9e3ef169302c6148eeb9a04","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-07 20:26:00.000000000","message":"Patch Set 2:\n\nrecheck\n\nerror pulling image configuration: received unexpected HTTP status: 500 Proxy Error","accounts_in_message":[],"_revision_number":2},{"id":"14343413a80b421c23e5ec789cc6b308b9a7ef8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 22:07:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/384c47fb63e94fe09532ace22ba8e50f : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71875f106b13438e81c771a6e0e26dde : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0897822857f348c69344482f31056bd3 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b09aa86de3047dea0d552939b4360f9 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d6556e194964c8f931e342c3a3b0e34 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7fc853c611548fca22f8948879af3f6 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5de5104fb77c4864830045f12fd121f1 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eddb9ff33f08424488ad64c9b77e0754 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a137065a9f2e4e17a32726254ec07462 : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b9803f6ec0a0446796d94998845a6da7 : SUCCESS in 20m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3a3d981b93c74bf2bfd9046b0afef71c : SUCCESS in 1h 04m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6fab65566374a1b95ab24e990484078 : SUCCESS in 47m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3a1ac8953e384fcebd27fb8aa46c3754 : SUCCESS in 1h 06m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8a02adda3ab644a7b1af7724b30e0a4a : SUCCESS in 36m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/902f2f357d04435ba48e2cf59f8b23a1 : SUCCESS in 12m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/71264f0e0dca43c08db2dce7a2ac41c5 : SUCCESS in 1h 39m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/02e2479c36a54e64b0d39baddeddad55 : SUCCESS in 25m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e01f297d5fd4f078a617c55190043ff : SUCCESS in 23m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/20533584a9144a3486d06d04deb83ded : SUCCESS in 25m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/70f32c770fee4a0c9072967e21e6c3ff : SUCCESS in 12m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9b5bcf6a50842bf91f45fe4c27289e6 : SUCCESS in 9m 39s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"1ff04beb239681a8fdad22ba85a979692e4fe1fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 22:07:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a5516e198fa9fed459c3b381c796be76ec3da473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 23:59:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/948a2f1e7b334008aea7e8cc4be91096 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5c666e3dd3f4c0788439f8e24b796ff : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa384d12c1e34e4185ab3a16622eec9e : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7147c77a0abf45808fc1756ef059e43f : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba2aa58529b5484aae6178b8bac5841a : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cdfa396f756340d992fd120900240372 : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09e0cf999c964c438007e0feeb1a7d0b : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9163ecaff023411b99d2f3bf68bc8151 : SUCCESS in 21m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/15b49f58d1ca4244ba513f82e524e277 : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f36ccaee7ad4bdfb00ae19aa736fd26 : SUCCESS in 46m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/342afe0f98a149faaa43753f36d70125 : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ae75748adb584db58a41f72f210ca5e8 : SUCCESS in 18m 00s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9617b6ffa16a4888a50fb84d7613f290 : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/61c9ac8abb274f93ad9894ef1a5e13fe : SUCCESS in 1h 50m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/35f4ca7079924d75bf544d7343b3087d : SUCCESS in 24m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34c06a5b6d334e83bbfa4107da531a59 : SUCCESS in 21m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/13e7fc0dced845e1a6b5520a1306785c : SUCCESS in 25m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/eadf00232eae4165a75f4065684e0c6e : SUCCESS in 12m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8ab6c326d4e453d9a0a5ddf65430db4 : SUCCESS in 10m 34s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"6ad494d72436529f036625a234dfb4c5113dfc7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 23:59:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d0abe5724dad588aefaa1e05381a6e5bcdf92495","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 00:00:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fc53d3e25d643f7b1b626b809e35926 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1a7ae5fa0684b498e59c7e5853d3ca9 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9c8f74993d63d73dfead6d1edb2b2827ae0e71bf","revisions":{"47eaed441fa72482dd966cc6045fb0b149debcb5":{"kind":"REWORK","_number":1,"created":"2021-06-03 14:25:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/794599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/794599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/794599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/794599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/794599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/794599/1"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:19:29.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:19:29.000000000","tz":60},"subject":"Fix container image builds","message":"Fix container image builds\n\nContainer image builds broken due to disabling docker iptables\nmanagement in\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/751795. This\naffects Wallaby and later.\n\nThis patch configures kolla-build to use network_mode\u003dhost.\n\nChange-Id: I608c0362e6eeea65ffafb32cda5c19ccb6ec8b5c\nStory: 2008942\nTask: 42560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47eaed441fa72482dd966cc6045fb0b149debcb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47eaed441fa72482dd966cc6045fb0b149debcb5"}]},"branch":"refs/heads/master"},"9c8f74993d63d73dfead6d1edb2b2827ae0e71bf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-06-07 15:18:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/794599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/794599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/794599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/794599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/794599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/794599/2"}}},"commit":{"parents":[{"commit":"dc53d29eaa0e5f67335aef3db23fc11a17dc21b0","subject":"chrony: cleanup during overcloud host upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc53d29eaa0e5f67335aef3db23fc11a17dc21b0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:19:29.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 15:18:53.000000000","tz":0},"subject":"Fix container image builds","message":"Fix container image builds\n\nContainer image builds broken due to disabling docker iptables\nmanagement in\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/751795. This\naffects Wallaby and later.\n\nThis patch configures kolla-build to use network_mode\u003dhost.\n\nChange-Id: I608c0362e6eeea65ffafb32cda5c19ccb6ec8b5c\nStory: 2008942\nTask: 42560\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c8f74993d63d73dfead6d1edb2b2827ae0e71bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c8f74993d63d73dfead6d1edb2b2827ae0e71bf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
