)]}'
{"id":"openstack%2Fkayobe~794601","triplet_id":"openstack%2Fkayobe~master~Id7378dd844aca7cb5634535308eba0f60342c81d","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Id7378dd844aca7cb5634535308eba0f60342c81d","subject":"CI: test base container image build in seed jobs","status":"MERGED","created":"2021-06-03 14:35:23.000000000","updated":"2021-06-09 14:09:00.000000000","submitted":"2021-06-09 14:06:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794601","meta_rev_id":"3ee69447f8d7dcd8f6193593c0de85ce39b4cac8","_number":794601,"virtual_id_number":794601,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 14:06:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-09 09:29:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-09 12:07:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-09 12:07:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 15:12:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 09:29:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-06-09 12:07:04.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"16a3b4c62319bdfc1aa4baf86210df2fc7b82e3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-03 14:35:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d950a1398eb72cf8e7aa755683b5ff82662bb0e","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-03 14:37:46.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"9762ea02a187923db5d134d034e6a084c40364f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 15:12:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c11593d1f8944348aaed9dcf6e3664df : FAILURE in 29m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f1700b0866d41da828793f3a14089ea : SUCCESS in 31m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0717ee90c17984fbbdf88fdc74b9c32b07395a2c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 08:55:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c2c8e92d289da7971b9ae1463a5dcc8f4a094374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 09:33:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cfa2ee061e4a43ff8edf2a323ee026ee : FAILURE in 27m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ccf99a1ef9a0497a9af0fa018f55f218 : SUCCESS in 31m 06s","accounts_in_message":[],"_revision_number":2},{"id":"64822143731740c40db5a387390e979e7ea435f1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 09:41:59.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5fa7ac3f49669bc784c3478c4c370d13809d4475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 10:33:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9dd875ead041456184b389ea86f4ae7a : FAILURE in 30m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8578901551f9426ab083ef2b14c63b61 : SUCCESS in 23m 40s","accounts_in_message":[],"_revision_number":3},{"id":"fc5366ffc197eef6279197fcf82cae1cf35abb00","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 13:22:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9ef1904524f46be75bd740819fe31ac22dfa51dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 13:58:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/826264156d494613bda77974af0802c0 : FAILURE in 29m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c77089f00e794584bad471b10b743145 : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":4},{"id":"f81b3e8ee32f93018051f2e6086f2f4538c0bafd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 15:27:28.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"326524dadd0faa5671ff91384ce5ba8c1e3aa209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 16:15:04.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/826cce29264449feb567253d870571d5 : FAILURE in 25m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e5c582d4c974a4b84960b03a6a4bf6d : FAILURE in 21m 34s","accounts_in_message":[],"_revision_number":5},{"id":"a65e39c607c1d75d90925c3fb7bf84a2cd6eea29","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 16:41:34.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"0006783f20184a9875ec10b4b2781c1593768bfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 17:24:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/30687689119845d395acfb30c9fc40ff : FAILURE in 34m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a0a198948c9474cac78459f684399fc : FAILURE in 22m 48s","accounts_in_message":[],"_revision_number":6},{"id":"9457756e0fb74bbcac5e1f25475089e69bd36297","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 08:01:31.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"22cecd92977e13f400097d74e83913b804a610cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 08:16:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f75131da62dd40a49a673d664556e17f : FAILURE in 6m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99bcd1837d4e4fa9ab7c27d80e90f63d : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":7},{"id":"772d1c53e497f978f71aceec9844d5efefbc34b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 08:25:58.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"1d16b680cf720b1141026e15d1885e23e3ea3a95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 08:58:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/037d38107c6d496da269bc723349f130 : SUCCESS in 31m 11s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91495e9ac55544ebba3186853d8a78ce : SUCCESS in 27m 01s","accounts_in_message":[],"_revision_number":8},{"id":"1d014a28bcfd6dfa414755f67f3f4aa6484941eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 12:47:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c3e878b6a4c43430320e4436c23ecec8db36c99a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:40:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f22618bc4064e66a63ee621af241b49 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52a0900f8be4b0bb855464307914c3d : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2e174c302634bf3b63148489761d73a : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2829d6fbd684949bac64f974d8e68e1 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67c9eb982c2c472ab0cf59d5d4aeabd0 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae3bb3ab58b46c383edbc65cd690468 : SUCCESS in 10m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60559c7a41874a99b9387e6b248ca42e : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bec9b11294cf42ffb52620820c6a0cd2 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b17a46468c640b183b721504593beeb : SUCCESS in 19m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4d17ed1637de45f39aa42ac7d3bbb1ab : SUCCESS in 1h 09m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bd97cc574a742ac8190ac6f873fc464 : SUCCESS in 47m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9d144ceba1164f6cb2d40f0268e8b278 : SUCCESS in 1h 26m 34s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9c98817986a947a38ff49e325208363c : SUCCESS in 19m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/883bbd41d2284c78a64366b6dd8bebee : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/68e2f3eb83454747bbe7c74100f698f1 : SUCCESS in 1h 48m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/cae7e34fca344ff1913241e108703067 : SUCCESS in 30m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4e1ba4d9d264b999c8729e73025317b : SUCCESS in 30m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/be7130c908f74191a0282fd94179087a : SUCCESS in 24m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/694cc3cb378140b3870a2a836407f7cc : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af117616f44d4080995233123cfcbab7 : SUCCESS in 11m 01s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":9},{"id":"3800f173e548928b283f7de1c7a131b40a0556f4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 15:00:33.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":9},{"id":"43dea58aa53b64a4ce126345e5fdd726f4b41a89","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-09 09:29:59.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"6f4fc417f3729e99fdf5ed35a8c9e1f811445f98","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 12:07:04.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"e47f2c5e38ba40bf6ac48165bd24036bd54b3d50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 12:07:19.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"22642f62fc90a0d61923eaddb0cfdd9e59829c54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:06:04.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27546617d724cb7b86b792643908883 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2188bc8e6c2544eaa3a84d964632b40d : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72a459bc97584c95ba02f1cacc0f2411 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed19c995a43348f6bb4051e653e94173 : SUCCESS in 9m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/12acfb4e02ab450da17f9ff4eeba33d7 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f7ca877ee2c84d67a6910439b716b191 : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3bb459595d634855a774514fe8902450 : SUCCESS in 20m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0f545b2bac9747b1bfae83619e49e7cc : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bd1312ac5c748d4ab739303da90f26d : SUCCESS in 43m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5a40307095e449efaa9eab821e03d9f8 : SUCCESS in 1h 00m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/09a0a4966ba943718b39e90e37ef1422 : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0fc6af24dc348f898eff52726a7ac30 : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bb3e4b4e26984280baed4fed576b66ba : SUCCESS in 1h 54m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/51e082d6ddf84199bb07116b16a1dc3c : SUCCESS in 31m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6a596e3ee9c49fe96d05d7410dad8e7 : SUCCESS in 27m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6a2e9a68da1a423c9337629132d6bd83 : SUCCESS in 30m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/587aedbcb8a64755989938646a8847ce : SUCCESS in 14m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46f8a4a494be4bc1829a62554dcb6126 : SUCCESS in 10m 36s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":9},{"id":"0db3726f8b88b3e4c8530442cc7e10a2c249fff5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:06:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"3ee69447f8d7dcd8f6193593c0de85ce39b4cac8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 14:09:00.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7ad7cb54013448f9fabe63f4543ab3c : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"c6c1cde3f9cf8bb232071b78ab89c79a0c7689b8","revisions":{"318b9d54d8c63b790858997356aaba421be2cc1e":{"kind":"REWORK","_number":1,"created":"2021-06-03 14:35:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/794601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/794601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/794601/1"}}},"commit":{"parents":[{"commit":"47eaed441fa72482dd966cc6045fb0b149debcb5","subject":"Fix container image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47eaed441fa72482dd966cc6045fb0b149debcb5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:31:43.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:34:50.000000000","tz":60},"subject":"CI: test base container image build in seed jobs","message":"CI: test base container image build in seed jobs\n\nCurrently we do not have any coverage of building container images. This\nchange adds a basic test of building the base image to the seed jobs.\n\nA potential future enhancement could involve pushing to a local registry\nrunning on the seed.\n\nChange-Id: Id7378dd844aca7cb5634535308eba0f60342c81d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/318b9d54d8c63b790858997356aaba421be2cc1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/318b9d54d8c63b790858997356aaba421be2cc1e"}]},"branch":"refs/heads/master"},"d7e65cb53f2cb8f92499ad308c82b2b6d9a37ab8":{"kind":"REWORK","_number":2,"created":"2021-06-07 08:55:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/794601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/794601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/794601/2"}}},"commit":{"parents":[{"commit":"47eaed441fa72482dd966cc6045fb0b149debcb5","subject":"Fix container image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47eaed441fa72482dd966cc6045fb0b149debcb5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:31:43.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 15:35:41.000000000","tz":60},"subject":"CI: test base container image build in seed jobs","message":"CI: test base container image build in seed jobs\n\nCurrently we do not have any coverage of building container images. This\nchange adds a basic test of building the base image to the seed jobs.\n\nA potential future enhancement could involve pushing to a local registry\nrunning on the seed.\n\nChange-Id: Id7378dd844aca7cb5634535308eba0f60342c81d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7e65cb53f2cb8f92499ad308c82b2b6d9a37ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7e65cb53f2cb8f92499ad308c82b2b6d9a37ab8"}]},"branch":"refs/heads/master"},"b744595432711a46e06ab401d591f37022016226":{"kind":"REWORK","_number":3,"created":"2021-06-07 09:41:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/794601/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/794601/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/794601/3"}}},"commit":{"parents":[{"commit":"47eaed441fa72482dd966cc6045fb0b149debcb5","subject":"Fix container image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47eaed441fa72482dd966cc6045fb0b149debcb5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:31:43.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 09:41:55.000000000","tz":0},"subject":"CI: test base container image build in seed jobs","message":"CI: test base container image build in seed jobs\n\nCurrently we do not have any coverage of building container images. This\nchange adds a basic test of building the base image to the seed jobs.\n\nA potential future enhancement could involve pushing to a local registry\nrunning on the seed.\n\nChange-Id: Id7378dd844aca7cb5634535308eba0f60342c81d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b744595432711a46e06ab401d591f37022016226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b744595432711a46e06ab401d591f37022016226"}]},"branch":"refs/heads/master"},"27923738ed7a1ed4a841ebdd425d67f9ef056b22":{"kind":"REWORK","_number":4,"created":"2021-06-07 13:22:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/794601/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/794601/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/794601/4"}}},"commit":{"parents":[{"commit":"47eaed441fa72482dd966cc6045fb0b149debcb5","subject":"Fix container image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47eaed441fa72482dd966cc6045fb0b149debcb5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:31:43.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 10:42:08.000000000","tz":60},"subject":"CI: test base container image build in seed jobs","message":"CI: test base container image build in seed jobs\n\nCurrently we do not have any coverage of building container images. This\nchange adds a basic test of building the base image to the seed jobs.\n\nA potential future enhancement could involve pushing to a local registry\nrunning on the seed.\n\nChange-Id: Id7378dd844aca7cb5634535308eba0f60342c81d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/27923738ed7a1ed4a841ebdd425d67f9ef056b22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/27923738ed7a1ed4a841ebdd425d67f9ef056b22"}]},"branch":"refs/heads/master"},"bc5ed3b04529d096d82972ed262ccf8306821eca":{"kind":"REWORK","_number":5,"created":"2021-06-07 15:27:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/794601/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/794601/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/794601/5"}}},"commit":{"parents":[{"commit":"47eaed441fa72482dd966cc6045fb0b149debcb5","subject":"Fix container image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47eaed441fa72482dd966cc6045fb0b149debcb5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:31:43.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 15:27:24.000000000","tz":0},"subject":"CI: test base container image build in seed jobs","message":"CI: test base container image build in seed jobs\n\nCurrently we do not have any coverage of building container images. This\nchange adds a basic test of building the base image to the seed jobs.\n\nA potential future enhancement could involve pushing to a local registry\nrunning on the seed.\n\nChange-Id: Id7378dd844aca7cb5634535308eba0f60342c81d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc5ed3b04529d096d82972ed262ccf8306821eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc5ed3b04529d096d82972ed262ccf8306821eca"}]},"branch":"refs/heads/master"},"06f779de24dc47125271613685ea1c8a49ff83ff":{"kind":"REWORK","_number":6,"created":"2021-06-07 16:41:34.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/794601/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/794601/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/794601/6"}}},"commit":{"parents":[{"commit":"47eaed441fa72482dd966cc6045fb0b149debcb5","subject":"Fix container image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47eaed441fa72482dd966cc6045fb0b149debcb5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:31:43.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 16:41:30.000000000","tz":0},"subject":"CI: test base container image build in seed jobs","message":"CI: test base container image build in seed jobs\n\nCurrently we do not have any coverage of building container images. This\nchange adds a basic test of building the base image to the seed jobs.\n\nA potential future enhancement could involve pushing to a local registry\nrunning on the seed.\n\nChange-Id: Id7378dd844aca7cb5634535308eba0f60342c81d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06f779de24dc47125271613685ea1c8a49ff83ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/06f779de24dc47125271613685ea1c8a49ff83ff"}]},"branch":"refs/heads/master"},"5493e2305806ae82766d6688e144fd9d88a72bc8":{"kind":"REWORK","_number":7,"created":"2021-06-08 08:01:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/794601/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/794601/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/794601/7"}}},"commit":{"parents":[{"commit":"47eaed441fa72482dd966cc6045fb0b149debcb5","subject":"Fix container image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47eaed441fa72482dd966cc6045fb0b149debcb5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:31:43.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-08 08:01:24.000000000","tz":0},"subject":"CI: test base container image build in seed jobs","message":"CI: test base container image build in seed jobs\n\nCurrently we do not have any coverage of building container images. This\nchange adds a basic test of building the base image to the seed jobs.\n\nA potential future enhancement could involve pushing to a local registry\nrunning on the seed.\n\nChange-Id: Id7378dd844aca7cb5634535308eba0f60342c81d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5493e2305806ae82766d6688e144fd9d88a72bc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5493e2305806ae82766d6688e144fd9d88a72bc8"}]},"branch":"refs/heads/master"},"1eba438602673ea562a0a71945062902d980b862":{"kind":"REWORK","_number":8,"created":"2021-06-08 08:25:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/794601/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/794601/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/794601/8"}}},"commit":{"parents":[{"commit":"47eaed441fa72482dd966cc6045fb0b149debcb5","subject":"Fix container image builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47eaed441fa72482dd966cc6045fb0b149debcb5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:31:43.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-08 08:25:53.000000000","tz":0},"subject":"CI: test base container image build in seed jobs","message":"CI: test base container image build in seed jobs\n\nCurrently we do not have any coverage of building container images. This\nchange adds a basic test of building the base image to the seed jobs.\n\nA potential future enhancement could involve pushing to a local registry\nrunning on the seed.\n\nChange-Id: Id7378dd844aca7cb5634535308eba0f60342c81d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1eba438602673ea562a0a71945062902d980b862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1eba438602673ea562a0a71945062902d980b862"}]},"branch":"refs/heads/master"},"c6c1cde3f9cf8bb232071b78ab89c79a0c7689b8":{"kind":"REWORK","_number":9,"created":"2021-06-08 12:47:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/794601/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/794601/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/794601/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/794601/9"}}},"commit":{"parents":[{"commit":"05225342a8b275e9621d085830f290b831b8bee6","subject":"Merge \"Fix container image builds\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05225342a8b275e9621d085830f290b831b8bee6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:31:43.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-08 12:47:04.000000000","tz":60},"subject":"CI: test base container image build in seed jobs","message":"CI: test base container image build in seed jobs\n\nCurrently we do not have any coverage of building container images. This\nchange adds a basic test of building the base image to the seed jobs.\n\nA potential future enhancement could involve pushing to a local registry\nrunning on the seed.\n\nChange-Id: Id7378dd844aca7cb5634535308eba0f60342c81d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6c1cde3f9cf8bb232071b78ab89c79a0c7689b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c6c1cde3f9cf8bb232071b78ab89c79a0c7689b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
