)]}'
{"id":"openstack%2Fkayobe~794767","triplet_id":"openstack%2Fkayobe~master~I4613e8973ccaef4534e44e3c79e910de5041add0","project":"openstack/kayobe","branch":"master","topic":"ironic-inspector-plugins-order","hashtags":[],"change_id":"I4613e8973ccaef4534e44e3c79e910de5041add0","subject":"Optimise order of inspector_processing_hooks","status":"MERGED","created":"2021-06-04 10:13:59.000000000","updated":"2021-06-04 17:49:21.000000000","submitted":"2021-06-04 17:48:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794767-ironic-inspector-plugins-order","meta_rev_id":"b3bb89037679330495488e099d84298006847884","_number":794767,"virtual_id_number":794767,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-04 17:48:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-04 15:58:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-04 12:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-06-04 15:58:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-04 12:19:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 12:03:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 12:19:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-04 15:52:13.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-06-04 15:58:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"24c6a51c80d3c7b9b3343e3b0a64da7128e0e840","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-04 10:13:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f29385b8767de0ded2e4b82472e4b88dc7a9a07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 12:03:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ab8e09d56594d43990a096af6e9ffcf : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efe52461f51d4b9bacf853dad2dec00c : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19419ce830464515934881d4be4578ac : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/762c7c0bacc14fb1ad11e647d773cd6a : SUCCESS in 10m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6761b6c2baca44189e1051bad130c019 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5612c94360694c989d96ceeaf268c9e6 : SUCCESS in 9m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bf6eab4a94464c07a7828812c18e6f57 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/89b1b8581e9646bea52cab3e34f4e7ae : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37a2b423463e4a8bb44e9f6d18c3b87c : SUCCESS in 21m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9589016befb04a18bdf468c75dac8053 : SUCCESS in 1h 02m 10s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b80f56a0850244918678313638667190 : SUCCESS in 43m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c4a63dc82680479cb42450308e5ce70b : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/485e6c4f79de4448afaf96fb5684e8a7 : SUCCESS in 17m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54fbd143df8240168a3dd54cae9273e2 : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bf8c6fc3294a4c4c9bd7e6767a53bc07 : SUCCESS in 1h 48m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d22f17a6796747b7ac8782166619dcf1 : SUCCESS in 26m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d37a373c2b78451b9652a80ac1bc8ac2 : SUCCESS in 23m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cadb1de797c34835a32775a6bfbf9d44 : SUCCESS in 26m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e09f82e6eebc4e0aa8aa56d6dea4ba1d : SUCCESS in 12m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0802be29d1d84ca9933cb6ca3018e592 : SUCCESS in 9m 09s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"d38d70ce11e057e2a65e2f0109ea8d6c7feeca60","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-04 12:19:25.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2ba26e047a53747e9607b195634aee61725916e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-04 15:52:13.000000000","message":"Patch Set 1:\n\nIs there an issue that this fixes?","accounts_in_message":[],"_revision_number":1},{"id":"08083fea5f14e11699df5a18bf8c78e121553bf4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-04 15:53:40.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Is there an issue that this fixes?\n\nNo bug to fix, it\u0027s only an optimisation of the flow (parsing lldp once).","accounts_in_message":[],"_revision_number":1},{"id":"23961e24e4c5e12fc60712c5f5cf39f6021a2639","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-04 15:56:46.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Is there an issue that this fixes?\n\u003e \n\u003e No bug to fix, it\u0027s only an optimisation of the flow (parsing lldp once).\n\nAlso, in the general spirit of \"do it like the others do\", it proactively shields you from \"only us\" issues. 🐨","accounts_in_message":[],"_revision_number":1},{"id":"c318b086c05ba40a46790c9550a13c50125d8192","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-04 15:58:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0a46181c01d2256ff6bfad31952d903e0581a60c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 15:58:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7de7769bbab12a3c57f84ad37c7c1e949872b47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:48:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/867bc6117bf2470a88e79de0ed0302e4 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de22ee8a97fc4416b7f94f3d33dac4a9 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/454e639a8a9d4099836155851627b4eb : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1980c9609f240b18d7de899923127d5 : SUCCESS in 10m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae79a3b3679c4004ad71541709d26d15 : SUCCESS in 6m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c437b91ffe6f4f0787312fb97ff794ad : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/216a3fb4d5714a43bcce8b8cab814b69 : SUCCESS in 19m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d672a3e2b6884c7ba037551aef5e74b5 : SUCCESS in 1h 04m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3662f995d3384ebaab50ce9e611a9a90 : SUCCESS in 47m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8aa553dc7d3f4f6c8bf1ad74579c209c : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7968bce0b1164a2288c4a644d1c7183c : SUCCESS in 16m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e1819f89cc53446586cbc8736e618352 : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb19a4dd6b514c628682159f2a6e7a27 : SUCCESS in 1h 42m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8fe489d3be7a4b498b86a9c18cf7b03e : SUCCESS in 25m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cbb069d4fdd4f23b6394f4ba8402448 : SUCCESS in 20m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3dc012832be240c0b13b0e33322f3c54 : SUCCESS in 29m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3f145e157074440eb812e7ec360bb47e : SUCCESS in 13m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae53c4b5562a44e2a632ce1caf9d8338 : SUCCESS in 10m 05s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"99a7d9d67c8ee8b980e3bf69eeae0de3df58e99b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:48:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b3bb89037679330495488e099d84298006847884","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:49:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/401f7c16e2254251a71adf20ea127961 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a52b3b27be4ddabc49c69ddd8ea2b264427d630c","revisions":{"a52b3b27be4ddabc49c69ddd8ea2b264427d630c":{"kind":"REWORK","_number":1,"created":"2021-06-04 10:13:59.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/67/794767/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/794767/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/794767/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/794767/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/794767/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/794767/1"}}},"commit":{"parents":[{"commit":"9883726bf2b43d7ff6abc4a071cb18d212623938","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9883726bf2b43d7ff6abc4a071cb18d212623938"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-04 10:10:23.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-04 10:13:57.000000000","tz":0},"subject":"Optimise order of inspector_processing_hooks","message":"Optimise order of inspector_processing_hooks\n\nIronic Inspector docs [1] recommend to use lldp_basic before\nlocal_link_connection.\nIt is also how TripleO applies them.\n\n[1] https://docs.openstack.org/ironic-inspector/wallaby/user/usage.html#plugins\n\nChange-Id: I4613e8973ccaef4534e44e3c79e910de5041add0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a52b3b27be4ddabc49c69ddd8ea2b264427d630c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a52b3b27be4ddabc49c69ddd8ea2b264427d630c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
