)]}'
{"id":"openstack%2Fkayobe~794818","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I4613e8973ccaef4534e44e3c79e910de5041add0","project":"openstack/kayobe","branch":"stable/wallaby","topic":"ironic-inspector-plugins-order-stable/wallaby","hashtags":[],"change_id":"I4613e8973ccaef4534e44e3c79e910de5041add0","subject":"Optimise order of inspector_processing_hooks","status":"MERGED","created":"2021-06-04 18:33:22.000000000","updated":"2021-06-07 11:31:00.000000000","submitted":"2021-06-07 11:29:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794818-ironic-inspector-plugins-order-stable/wallaby","cherry_pick_of_change":794767,"cherry_pick_of_patch_set":1,"meta_rev_id":"675fa8f3da50beda393e51d02346c4152ec7ea10","_number":794818,"virtual_id_number":794818,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-07 11:29:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-07 09:22:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-04 20:03:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-06-07 09:22:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 20:03:37.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-04 20:24:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 09:22:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"def343888245c41b80f0a93a9dd23fefaa6ed318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-04 18:33:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9e866bb05a957bedb8dc43cc036db0ba7fc8bf31","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-04 18:33:42.000000000","message":"Patch Set 1:\n\nbackporting to wallaby as it has not been released","accounts_in_message":[],"_revision_number":1},{"id":"e9398b5abd00e8dd7c3ae28934eee74be16e251f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-04 20:03:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fad2aedad71bd0ab42dd00e01cd9509f960ebeb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 20:24:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b67fe58e59624b51ba8a843a25ffacaf : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dbc25f297764b408c9878ddb73c8b26 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/592acd2cea60472b89dc2bcac82af56b : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d50b7de7b47d49639509bddc187e11e5 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09e30eccc16b42d0afe4e35717d8e63a : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2ec39e65c64a1e871e26ee97890f63 : SUCCESS in 8m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e354d37f148240279e98791e86bf22df : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4cf6ddd0edbd4c1e8d80b39bcd55db99 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/79618e7f2a974a379f3c8a6b57a6d223 : SUCCESS in 18m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4bd15fc4851f4eeb9992376126b1c5d3 : SUCCESS in 1h 11m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02afbe1ecba645afac5788da50e95587 : SUCCESS in 42m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d590fd84668f406eac47d0dc4c96752c : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0cda13916d9147feb0a9dddea42a2dec : SUCCESS in 17m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02ef4e7fc65740919620eab11974b56e : SUCCESS in 13m 05s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6edc825ec81f472f99714594691b74dc : SUCCESS in 1h 49m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/00ce998f4b3443728f4c083cdb94298d : SUCCESS in 25m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f242f222376a4b2f899dd89fcaa993f4 : SUCCESS in 22m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d7def2759073456a9327a078493b4c0d : SUCCESS in 25m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae06c1f1418d4d1a87d1f1c1473171f6 : SUCCESS in 12m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52243f5e85364940a226e0078f5865c7 : SUCCESS in 10m 29s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"e2fc1ec2d3bf5ff60ea11c0e206c7b848b6ba58b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 09:22:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e78e6e7f1548df8ce5853f24effd8d64abe22680","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 09:23:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97aafd8e2d494dc1b5e7902e7d2bb74f12609738","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 11:29:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a26d3d418ac45ed850fa2e0cb1bae65 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc98a9a057d54a96b19ade2bea14a97b : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12ed94ef2d4541e9be4ab8cca8cc6835 : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be92275c3fd46428e16b85f86aadb88 : SUCCESS in 9m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c7bbc31514024edd9d68fa33825ddee9 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cace271cbeb84e79ac425e4ca62af6f8 : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7dc69519ff6e43e6ad2d517500b98f5e : SUCCESS in 22m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a40925ae12e64b348ea34ebc48bacc1a : SUCCESS in 1h 20m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81468df5c544489ba8775fd4f853752e : SUCCESS in 47m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/02add34ed0d94e3580e4cbbe416d01af : SUCCESS in 53m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bdd658d3f1ff424ead2e70fb133f41f0 : SUCCESS in 19m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4db3f7f485664fe4925af0bd642db9db : SUCCESS in 14m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8ce2328276b048c8997694a9a9d1905d : SUCCESS in 2h 01m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0f2cf2f01a49448b8c1fe130b5e5e038 : SUCCESS in 28m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bc36c5a7a7745e7a71dc1590375e98d : SUCCESS in 25m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d87c9b126abc4e0b8a248106bb0d3336 : SUCCESS in 22m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a01b15d213cf436db02b45e1d8f3fac5 : SUCCESS in 13m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/305e85cc84494b8aa49077ca76174caf : SUCCESS in 10m 30s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e0145447c51f14c504f064927cbeeec71361ee70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 11:29:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"675fa8f3da50beda393e51d02346c4152ec7ea10","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 11:31:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024b8ba7ad2d432c9c24661519f804b4 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a374af8d4616d3a5c553ba70707d74e7091ef6e","revisions":{"9a374af8d4616d3a5c553ba70707d74e7091ef6e":{"kind":"REWORK","_number":1,"created":"2021-06-04 18:33:22.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/18/794818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/794818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/794818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/794818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/794818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/794818/1"}}},"commit":{"parents":[{"commit":"4648b90bcf31cd6c293653c7500b5f4ae0a2eddc","subject":"ntp: check for ntp group in inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4648b90bcf31cd6c293653c7500b5f4ae0a2eddc"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-04 10:10:23.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-04 18:33:22.000000000","tz":0},"subject":"Optimise order of inspector_processing_hooks","message":"Optimise order of inspector_processing_hooks\n\nIronic Inspector docs [1] recommend to use lldp_basic before\nlocal_link_connection.\nIt is also how TripleO applies them.\n\n[1] https://docs.openstack.org/ironic-inspector/wallaby/user/usage.html#plugins\n\nChange-Id: I4613e8973ccaef4534e44e3c79e910de5041add0\n(cherry picked from commit a52b3b27be4ddabc49c69ddd8ea2b264427d630c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a374af8d4616d3a5c553ba70707d74e7091ef6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a374af8d4616d3a5c553ba70707d74e7091ef6e"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
