)]}'
{"id":"openstack%2Fkayobe~795068","triplet_id":"openstack%2Fkayobe~master~Ie94099075bae3c491f9cf830c38e6cfc8af605a6","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ie94099075bae3c491f9cf830c38e6cfc8af605a6","subject":"Ubuntu: ensure /usr/bin/python is python3","status":"MERGED","created":"2021-06-07 09:02:37.000000000","updated":"2021-06-07 18:08:20.000000000","submitted":"2021-06-07 18:05:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795068","meta_rev_id":"9dfdcea366b16af84c5f810b04770acdb688e09c","_number":795068,"virtual_id_number":795068,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-07 18:05:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-07 09:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-07 15:27:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-07 15:27:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 09:11:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-06-07 09:45:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-06-07 11:01:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 15:27:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"13910755c76b17840ae2b2d87910d1358c963cd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 09:02:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3f372ddb54dc70a7f969cdcc80fc9ff5a1d6139","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-07 09:11:44.000000000","message":"Patch Set 1:\n\nI thought the only place where we use /usr/bin/python is virtualenv - and we create it with -ppython3?","accounts_in_message":[],"_revision_number":1},{"id":"1654c9b766a03d2f9c548fac94c31755b43dadbb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 09:21:48.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I thought the only place where we use /usr/bin/python is virtualenv - and we create it with -ppython3?\n\nThis affects the control host. Ansible is usually executed from a virtualenv there, but the remote execution will default to use /usr/bin/python if present.","accounts_in_message":[],"_revision_number":1},{"id":"3cd5366f6ca621606ca61b29bc1a3e1480f8f9c2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-07 09:45:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"daded1470c0860ae7437d36972456059d1b1edc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 11:01:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc1cc0a63c3749cbaaa2dd873598644a : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98f02d963ef941a7830e8096ed265dff : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52902eff82f2439e83697380bda2aacf : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7faa7838a4684e2cb0f84c50b7726c1e : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b760ff25a1b4d288e98bf735658254a : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d93273efa16448a0947f74ed1fffafb1 : SUCCESS in 9m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2de9738641cc47f7a998b6612cf15b72 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/acb8fbd48e6e4495953e2cefae33f18f : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16ba09e4ca3c41e4b77b20246a9f829d : SUCCESS in 21m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c25d2cb84c9a4f11807e951f183ae6b4 : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a45b56bed604dfba6ce84d768c4f6a3 : SUCCESS in 47m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4406091176e140308851c0675fff22dc : SUCCESS in 1h 05m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d27d8a94412c484e87ee9ee6de00d41d : SUCCESS in 16m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/489dae47114b4a39b73ce4482fb9ae16 : SUCCESS in 14m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b6cb49021f014fce8f7532043ba3bf5d : SUCCESS in 1h 50m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d12def1b4f8d4331b153f82007c521bd : SUCCESS in 25m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9f4b21a7cdb4c979181af305c917dcb : SUCCESS in 19m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0fdc27a92b784a13988ab85148482993 : SUCCESS in 22m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8ca6fa5ea9d04059b9c41fcaa8e0921d : SUCCESS in 13m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f601038f84a4ca9ac5bf203d4f08a19 : SUCCESS in 11m 02s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"bd5f09d0bb3f4386ad981c85e55271ad12fe878c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-07 15:27:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e6ef2b8fb13169d39039b1ff97568816ef5892eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 15:28:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6d06fd4cbd4d9bbe883bdf871ecf53f915d10fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 18:05:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a71f0f020a9d43c6a78b53d359818bca : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13ea6870bbab4052bc246b3ead321ef1 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96a0179308a14aff906657c03688a124 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76801571cf5a4e0487479fb2d0b929b2 : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/021b20b4a7094f3f80ada914897973ab : SUCCESS in 7m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d9e4652d8ff4a6f843f98368024fb81 : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7841be79c87e4d4d99e736b7e5c93a98 : SUCCESS in 20m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/420c3e33f7d24031ba878690554843f2 : SUCCESS in 1h 08m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d742747164b54383bc8877c58752c4b0 : SUCCESS in 45m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f243f3e14ef449b294259344a11cb61f : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/06e4cfd266fc405ebcc9c06765d13e8f : SUCCESS in 19m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d17eb5ad971f4903bf730b39f61cd814 : SUCCESS in 15m 42s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c47f2a2c7a5b4a598acbc4c7cdf69542 : SUCCESS in 1h 55m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/86c69921ca1b4080b4235bff0fe16f56 : SUCCESS in 26m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca8aa38e77ac4091980b0b97e51e7542 : SUCCESS in 25m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0c431e7cf7eb4fc1a50ef6bf0ee40f41 : SUCCESS in 23m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5f7b321c4211420da0373faee0bceff8 : SUCCESS in 12m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f5f971a74a84739b086ef193238b775 : SUCCESS in 9m 56s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c77c27f04c4634fe0613b35cc25e1b3356110f54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 18:05:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9dfdcea366b16af84c5f810b04770acdb688e09c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 18:08:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6fd0894d3049f3b3ff6f8d19e94be0 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82def9a51641106ed92569e6d132527934350bac","revisions":{"82def9a51641106ed92569e6d132527934350bac":{"kind":"REWORK","_number":1,"created":"2021-06-07 09:02:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/68/795068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/68/795068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/795068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/795068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/68/795068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/68/795068/1"}}},"commit":{"parents":[{"commit":"211c68aa6c1bb13ab5bd259f1ae45142871a8553","subject":"Merge \"Support additional pass through host variables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/211c68aa6c1bb13ab5bd259f1ae45142871a8553"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 08:58:31.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 09:00:40.000000000","tz":60},"subject":"Ubuntu: ensure /usr/bin/python is python3","message":"Ubuntu: ensure /usr/bin/python is python3\n\n/usr/bin/python may be python 2 on Focal, which causes problems with\nAnsible on the control host. By installing the python-is-python3 package\nwe ensure that the correct interpreter is used.\n\nThis change updates the installation documentation and development\nenvironment scripts.\n\nStory: 2004960\nTask: 42579\n\nChange-Id: Ie94099075bae3c491f9cf830c38e6cfc8af605a6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82def9a51641106ed92569e6d132527934350bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82def9a51641106ed92569e6d132527934350bac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
