)]}'
{"id":"openstack%2Fkayobe~795116","triplet_id":"openstack%2Fkayobe~master~If14fc6485620a1c9ef01ac37a1914c64606a2f70","project":"openstack/kayobe","branch":"master","attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-02-07 15:33:13.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"If14fc6485620a1c9ef01ac37a1914c64606a2f70","subject":"WIP: add support for mitogen","status":"NEW","created":"2021-06-07 14:32:49.000000000","updated":"2025-02-18 15:28:46.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"cfac041f4f2f3d479a0dab50845ea07f49ccc841","_number":795116,"virtual_id_number":795116,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-09 13:18:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"CC":[{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"}]},"reviewer_updates":[{"updated":"2021-06-07 15:42:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 15:33:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-02-18 15:28:46.000000000","updated_by":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"reviewer":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"state":"CC"}],"messages":[{"id":"d08383068829cbe0d8d2eba6b504e9c33f2f67db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 14:32:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdf879b43e0e62dbb8151e674451b930b6e4642d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 15:42:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7875f9244b9246a5b2abf5120dfa26ad : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e0557fa5299415ebc8498a311983925 : SUCCESS in 48m 02s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4fab2424fbb746bfb24ecad415a5c3cc : FAILURE in 13m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0afc00a9199741f5be411ff153035e88 : FAILURE in 11m 03s","accounts_in_message":[],"_revision_number":1},{"id":"e2d230f2cd61117d8e7d0918c8d3be089c0c247a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-07 16:23:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e05f9cc4490828bb8884b86a10b590b6f8810568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 17:51:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c81753ea2eb04c00b5fc2d86e3c760cc : SUCCESS in 1h 05m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88c9a6e65dcf4125a46aa30cdbc9bac8 : SUCCESS in 47m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fad7dfba8b3747aaa6d7b226c2fed327 : SUCCESS in 30m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38b45685b3284042bb38b4b001cd7c86 : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":2},{"id":"9c406df1ae278acaf2ecc8734f60dec76ec73a29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 08:11:31.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"941d826501496cf33c9a7a0ebfb8b10c84994d11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 09:05:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cb848c754f25449b915327e0ce47c67f : RETRY_LIMIT in 6m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da10755ce84b41a9a47bb002939632c8 : RETRY_LIMIT in 5m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/29abbe20c6f04577a90e524f27d07668 : FAILURE in 8m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eca735d02a4447eba1938e8efdc75fb5 : FAILURE in 7m 38s","accounts_in_message":[],"_revision_number":3},{"id":"6332eb0bfea5ffae67936f4c57263877acb7bacb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 14:45:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"82a98d9a5f15cc923835d03ce7ad10df03dd8429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 15:48:03.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c53b0538e21547fa90d67d5cff6f57b1 : FAILURE in 9m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0e8a275867546d4a5f9899018e7b521 : RETRY_LIMIT in 6m 15s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/92660b6dcbe141ee8288bcfa17f57322 : FAILURE in 9m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a52b3839b5ea4fa3b5e95c7b1a55a9d9 : FAILURE in 6m 56s","accounts_in_message":[],"_revision_number":4},{"id":"c71aa045811a2b5964b2e554cc0ef1fc2c9af8a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 12:45:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"de2ea2497643c1a250e3b6787803698a88d2bb77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 13:18:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2f483417806949daa8883539a3818fc0 : FAILURE in 10m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdb70c86eb4c4803aae3d3b39cd4c5b9 : FAILURE in 19m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/df90b52323444543add2c1de4eccd17f : FAILURE in 10m 11s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d1e79b578a84d85a3301947405fdb5a : FAILURE in 15m 23s","accounts_in_message":[],"_revision_number":5},{"id":"c2c8dddf4aa36de58dce0d349ddd4a7abf55d6c6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-16 13:42:19.000000000","message":"Patch Set 5:\n\nPackage module issues again.","accounts_in_message":[],"_revision_number":5},{"id":"fa9c7d7c1595c0fac60a990c4eb6138cc80693dc","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-16 13:42:22.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d7372aa762d0b9edb831e5b7489b99b3242c0f94","revisions":{"a3706bd88e7f01aa904f44663972edbed6915404":{"kind":"REWORK","_number":1,"created":"2021-06-07 14:32:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/795116/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/795116/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/795116/1"}}},"commit":{"parents":[{"commit":"091a262908b137a9e87b93ffc4503c365339d29f","subject":"Merge \"Optimise order of inspector_processing_hooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/091a262908b137a9e87b93ffc4503c365339d29f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 14:32:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 14:32:42.000000000","tz":60},"subject":"WIP: add support for molecule","message":"WIP: add support for molecule\n\nCurrently the master branch of molecule is required to use Ansible 2.10.\n\nChange-Id: If14fc6485620a1c9ef01ac37a1914c64606a2f70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3706bd88e7f01aa904f44663972edbed6915404"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3706bd88e7f01aa904f44663972edbed6915404"}]},"branch":"refs/heads/master"},"628ddedd2f6822a781d03406ffb3b92cd392dee1":{"kind":"REWORK","_number":2,"created":"2021-06-07 16:23:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/795116/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/795116/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/795116/2"}}},"commit":{"parents":[{"commit":"091a262908b137a9e87b93ffc4503c365339d29f","subject":"Merge \"Optimise order of inspector_processing_hooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/091a262908b137a9e87b93ffc4503c365339d29f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 14:32:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 16:23:12.000000000","tz":60},"subject":"WIP: add support for molecule","message":"WIP: add support for molecule\n\nCurrently the master branch of molecule is required to use Ansible 2.10.\n\nChange-Id: If14fc6485620a1c9ef01ac37a1914c64606a2f70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/628ddedd2f6822a781d03406ffb3b92cd392dee1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/628ddedd2f6822a781d03406ffb3b92cd392dee1"}]},"branch":"refs/heads/master"},"534e33955c7f87a4f0fb2335a7946093ecd50f37":{"kind":"REWORK","_number":3,"created":"2021-06-08 08:11:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/795116/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/795116/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/795116/3"}}},"commit":{"parents":[{"commit":"091a262908b137a9e87b93ffc4503c365339d29f","subject":"Merge \"Optimise order of inspector_processing_hooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/091a262908b137a9e87b93ffc4503c365339d29f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 14:32:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-08 08:11:27.000000000","tz":0},"subject":"WIP: add support for mitogen","message":"WIP: add support for mitogen\n\nCurrently the master branch of mitogen is required to use Ansible 2.10.\n\nChange-Id: If14fc6485620a1c9ef01ac37a1914c64606a2f70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/534e33955c7f87a4f0fb2335a7946093ecd50f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/534e33955c7f87a4f0fb2335a7946093ecd50f37"}]},"branch":"refs/heads/master"},"6f18c2af000c82e7d3536bb0e44ba606de5cb10b":{"kind":"REWORK","_number":4,"created":"2021-06-08 14:45:56.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/795116/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/795116/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/795116/4"}}},"commit":{"parents":[{"commit":"091a262908b137a9e87b93ffc4503c365339d29f","subject":"Merge \"Optimise order of inspector_processing_hooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/091a262908b137a9e87b93ffc4503c365339d29f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 14:32:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-08 14:45:38.000000000","tz":60},"subject":"WIP: add support for mitogen","message":"WIP: add support for mitogen\n\nCurrently the master branch of mitogen is required to use Ansible 2.10.\n\nChange-Id: If14fc6485620a1c9ef01ac37a1914c64606a2f70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f18c2af000c82e7d3536bb0e44ba606de5cb10b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f18c2af000c82e7d3536bb0e44ba606de5cb10b"}]},"branch":"refs/heads/master"},"d7372aa762d0b9edb831e5b7489b99b3242c0f94":{"kind":"REWORK","_number":5,"created":"2021-06-09 12:45:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/795116/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/795116/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/795116/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/795116/5"}}},"commit":{"parents":[{"commit":"091a262908b137a9e87b93ffc4503c365339d29f","subject":"Merge \"Optimise order of inspector_processing_hooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/091a262908b137a9e87b93ffc4503c365339d29f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-07 14:32:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-09 12:45:48.000000000","tz":60},"subject":"WIP: add support for mitogen","message":"WIP: add support for mitogen\n\nCurrently the master branch of mitogen is required to use Ansible 2.10.\n\nChange-Id: If14fc6485620a1c9ef01ac37a1914c64606a2f70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7372aa762d0b9edb831e5b7489b99b3242c0f94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7372aa762d0b9edb831e5b7489b99b3242c0f94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
