)]}'
{"id":"openstack%2Fkayobe~795178","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I275102ec6b5bab6982577b52fd29654c874446ce","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"I275102ec6b5bab6982577b52fd29654c874446ce","subject":"chrony: cleanup during overcloud host upgrade","status":"MERGED","created":"2021-06-07 20:28:18.000000000","updated":"2021-06-08 09:43:17.000000000","submitted":"2021-06-08 09:41:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795178","meta_rev_id":"761f7673a03d748a2763497f798077cd62006287","_number":795178,"virtual_id_number":795178,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-08 09:41:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-08 07:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-08 07:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-06-08 07:57:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 20:28:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-07 22:17:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-08 07:56:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"b745369838fdc347ac15abe225cb191f391bbf86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-07 20:28:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1762637f9aa76e13cd47de59c3bbf3b72e1abd75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 22:17:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ea6fcf571404088b00568e1c3524283 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/775c7862f3d24f18bde34619b0a31fa1 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd4796ebcf7b4824b4293ea02a871519 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71074cad0dae40cfbabd92b04b7ca764 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8348b16bc2ac48b4bb62c74a4c7003d0 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/959b21e3512144f2ac72e6618859fdaf : SUCCESS in 9m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0d4bd3f84b74c928ebe31d4c0573eec : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff4a0888edb848dbb5f2fe83739f0c14 : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/380840dc4289480cb5b4b24966b8ae7f : SUCCESS in 18m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6725f993c5454458ae01556ee42c181f : SUCCESS in 1h 13m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2716c5991bfd456aa2b14da9aa207449 : SUCCESS in 47m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3855cad2deeb473d8e43fc267cbefc59 : SUCCESS in 1h 12m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a3a94fd654f744d9918197c402b75824 : SUCCESS in 18m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0765ee81fc1747228e6b1be3489aeb5a : SUCCESS in 13m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a16b45b48a74435dad0bb0827e2a3c14 : SUCCESS in 1h 48m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c75f5137f32b4c6195bad358b8c6ad96 : SUCCESS in 26m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/482a78ef866e43c49256767c24d14e20 : SUCCESS in 22m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4390d0be3d1a450babb70408f2987985 : SUCCESS in 26m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3fee3a40391d4fb4bd9b0efe8a323751 : SUCCESS in 12m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06b2a3af7a6743359e0cf9e87c28f32d : SUCCESS in 11m 28s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"90c1ba029b7eb1e994d0bf3cafd3ff0718a0d690","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-08 07:56:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"662d1f1ff1ee3799c23be3189cd7842533bd1f56","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 07:57:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e65283b2f5634e9e7a0fe5c4bc7c49b252c9917","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 07:57:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b33b9b525be8e833ca4ca6091c56144e6376a9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 09:41:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/258f5bb9f7f44f2b8e74f10e417053c2 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e39e5b6729f445782427e7bf79ee075 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bb9e9f3922b4d6092ddb735faf3b607 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/131b8a92538e4e3db99ff66069e917c4 : SUCCESS in 8m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8828ed95c7d840d8b029d6152767c710 : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/055f749a52764b84883750b865d69d15 : SUCCESS in 12m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1284af50db824602b89e012413414302 : SUCCESS in 22m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/175f623a756446bba79362ede814f5d9 : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/746cfa6d0dc14ddf9b0ac5aad266dd50 : SUCCESS in 50m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b71cc801c8514818a18cfbc93c37b3a7 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f7c72ce28ca74e51be4a12455302bb3e : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d526d7d781446bfb9136877bd2368b3 : SUCCESS in 12m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/910380756f2648918902b6730362c275 : SUCCESS in 1h 39m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/498bfc619bcb4ee0b00880122d776c2a : SUCCESS in 24m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4648d8519ab84fc7ad051c2f4d652152 : SUCCESS in 20m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fbdae07496534fa09b7244c5c0fdb44b : SUCCESS in 25m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bf54e5ec2c0e44869caf58e0cd458966 : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d42d496f3ada496ca6ca2d6e7bdd3797 : SUCCESS in 9m 55s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"cfc8ffc0cffd7bfda0753707552fc542e023bf6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 09:41:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"761f7673a03d748a2763497f798077cd62006287","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 09:43:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db2d4ee93dbb42f59c553b33523ddbb2 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5fe1a7edd8d3c246a3ba10194a9ff5cebd84faea","revisions":{"5fe1a7edd8d3c246a3ba10194a9ff5cebd84faea":{"kind":"REWORK","_number":1,"created":"2021-06-07 20:28:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/795178/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/795178/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/795178/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/795178/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/795178/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/795178/1"}}},"commit":{"parents":[{"commit":"9a374af8d4616d3a5c553ba70707d74e7091ef6e","subject":"Optimise order of inspector_processing_hooks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a374af8d4616d3a5c553ba70707d74e7091ef6e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-19 18:47:36.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-07 20:27:34.000000000","tz":120},"subject":"chrony: cleanup during overcloud host upgrade","message":"chrony: cleanup during overcloud host upgrade\n\nThe Kolla Ansible chrony container is disabled by default in the Wallaby\nrelease. A new kolla-ansible chrony-cleanup command can be used to clean\nup the container.\n\nThis change extends the \u0027kayobe overcloud host upgrade\u0027 command to cover\ncleaning up the chrony container (if disabled) and deploying a host\nchrony daemon.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/795036\n\nChange-Id: I275102ec6b5bab6982577b52fd29654c874446ce\n(cherry picked from commit dc53d29eaa0e5f67335aef3db23fc11a17dc21b0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fe1a7edd8d3c246a3ba10194a9ff5cebd84faea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fe1a7edd8d3c246a3ba10194a9ff5cebd84faea"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
