)]}'
{"id":"openstack%2Fkayobe~795232","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I608c0362e6eeea65ffafb32cda5c19ccb6ec8b5c","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"I608c0362e6eeea65ffafb32cda5c19ccb6ec8b5c","subject":"Fix container image builds","status":"MERGED","created":"2021-06-08 07:57:23.000000000","updated":"2021-06-08 20:34:49.000000000","submitted":"2021-06-08 20:33:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795232","meta_rev_id":"0d7cda748dd81645552c018c50f363bc31e3daed","_number":795232,"virtual_id_number":795232,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-08 20:33:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-08 08:01:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-08 07:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-08 08:01:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-08 07:57:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-08 08:01:31.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-06-08 09:51:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26e016ff86325820743d32704e82452c89d43831","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-08 07:57:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"abcddd9c8297c64714e9ab4633f1136d0249bf11","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-08 07:58:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e8973ab5f8d1fcdfea5a013eb04862af613db6c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-08 08:01:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f8f28964e26736d5554e895f8de40354f34abba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 09:51:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e55c2687365944a6b218fdbba82236d5 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15bc1c3a1a1f402a856398dc1077095a : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/946df6226c87445c929e16d10a6200d0 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb401dce486e4b09aed9ff88d785c7da : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfccb2796de94f5880097d6b21ee84a3 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569f9b21a6c9458092e2ca30b6f75dc3 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49cc3696e62c485ca1c0d30f588598b9 : SUCCESS in 4m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f6ed11fe97047bd8e248a8ababb051f : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2364a46b1bda47689173b886b58734b4 : SUCCESS in 10m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/23f64f7f30654e418092dbaa8ad0dab0 : SUCCESS in 19m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/977cae6ab81f4a7cb385096750d9c808 : SUCCESS in 1h 02m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c55cb92ea2a04b8e9cd125dd04a38a0c : SUCCESS in 49m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d17ee52452fc4f8e88ba6e281d9bedcd : SUCCESS in 1h 10m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8f7d6319bcb7458c92e50b7e321d29ac : SUCCESS in 18m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b0e9e513bf3498ca9e8dfec0952077e : SUCCESS in 13m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8fde947b561247d78ecced518a2f8f2c : SUCCESS in 1h 48m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2ed79bb04409483cb243a465fc54d091 : SUCCESS in 27m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b85fc72c1ad44ea845c2c5157d17403 : SUCCESS in 20m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/afa43644d3154f68b1d9aa044c9f9826 : SUCCESS in 26m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a590dc7b1a74445288fc99c65c1c4e5a : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20c9b408692141998e580dc37b2386af : SUCCESS in 9m 52s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"9abc077ab1028857141b53aabd24731fddaf648c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 09:52:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebe5ce595f16064651fa8267a03f0983671e3198","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 10:46:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16098ad976704e68af254ba7474e4c5f : FAILURE in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/161ba60742654bb89a77e7d555f5af50 : FAILURE in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52208f348dff45889e1d54906c2500a9 : FAILURE in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8514c4353a5645899daf0cf4056ee605 : FAILURE in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8848ac57ec2f48e6ba4948b610e2df02 : SUCCESS in 5m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a5b6239908cb49fea100f54977bea3bb : FAILURE in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b6024ac4da054ac8967fdc3b4d3d379d : FAILURE in 4m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/02516df5a4264186a806cb34261c35e6 : FAILURE in 5m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e379d37f80674ab8b023bf3de44d8826 : FAILURE in 47m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07324b659d114202b17c006f2ac6ef0e : FAILURE in 37m 58s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0a3b278e8d36487d98ac59ad43704aa4 : FAILURE in 52m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/89504697ff724317b2ed90c59ba766bd : SUCCESS in 19m 46s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a94a958dd7d443681c83bcb5d6accf0 : SUCCESS in 15m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/028b803022d646939da50a120c507ca8 : FAILURE in 5m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7f4a1d9aeafc4b49bfee1701f5f9ed9c : FAILURE in 17m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/232e826353b04cdc9a9dbfc73c7e207a : FAILURE in 13m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3181863fceb542a4a947ee888ab978b1 : FAILURE in 10m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bff39330b3a74883b9c7f600681c10ad : SUCCESS in 13m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14cf69725be4460fa5ede90f3541a28b : SUCCESS in 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"69b687dfae56fd58ea9e506c659259593fe77278","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-08 12:26:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5a32e0f17130e63eaecd432d09a89d366e686cee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:28:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bd76a4fa8c148819d1b29ab7dee2fc9 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea82388f317a4798bd5078c335d85960 : SUCCESS in 9m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f3f3ea5c8104a05bddc76ec7a89c628 : SUCCESS in 7m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0709321ed40c4eb3b030e4aee290afba : SUCCESS in 8m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/906cc149fee54c7c83be98692061f420 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/937144d3d71e406eb98fb7a53bd4ecab : SUCCESS in 11m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9cad0f961fc4d09bf5850a5f95d2c92 : SUCCESS in 5m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e055c333abfd455db066a587c89dcd42 : SUCCESS in 7m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9172a6898caf4871a78153b1039d4a61 : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d1c46e0dd17415586f99f9e5ad8bde3 : SUCCESS in 21m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3463cdcdfe814787a0e43c3b253da065 : SUCCESS in 1h 06m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e95d7d83a9f54024b0b78466e98f34d4 : SUCCESS in 51m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f74662938a114c69bc9de61246bf8056 : SUCCESS in 1h 08m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2ae52c13fef74888abf1e2104f51f287 : SUCCESS in 19m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a68b383da40c401aa91e54ebc23ea959 : SUCCESS in 14m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/aea679fc014b4949aa2b4023c1136a1b : SUCCESS in 1h 57m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b45b44981c8d4ace90f5342dfd4cfb8d : SUCCESS in 28m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e2591806f32485e98b43b1e3736969d : SUCCESS in 23m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c315875c44674c2c9b78cd7aa243cf75 : SUCCESS in 29m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7d54ccf2f1c34cc381a4308db08db923 : SUCCESS in 20m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/748299b991f141a78beca71ebfb6fadf : SUCCESS in 10m 45s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"8de8695b8e8ec1462699cb455f0517ef5fa11f31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:28:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3d2746912f5bd3adef83acfbfc7a4256a4fe084","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 16:20:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22c816584fdf48be8fc9a60fd3b95015 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34749c0df8154c44aab1b2087c1ee6cb : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54611a2facf44c3d8e363e3b3ea1d039 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2505897844fa4b33a3a7ff7bf76a9468 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abc32f04fa2a4072a0abf57960a1344c : SUCCESS in 4m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8cf6c0b85a8f40bb85ee376f720073bd : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b619c720a2244a6863bcd05cc5d9996 : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75928efafeb249d99c72bfd035414d45 : SUCCESS in 19m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/157b3bd6e70b48edb15429db2df205ad : SUCCESS in 1h 28m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3626ceeb8ce4358a1ce929769e98ee4 : SUCCESS in 47m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/900a55c3d43245c3878a281a5feb6172 : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/000a9ce141dc4d43a19ae3a9b74eb82b : SUCCESS in 18m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4db318ef37b4e05970e29aeff54d35b : POST_FAILURE in 17m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1afab2f73d6a4edca99d67800cf04112 : SUCCESS in 1h 46m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/efa143efaaee491c815addcc9d957a65 : SUCCESS in 36m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3afae6fb99444e04b6218dfcf11ce9e6 : SUCCESS in 29m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a499568779274413be3b0700a360cc1d : SUCCESS in 27m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e4aa2846cd8642e18a6eec3b019ac95e : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74090843932544359978df87a5680190 : SUCCESS in 9m 42s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"ed81c1af797242fce129e26d6ec5717db58c9c27","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-08 16:20:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3bf149b81146a76d7629dc476182938491f9c0ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 18:13:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a946b15ea0194ff69d82b69c5ac2bce6 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f4d591b01dd4fce9b3279444c96f087 : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/453651d613204842a499ce1b8025fdc9 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69075662c6d649249cc00dd8203d5d28 : SUCCESS in 8m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d77fcb4379ce4e06a3f704ee262acd7c : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ff9faa7db3a49cea1985cbe28e38ec1 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88f280f7e5d246289d930214031ee289 : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c1f4469928834f5fb2bd1f34205ea4a5 : SUCCESS in 9m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/337337aa302144308c4441d81447c443 : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f016d63b173644449d932cc922b74a52 : SUCCESS in 19m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a8bd20a2d26a4d659422fd4ba1fbcf40 : SUCCESS in 1h 03m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c284c78f4c54c3abe96e9dcbe0525b7 : SUCCESS in 44m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8c52cdd8f44b44e981f2b3b397158bb3 : SUCCESS in 1h 03m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9a4e2a1de06b44b9ab71e57174383980 : SUCCESS in 21m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7f289345c604b2498e9770a0dc39fc2 : SUCCESS in 15m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e0f1fff1ebe743b8bba2c492aaf3702a : SUCCESS in 1h 49m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/aad31bbbd2464b5ba16941b407247ae7 : SUCCESS in 26m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d032923a6874e0ea8fcdd41dfd2c159 : SUCCESS in 26m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5a8e132da6ab44c8affedbf8f0d22172 : SUCCESS in 28m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4a6689e5216a4fa7b348108e4880eac7 : SUCCESS in 13m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4570698c1174fb3beaa612b4eb50464 : SUCCESS in 12m 27s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"196ceba2d241620f6853a8a76d34277acb832963","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 18:13:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70d74f1177d84ec637608bcf8ef28234c4b61031","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 20:33:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e182dd1949444821a0e18a60894cce41 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ef8e38fa3b644639b507ff57483259a : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aab1be54da704420978cd28399842312 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f1dcddc9d447f0bd5a13d592158a0f : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28cde2c9fb2e4e1093dd927dea730673 : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c0ba7e95d0294cbb9f584ad7c4ea7a35 : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/409c65a8522748b4881265faa26cda01 : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afce8ae6c99244fc95fdf4817782ffdf : SUCCESS in 25m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/daa2c150e5a54e0ea79676871e4e75fd : SUCCESS in 1h 19m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f3db59b51374d1baa141f36c5bb22c1 : SUCCESS in 1h 16m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e4fb5e7e66e04139b67f65528237b007 : SUCCESS in 1h 26m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/406c7d64f9714d0fb221c069fd976538 : SUCCESS in 19m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61f18d82223d4d81ad2d40c8ea4ff8b5 : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/29da2077032f4045b0d0e697da2b83ec : SUCCESS in 2h 09m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/677e86c7c96b41149249b3e3c84f205d : SUCCESS in 29m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c57e136f08e4fd5b529f5b8fa8600bd : SUCCESS in 26m 14s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/53ef626a13854a0d84eb3ff23e6bcdc1 : SUCCESS in 30m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/00a0bf3a254e4a9d842b9f2f48597eea : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ede1a8cc34db4c9e8a577f55c800814b : SUCCESS in 12m 03s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"52b3176a562daf2e292ba8ec5404c0c67820da3c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 20:33:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0d7cda748dd81645552c018c50f363bc31e3daed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 20:34:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2be455d1d22245e99c647d9f0788ec68 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5c2047a329a4b23b86d2449000d81db : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e6ca694da906d415364d412887b40ab1cbb43f9","revisions":{"7e6ca694da906d415364d412887b40ab1cbb43f9":{"kind":"REWORK","_number":1,"created":"2021-06-08 07:57:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/32/795232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/795232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/795232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/795232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/795232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/795232/1"}}},"commit":{"parents":[{"commit":"5fe1a7edd8d3c246a3ba10194a9ff5cebd84faea","subject":"chrony: cleanup during overcloud host upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fe1a7edd8d3c246a3ba10194a9ff5cebd84faea"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:19:29.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-08 07:57:15.000000000","tz":120},"subject":"Fix container image builds","message":"Fix container image builds\n\nContainer image builds broken due to disabling docker iptables\nmanagement in\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/751795. This\naffects Wallaby and later.\n\nThis patch configures kolla-build to use network_mode\u003dhost.\n\nChange-Id: I608c0362e6eeea65ffafb32cda5c19ccb6ec8b5c\nStory: 2008942\nTask: 42560\n(cherry picked from commit 9c8f74993d63d73dfead6d1edb2b2827ae0e71bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e6ca694da906d415364d412887b40ab1cbb43f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e6ca694da906d415364d412887b40ab1cbb43f9"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
