)]}'
{"id":"openstack%2Fkayobe~795461","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I14289df5e248b5b47fb8e60c6eea1ac845f50fd3","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"I14289df5e248b5b47fb8e60c6eea1ac845f50fd3","subject":"Fix overcloud post configure Python interpreter issues","status":"MERGED","created":"2021-06-09 12:02:49.000000000","updated":"2021-06-09 18:03:56.000000000","submitted":"2021-06-09 18:02:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795461","cherry_pick_of_change":790400,"cherry_pick_of_patch_set":2,"meta_rev_id":"64fd4efe6e1b5bc455cedaa642d5992ad439dd7c","_number":795461,"virtual_id_number":795461,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 18:02:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-09 12:22:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-09 16:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-09 16:09:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 12:02:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-09 13:56:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7edad2333bcbaf37517c874c7484f0adb65b211a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 12:02:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c0a19e7a9d70d61246b8e33142ff31d9e8b3f35d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 12:22:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ff1e0385b7b2de6ac4f609540722351f4341377f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 12:23:14.000000000","message":"Patch Set 1:\n\nShould we backport further?","accounts_in_message":[],"_revision_number":1},{"id":"0d8e3c8fc0fb1f290876b7c08f2d97bdcc709e97","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 12:42:41.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Should we backport further?\n\nNo adverse effect expected from the role bumps?","accounts_in_message":[],"_revision_number":1},{"id":"a674815dd31d9f01da8ee850d68423f036e0a12a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 12:49:18.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Should we backport further?\n\u003e \n\u003e No adverse effect expected from the role bumps?\n\nProbably not, but I\u0027d need to check.","accounts_in_message":[],"_revision_number":1},{"id":"45c1c9564257302fc1230eb10eb5539f6a76941d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 12:54:09.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1:\n\u003e \u003e \u003e \n\u003e \u003e \u003e Should we backport further?\n\u003e \u003e \n\u003e \u003e No adverse effect expected from the role bumps?\n\u003e \n\u003e Probably not, but I\u0027d need to check.\n\nos-images just has the necessary fix. os-networks also has a change for upper-constraints:\n\n+\n+# Use Train upper constraints when running with Python 2, to avoid\n+# incompatibility with newer OSC releases.\n+#\n+# Use Victoria upper constraints otherwise, due to a bug in openstacksdk 0.53:\n+# https://storyboard.openstack.org/#!/story/2008577\n+os_networks_upper_constraints_file: \"https://releases.openstack.org/constraints/upper/{% if ansible_python.version.major \u003d\u003d 2 %}train{% else %}victoria{% endif %}\"\n\nWe override os_openstacksdk_upper_constraints_file in kayobe so I don\u0027t think it matters.","accounts_in_message":[],"_revision_number":1},{"id":"ce83ad4ded3605fadff219cfb9f46bec1dd70f10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 13:56:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/813892dcee7d45a5a2bd4905d1eb46a7 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bae250a790d44553b35f82e1ed111469 : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/905e8175583b455dade51f1f96884b7e : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcbbb1c755544e59bd76878db30e928a : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1059f5e20995492ca1ee76549d5b0e9d : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27eb8f43ac9641419fde25dad2d42f3f : SUCCESS in 8m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb6d0094e7a84863b2d36a87a311dcfe : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d622a75ac0e4a2aaf16802a3d20eba0 : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f2f43f1a4da04c4dba31f7fde80d5b20 : SUCCESS in 21m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e2e7287b0dc44b57ac54e8bac6a62884 : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dba9316444fb4528b014adbac5637f05 : SUCCESS in 44m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/91abfdecdbc04c3c86c2180580141a3e : SUCCESS in 1h 09m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/350377de0d594b0fa5bf5c6e2a1a5962 : SUCCESS in 18m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad86a63debd24affa6221b2428ecca3b : SUCCESS in 13m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/57807cab26e44e64ba0eb532fb36d2a7 : SUCCESS in 1h 50m 01s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c4ef5ab72d014c73a7f15107d0d96700 : SUCCESS in 27m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c997fd89b1a4b50a529c3f67d629870 : SUCCESS in 25m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1d3e50634ca04e8db82333b47b1329ac : SUCCESS in 24m 59s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/38c166964db84e4181b18a815c69b627 : SUCCESS in 12m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35823ff1d73748d5a56c5e2fac2abd76 : SUCCESS in 11m 03s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"58694698bb4bc12d9e1ed78bf424456b06376293","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 16:09:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"25ee8bfa470490d6c28e2123c6f7cd0414b93123","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:09:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b90325aad4835f80c468d77172203e74c617c256","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:02:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3ccc203381b4ced97dd2eab98dd1ff1 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a51db772a734bbb96dbee77c6eff791 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a512fc80c0f42b5a4b5ab5385feda4a : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcaf95c37dba40e397e10a7b09e65fba : SUCCESS in 8m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8cee7d228bcd4bb4ab7e663ca87d86f2 : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a9a262d830c43208140f1ab89273a14 : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/da1858c9de5d4158827c7b229a7669bc : SUCCESS in 20m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/540e8f7873f04b0497247c932a7ccd5d : SUCCESS in 1h 07m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d8b01e63d564835b3b6c844a630da7e : SUCCESS in 43m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e6a459b4509346f1b7e07690c691bbe5 : SUCCESS in 1h 08m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bfa2d2b7c19d451c80396ed5bb9c76c9 : SUCCESS in 18m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6003127dffa84e909abfd4a87cd0da9d : SUCCESS in 12m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/99c82d2be0e14614b2410d32a72971e3 : SUCCESS in 1h 51m 58s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/aaa39033553e4660a7539aa734c1f31f : SUCCESS in 27m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9df021653624e3fb42bdd41ab41275b : SUCCESS in 28m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d7f1bad3917a4803aa93da62009b7db9 : SUCCESS in 28m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e03e8340855b41a2bfecfeee87b29b19 : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1167031566af4dd9b9016f580ee07cf2 : SUCCESS in 11m 20s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"79a2d234d4156005c05ca8b87ede19141b09a8eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:02:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"64fd4efe6e1b5bc455cedaa642d5992ad439dd7c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:03:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5299119df74597a82945e5539e8fa9 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e221681b228bbd3a7e41f6e8e40dcfcf2666355","revisions":{"2e221681b228bbd3a7e41f6e8e40dcfcf2666355":{"kind":"REWORK","_number":1,"created":"2021-06-09 12:02:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/61/795461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/61/795461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/795461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/795461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/61/795461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/61/795461/1"}}},"commit":{"parents":[{"commit":"85389d1cc73bc4008909683191aaccdd03f703a0","subject":"Merge \"Fix container image builds\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85389d1cc73bc4008909683191aaccdd03f703a0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-10 10:33:14.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-09 12:02:49.000000000","tz":0},"subject":"Fix overcloud post configure Python interpreter issues","message":"Fix overcloud post configure Python interpreter issues\n\nSome time ago in Kayobe we switched from using set_fact to override the\nansible_python_interpreter variable to using a task variable. However,\nnot all of the galaxy roles were updated to match. This can lead to an\nissue where a fact set by the os_networks role overrides the task\nvariable set in Kayobe.\n\nNote that we haven\u0027t seen this in CI since the code path requires the\nuse of a VLAN provisioning network, and CI uses a flat network.\n\nChange-Id: I14289df5e248b5b47fb8e60c6eea1ac845f50fd3\nStory: 2008284\nTask: 41158\n(cherry picked from commit 4a59ea483bc1baf57a6423812b597147d5e24b2e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e221681b228bbd3a7e41f6e8e40dcfcf2666355"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e221681b228bbd3a7e41f6e8e40dcfcf2666355"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
