)]}'
{"id":"openstack%2Fkayobe~795463","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9","subject":"docs: improve automated seed development info","status":"MERGED","created":"2021-06-09 12:03:39.000000000","updated":"2021-06-09 15:54:41.000000000","submitted":"2021-06-09 15:51:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795463","cherry_pick_of_change":793059,"cherry_pick_of_patch_set":3,"meta_rev_id":"70ac4a0275e1dd81deab7922b5262c16a58187dc","_number":795463,"virtual_id_number":795463,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 15:51:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-09 12:24:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-09 12:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-09 12:42:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 12:03:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-09 13:53:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"494058da92fca74cbff5ada0c8f195418affb704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 12:03:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9d9b30c482fb84e2ff5b0248868fd78d61848d83","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 12:24:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"acbbd5a0d68735406e256fd09cf91102a770f9d6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 12:42:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d52238bacfd90dc9161f919181bc08f223d21c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 13:53:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a66ba16ae36644d68e500e5a06f90166 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69cb255be4b04c04b50ee8bb8a9c0241 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82f4e36257dd4644876a30a056c3c44d : SUCCESS in 7m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85660ad79ac245d18b6058aaf4c2251f : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42652a87958846f7b1b4fcf0e4a20e4c : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5581071864224c0f9ccfb11a48a94721 : SUCCESS in 8m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d9652948ff144c09abd3153a1b05f9e4 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c730c48729f48a4827442dcfd21effa : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/88b1f0b4252a4aedb1fde32e39d58709 : SUCCESS in 20m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/68dfb64770b3465f86296ad5241ce0f9 : SUCCESS in 58m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d53041384f5c4cf79af6531fe8fad57e : SUCCESS in 47m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/87bb9bd4d15845bd8854d1d254ab75ad : SUCCESS in 1h 00m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/22748e0e22a94cecbe3d1ebfdda26df5 : SUCCESS in 17m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a8de17f3a324a0e905158c5f51a958f : SUCCESS in 12m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e32a2f578094c429d9a8c18f59a1f80 : SUCCESS in 1h 48m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/10a518bfcb89456c997e46af247d6d55 : SUCCESS in 27m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82d12ef72410404ca69d3d46628f58cf : SUCCESS in 22m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c917cff04bf4920851db2c80346eb7d : SUCCESS in 30m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6bd7c0811f134f20a39d41a297f0840d : SUCCESS in 11m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b5b2fa9857c444d837138ca256e37eb : SUCCESS in 10m 57s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"1f22713bdeac5145a5bfa960fafcbf0b32f84ece","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 13:53:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c4b1756a4fe93f2ef62b9b8457afa8c4cbecd06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 15:51:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a011f47d0af4cd5a1600bd900d4daf3 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64ccb267474946158ee1ed981d4423bd : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6489816dc102455ab823ab1f6525d12e : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e110beb1ec6a442882b583c3e6fc8385 : SUCCESS in 7m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/77ff404b41334a9e886f0d5e7fd73ef8 : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fcae51341e2f401a8e85b4c2e4c34ce2 : SUCCESS in 10m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/48480e6a339c44808c9f6f43742faaef : SUCCESS in 19m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a8509c2915f6487497e449e157c17fef : SUCCESS in 1h 32m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa7bfacac24e4cf088ac04ca7b656d0d : SUCCESS in 48m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/32ab354a5fdb4e98b00774197972f21d : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5ee37e87473243c2a5ab4db77ffa960b : SUCCESS in 19m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d17a3f1edeb44ec98640f9b88b949d92 : SUCCESS in 15m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/219702212d2d4e84a3b368922bb74fdb : SUCCESS in 1h 52m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ac25dc17a63640f7a042d18317887fde : SUCCESS in 29m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d217e7cfdd31495490b97cf67a1be1ac : SUCCESS in 27m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fab6d5dfc25b4c6683f89d077e3c49ff : SUCCESS in 26m 42s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e096cc20518c45eea79adbdcccb4cb44 : SUCCESS in 14m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bb2f2d0774a482f92c9c3b8e0bfdf50 : SUCCESS in 11m 34s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"a36517cbc92193f15dfc89b88ab90d6faf4d4747","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 15:51:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"70ac4a0275e1dd81deab7922b5262c16a58187dc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 15:54:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ecd82e453f42648b2ff20b8d989eee : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d56e8fd2e2357ed612c200fc61e1e047872cbdc","revisions":{"7d56e8fd2e2357ed612c200fc61e1e047872cbdc":{"kind":"REWORK","_number":1,"created":"2021-06-09 12:03:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/795463/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/795463/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/795463/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/795463/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/795463/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/795463/1"}}},"commit":{"parents":[{"commit":"85389d1cc73bc4008909683191aaccdd03f703a0","subject":"Merge \"Fix container image builds\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85389d1cc73bc4008909683191aaccdd03f703a0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-25 16:14:06.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-09 12:03:39.000000000","tz":0},"subject":"docs: improve automated seed development info","message":"docs: improve automated seed development info\n\n* add \u0027bridge_type: linuxbridge\u0027 to tenks configuration to avoid\n  dependency on OVS\n* extend seed development environment testing to include overcloud\n  provisioning and deployment\n* remove seed hypervisor and seed VM environments. These are very\n  stale, and largely replaced by a-universe-from-nothing. Add a link to\n  that workshop on the same page\n\nChange-Id: I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9\n(cherry picked from commit d2e3f6f1aec2c1ed301144ed26872394799346fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d56e8fd2e2357ed612c200fc61e1e047872cbdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7d56e8fd2e2357ed612c200fc61e1e047872cbdc"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
