)]}'
{"id":"openstack%2Fkayobe~795465","triplet_id":"openstack%2Fkayobe~stable%2Fwallaby~Id7378dd844aca7cb5634535308eba0f60342c81d","project":"openstack/kayobe","branch":"stable/wallaby","hashtags":[],"change_id":"Id7378dd844aca7cb5634535308eba0f60342c81d","subject":"CI: test base container image build in seed jobs","status":"MERGED","created":"2021-06-09 14:13:47.000000000","updated":"2021-06-10 16:00:13.000000000","submitted":"2021-06-10 15:58:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795465","cherry_pick_of_change":794601,"cherry_pick_of_patch_set":9,"meta_rev_id":"22eda908aa7d14f26a8fd710436ce0aed916fc0c","_number":795465,"virtual_id_number":795465,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-10 15:58:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-09 18:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-09 16:10:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-06-09 18:40:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 14:13:47.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-09 16:17:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"76c60cb9854a8694b7e50e9d1163e1d104723b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 14:13:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8cd490c04d3447a4c106dbb4021d07fb366284ce","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 16:10:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"685e9e014ef3b565d6ee2a79b8b281d5b8dafb96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:17:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae0511b9eab9431e8cd8edbfa070ba34 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0d183f1b3074db49c7b87c24e7ca9f0 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5590e71621145418ef7d1e5b1de4bc7 : SUCCESS in 7m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcf73d5449c1443cb6bd6e47254a8192 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/780f4907cf9f4b7185b32dc5a0a375b8 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f92fa9f5c36143e8a572f15f4be743cb : SUCCESS in 10m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e57f53808ba045e599ab737bf72d749c : SUCCESS in 7m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/059726bfa7f142e8aaf4da1173de50ec : SUCCESS in 13m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/af75bc20f9b240b19f6e3abef266c3ab : SUCCESS in 22m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6655372e07234b25966ee14ce9bfb48c : SUCCESS in 1h 10m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/243feb6d230b4a4cbc7f7ca647d645fd : SUCCESS in 52m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a4e49dffdd4045fcb165bb60d4ed9410 : SUCCESS in 1h 14m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/013a051065f84323a50770bbff5c616d : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96711c3aebee47b3b2627524b3101a36 : SUCCESS in 12m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d42b94de331442c2bb51ed5b809de631 : SUCCESS in 1h 53m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/738ea9c0f2b6459abef5a684c2c24052 : SUCCESS in 30m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6108d92740bf4c50b6cb3adb6f0bb8c4 : SUCCESS in 33m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/52018978333b492392d4d8a2927b7c25 : SUCCESS in 24m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6f5ebe489b4a498da0ade9cd151aecd0 : SUCCESS in 12m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02a96a95158d42afadac412c20d6949d : SUCCESS in 10m 31s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"f2b5e5898d602f1f75af726e4f8dece2db35a03c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 18:40:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc3b0ac894235dd6db26bbda22673323203d05be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 18:40:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef0eaf9db13ef62b6306ef1c020143d94781323f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:32:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bbf9356a28443cea090ed6ddd81d093 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c940a90d671d4b739c207e3069558903 : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7c6ba8840a747edbf476b4f98c6be7b : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac62c02aeb4b436096817e442232ab4e : SUCCESS in 8m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a64e4e6f8cc64274a837ddb400e2540f : SUCCESS in 7m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/325fab52425640ec93ddad995cc338c7 : SUCCESS in 12m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/32b1fd9f85c947e7a0d0da04cd074649 : SUCCESS in 21m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7043fac9a64c4014a3d4e08f0b7efe8a : SUCCESS in 1h 09m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bc383397468484f8a3f3aa92f3d6fb5 : FAILURE in 40m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e3939842844a439b915b29e10aae9231 : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/379a8791bddd4b6d882a6e07b4da24a7 : SUCCESS in 16m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5645f9bda00f4074b22e6a747cb71be8 : SUCCESS in 14m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6aa65fc4c51d428689907985e88716e0 : SUCCESS in 1h 49m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5ccaa634271343c599a1aa03e3bd2eb6 : SUCCESS in 35m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fef0fcd121c4ca4baaa9ad2bae21fff : SUCCESS in 25m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3cf10d749d05494496936c8a2911faf4 : SUCCESS in 28m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ce3a173f83944b53983d1399c1c06dd1 : SUCCESS in 12m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84288f4bc3ad4db299260bca3b00e088 : SUCCESS in 11m 34s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"2083f0dced93f589b49a9a881a2f0e31152da356","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-10 04:50:14.000000000","message":"Patch Set 1:\n\nrecheck\n\nThe following packages have unmet dependencies:\n qemu-utils : Depends: qemu-block-extra (\u003d 1:4.2-3ubuntu6.14) but 1:4.2-3ubuntu6.16 is to be installed\n              Recommends: sharutils but it is not going to be installed","accounts_in_message":[],"_revision_number":1},{"id":"694cfeb868957368106bf57786d2ab71bd587a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 06:42:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d528dc085a234071a9d159c1b201ef7f : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e709fe6e50834e09b6e688b1094fe4e2 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a234fcf4b7d645e29ecad3f9d83045cc : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c91b609333de4c2e94b8ad91f0e64139 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d972032a58c4977ad1e4226b8e70464 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd6ac1c24524179ac65bc8528f27601 : SUCCESS in 11m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f77a27b143d047058d171db086d406f1 : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f379b233aad04124b6c75cf79e64747e : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/64299a7e0fc04fbdbed4cd11eb7b20cd : SUCCESS in 20m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/920edc8749a8406ea2c727b1b5585994 : SUCCESS in 1h 03m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2328a2d34c284043b6ecfce50e81b215 : SUCCESS in 44m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b940d8230a3c4d7f8eb4a15bf6afee05 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/472f5799d4614e888c8b658e96497756 : SUCCESS in 17m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9533f50e1334aafb559ccfee5016412 : SUCCESS in 14m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/816d7e0801fe4f08b25bc69988841f61 : SUCCESS in 1h 51m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/517b804acf014f6a9360890a5439b953 : SUCCESS in 31m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5202bfdf3003400cadb5494adba95afa : SUCCESS in 33m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f72c48d60cb840aca16bd7cadef212a3 : SUCCESS in 25m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9c05753414694cb8a1c5c93bab6aac09 : FAILURE in 9m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1396d64d6bdd41ffa2d80c7dd8d1168f : SUCCESS in 11m 37s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"2a4601711d0936e398d8a0dde893d0c2c5beb11a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-10 06:51:14.000000000","message":"Patch Set 1:\n\nrecheck\n\nurlopen error timed out while fetching ansible roles","accounts_in_message":[],"_revision_number":1},{"id":"ece6e838eec8308ca00f6111713ace58d8a9a3e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 09:07:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91b0655d56d34909874b34b74dc83b1b : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea401545b5648c9abaf3fe52931cfc8 : SUCCESS in 7m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/05d5200868ac4a549127730f1f9fcbc4 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a860016ca3a443fd98fe5b3191bcbb15 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d9dd46aea21241489e0ea736ebcb8307 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60a65b32085b48a2afc915ba1d1b79be : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eabc78b6bf6b4fa3a95aaaca806d3481 : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/365a85525f124f6e8ac00bc1d9cd726a : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8e46b79ee8f54c629b74703ede2e274a : SUCCESS in 18m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3fbdf7e306bf4cd1836583bb4bb4c63c : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc0b026b3e98478c92f8ac4899354b61 : SUCCESS in 48m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b6570cf48ce9462bb049e88b59b106a9 : SUCCESS in 1h 00m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/62d8f5344d5d45948544cd98894388ee : SUCCESS in 27m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d78b4c7dd5f4c7c919fdcc0e8f23053 : SUCCESS in 21m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a33fd3a61497488293bcfe3a8df37412 : SUCCESS in 2h 09m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/837282817c544d1c8137d9685639d83c : SUCCESS in 37m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/509ee2c9a49c49458230cf8d8575c0f0 : SUCCESS in 33m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/86244c748aec409eaccc47c5bba17c24 : SUCCESS in 25m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b1cda01d4aaa42dd83cb3f4b666cbc58 : SUCCESS in 12m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b9af4f076b94b3ebf7fe32c274d1451 : SUCCESS in 11m 08s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c226bbeb066520159f848383e5e571cf3c3bf2c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 09:07:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab46d6ea27f6008464d7627d18538adb1b4a3fb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 11:03:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c53ce99657ef46089f73675b18f46fee : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c672128ee544ebc995f8e78685d4683 : POST_FAILURE in 19m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76ad85307d174250a45a25d0259c1f16 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d1b18e52eb745498c0c8cb30013e040 : SUCCESS in 9m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2a6eb52a60d04982895f4e216ed46c0f : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7663fc399674610be79dc4f619bc51f : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c89890211cc451ebb0322a17145fab9 : SUCCESS in 20m 31s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7161614776964aeca12e269ddcdc7cf2 : SUCCESS in 1h 02m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b3a52c397804c9d8fb23efe57c623dc : SUCCESS in 45m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/12e7d975b75e418ab42b0adf227532d5 : SUCCESS in 1h 11m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/82c8d2a3bc514afe9eabf193ed512830 : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a5a77813995443ba54c88f675f7739a : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/657d4919dea541ffba6460f1f6fb7727 : SUCCESS in 1h 52m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/52d9575c3ca4444e9064647f37dddf9b : SUCCESS in 29m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd305bbc2b4f40b98e87fab5fcf71147 : SUCCESS in 29m 03s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b466dc902e834581821a24f489c1003d : SUCCESS in 23m 45s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ca75f64b0768450cb949963a30a7da3b : SUCCESS in 10m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddc683e0873f4d17a30b6c72bc6f638a : SUCCESS in 8m 53s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"853f32432830be9a064d1323904042639352ec81","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-10 12:19:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"498bd5d2288a59a9c528ad58e5d6a4b9623130a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 14:12:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/039fa35b041f42bcb8f7dbf008e961a4 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d3ecd5866f84cd8a7498f3a42e37a25 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da845ffda35b43feadd244d0a2211399 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11ae1401a9634cf49117a5d08677a2c5 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33ee16e2fa2f4cc9bfe4c62e2e91fb5d : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5978218c4007432581fd5eb7188a7758 : SUCCESS in 9m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1af510e4ee6147d5b8a0da0d5fd6e45b : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d70a861256134bb39618738a6a522715 : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f841b5ceaab4a5ababcc2d4bedd2c40 : SUCCESS in 22m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d7fedace7b9446119617a37190d9a877 : SUCCESS in 1h 04m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28f8357f61e94783a21000533fbdf515 : SUCCESS in 45m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2bf04435d26c41edae4c99581179e160 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3b43df7a62a24ee387d77f0005a0dab0 : SUCCESS in 19m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9884f91d2fa04483af89e4149717474b : SUCCESS in 13m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5495e9ca1e8e418f901a57ddc6c2e99d : SUCCESS in 1h 48m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ca150c517eb44bb7aa969ef366b18fda : SUCCESS in 33m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfe710a1bef240c08fe7104efc84ed9f : SUCCESS in 29m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/25d4e07cb5b94ae6b2c38a977ca2c533 : SUCCESS in 34m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dd9b7edbb7f1486993832bcae9ff3c8b : SUCCESS in 14m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/584cbd6eee5b4f5d8c57d5f72e64ed7b : SUCCESS in 10m 30s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"88be02469a73cbb0a480b1c013c55f240d7f08f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 14:12:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e3e59280b8b1a9ca882571215f84f263cff2b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:58:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de03694b5c4e4cb1a3df731a72bf86dd : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/457785231fff4b428de8953a3042a6f3 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55156a4402144b6a8db70ba03c6949bd : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef27fc294e10447a924b796b2fe61bc3 : SUCCESS in 9m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46894d4e8d48422e99a9a8b819d8d457 : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/89b1d5ecb68244aaaf04cfc06b130eb1 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2d364727e1cd42838ebea80fb1d5662c : SUCCESS in 22m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e341e7c24f3a4280a101467d3757cc60 : SUCCESS in 1h 09m 14s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6876e474cd548089352d6bbae6c63f5 : SUCCESS in 48m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3b6a0808adee4d23900cee79e86775f8 : SUCCESS in 1h 10m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/13e8f5d6b5be401e9cd8ae8b4f3d199d : SUCCESS in 20m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8b9c95b9ae348debff821248da7b95f : SUCCESS in 14m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e44b688ed7874f058c54e32edda8a53e : SUCCESS in 1h 44m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/74f535ee6c7945919245ad3de1adb592 : SUCCESS in 32m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a706f57309ec4c1a9844fe9e63b2a1b3 : SUCCESS in 29m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/79406502e6044dfaa7f76bc78726ef8d : SUCCESS in 25m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/97fb68a64f244103981ac4597c72c18e : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed4cf873690f4f07a13b7dcac360abd3 : SUCCESS in 10m 25s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"0903fa487e2cb7ca09dc25457881ffe36cba10ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:58:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"22eda908aa7d14f26a8fd710436ce0aed916fc0c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 16:00:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd4241925718453089bc8a8fc57def28 : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e010356e7b4789ae17b770e51fbc5e456411ee62","revisions":{"e010356e7b4789ae17b770e51fbc5e456411ee62":{"kind":"REWORK","_number":1,"created":"2021-06-09 14:13:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/65/795465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/795465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/795465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/795465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/795465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/795465/1"}}},"commit":{"parents":[{"commit":"85389d1cc73bc4008909683191aaccdd03f703a0","subject":"Merge \"Fix container image builds\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85389d1cc73bc4008909683191aaccdd03f703a0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-03 14:31:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-09 14:13:47.000000000","tz":0},"subject":"CI: test base container image build in seed jobs","message":"CI: test base container image build in seed jobs\n\nCurrently we do not have any coverage of building container images. This\nchange adds a basic test of building the base image to the seed jobs.\n\nA potential future enhancement could involve pushing to a local registry\nrunning on the seed.\n\nChange-Id: Id7378dd844aca7cb5634535308eba0f60342c81d\n(cherry picked from commit c6c1cde3f9cf8bb232071b78ab89c79a0c7689b8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e010356e7b4789ae17b770e51fbc5e456411ee62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e010356e7b4789ae17b770e51fbc5e456411ee62"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
