)]}'
{"id":"openstack%2Fkayobe~795491","triplet_id":"openstack%2Fkayobe~master~Ibe177fe078769204c7e89c04e40870890fe501ba","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ibe177fe078769204c7e89c04e40870890fe501ba","subject":"Revert \"CI: Fix libvirt errors caused by latest edk2-ovmf\"","status":"MERGED","created":"2021-06-09 07:49:09.000000000","updated":"2021-06-11 20:25:40.000000000","submitted":"2021-06-11 20:24:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795491","meta_rev_id":"53b2e20d271a6a5bf823066fddb10786327fdfe9","_number":795491,"virtual_id_number":795491,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-11 20:24:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2021-06-11 13:48:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-06-09 18:41:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2021-06-11 17:06:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"date":"2021-06-11 17:06:02.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 09:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 16:06:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-11 13:48:19.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"}],"messages":[{"id":"15334fcdf30ea1304ac99e04317eaec4f06cef9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 07:49:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79b7929f1d36c328dd4fb1318d15d3c3650fcebd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 09:35:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/978538fe61e14790a07b9406c8d5676c : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68575909f6244393be5b17e8ffab9da2 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a930ad1a20c74edd995fbbfb0bb92424 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3217ad74eda47b2b2546d78ffbfe94a : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a9216a9473043fd8d9d78798dd59ec4 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2580883913a544f69046406686886ef4 : SUCCESS in 8m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c9f9e838b1d47a9848f463e01450584 : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/18ec0ff2582b4232841daa4cc5b4ede0 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d542edf3e51a44ae8e8a583527ea0b22 : SUCCESS in 19m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/40aad4601ff74f3b8dc3a7decb591c8e : SUCCESS in 1h 07m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e474272d58e046b2b3672c3938852103 : SUCCESS in 44m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/39272d4cc65f49d2af61792b08ef8279 : SUCCESS in 1h 05m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ab8ea0f398a048aea4acade1a2c9c787 : SUCCESS in 21m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a3efdc8f1764370a93860a921897044 : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2d2016d665cc49adb8e524c4a4b31931 : SUCCESS in 1h 38m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/dfd677355df6444cb9f94ecebfb58da9 : SUCCESS in 28m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cfb15d5c8cf4100b070f1d659ee26b8 : SUCCESS in 23m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/74059a060b69482f89b5e31069cdfdbb : SUCCESS in 28m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c93cfca703f74945b7fe63126fa69dc1 : SUCCESS in 14m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f94e3193c1bb42719a64d44ae96b80eb : SUCCESS in 10m 32s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"135e5e83876fb7b13ebeeda8c59f809b217f2e4d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 16:06:45.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"71babcb4bd7b692433002bb7c1b06e894381ade3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 16:07:13.000000000","message":"Patch Set 1: -Code-Review -Review-Priority\n\nAlthough maybe hold off until libvirt 7 lands","accounts_in_message":[],"_revision_number":1},{"id":"3f431e488fadf5939c93ff219f275c252e5485a3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 16:14:38.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: -Code-Review -Review-Priority\n\u003e \n\u003e Although maybe hold off until libvirt 7 lands\n\nWe don\u0027t use the advanced-virtualization repository for installing the seed hypervisor\u0027s libvirt, so we will stay on what AppStream provides, i.e. libvirt 6.0.0. libvirt 7 will only be used in kolla\u0027s nova-libvirt.","accounts_in_message":[],"_revision_number":1},{"id":"a5b7d9e8c936772f4c60d18dca129646ac86a492","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-09 18:41:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"403cda3931eb3d4434f958508101323174045130","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-06-11 13:48:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dc29aefb9a2a92a732f43d2b253c841f6f6e78d1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-11 17:06:02.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"75505c746892553eaf374578013456002993a653","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 17:06:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a49119291f3de8b1a038213e43f4bf79f2aa2178","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 18:24:25.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f076ba15382d9003687f9aea3cb4ea4dd737b201","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 20:24:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cadf01ea82d746d3a12f7b70fa0cf446 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ec803c2304f4f6890716accc6acb1b6 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44d6f98d87574d42ae7d52bcc958167f : SUCCESS in 9m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11253c54858b4711a5fa4ffa8e479688 : SUCCESS in 11m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e163aedc233243f49a6eb73cd4d166bf : SUCCESS in 9m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f40716a72a29493996d3a32c4b26dcd4 : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b9d12309cad34209b756bc5ef8e02791 : SUCCESS in 22m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fb9fc33f3af9426c8756868f6dac4bcb : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02edad069a924234a9b51c54857d4dd2 : SUCCESS in 51m 16s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bb3411ff95d54bd6a520a6d612189c0a : SUCCESS in 1h 11m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b314e4f9f9cf4f70ae3e7f14bf3fc31a : SUCCESS in 20m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f0b797e87b846eaa9ba84c6f99dbde9 : SUCCESS in 15m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9cd261552ccc40b19fd989aea2d8ae87 : SUCCESS in 1h 51m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/596e01eb76e44aed89a03291e67590c5 : SUCCESS in 32m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d338b1064be048bb94ecbf8c23e72560 : SUCCESS in 26m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f4735122469e4dfca4b39b13ef5026d2 : SUCCESS in 41m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4b76dedededc44b79420f43d136dbbca : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c9922f0eaff44bdac7283a569612c2e : SUCCESS in 13m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":1},{"id":"9ff5d1043980db639123b43415f280201e4fc182","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 20:24:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"53b2e20d271a6a5bf823066fddb10786327fdfe9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 20:25:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c0e0cbc63f8483b838243153e10f85d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66e143ada2b9e902cae2c63536f276ac84664f9a","revisions":{"66e143ada2b9e902cae2c63536f276ac84664f9a":{"kind":"REWORK","_number":1,"created":"2021-06-09 07:49:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/91/795491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/795491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/795491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/795491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/795491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/795491/1"}}},"commit":{"parents":[{"commit":"910dd452d9e19ed919c6f8930dc1a872a4bb4acf","subject":"Merge \"Revert \"Use OpenStack Wallaby release\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/910dd452d9e19ed919c6f8930dc1a872a4bb4acf"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-09 07:48:33.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-09 07:48:33.000000000","tz":120},"subject":"Revert \"CI: Fix libvirt errors caused by latest edk2-ovmf\"","message":"Revert \"CI: Fix libvirt errors caused by latest edk2-ovmf\"\n\nlibvirt-6.0.0-36.el8 was built [0] and released for CentOS Stream 8. It\nincludes a backport of a commit [1] from libvirt 7.4.0 to avoid erroring\nout on unknown firmware features, which fixes the issue seen with the\nnewer edk2-ovmf package.\n\nThis reverts commit 1e14fa3a24fa64b84cc993c985e696a9cba78dea.\n\n[0] https://koji.mbox.centos.org/koji/buildinfo?buildID\u003d17918\n[1] https://github.com/libvirt/libvirt/commit/61d95a1073833ec4323c1ef28e71e913c55aa7b9\n\nChange-Id: Ibe177fe078769204c7e89c04e40870890fe501ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66e143ada2b9e902cae2c63536f276ac84664f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66e143ada2b9e902cae2c63536f276ac84664f9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
