)]}'
{"id":"openstack%2Fkayobe~795745","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ib91fda6420a8255ba5d3cbb7de98613c492eb9b7","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Ib91fda6420a8255ba5d3cbb7de98613c492eb9b7","subject":"Revert \"CI: switch to quay.io for container images\"","status":"MERGED","created":"2021-06-10 09:06:53.000000000","updated":"2021-06-16 19:29:59.000000000","submitted":"2021-06-16 19:28:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"795745","meta_rev_id":"e019d2e6372b409990e28d266eb3f8553714d9a2","_number":795745,"virtual_id_number":795745,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 19:28:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-06-16 14:26:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-06-14 14:09:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2021-06-16 14:26:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-06-16 14:22:26.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-16 14:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 10:44:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 14:09:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-16 14:15:00.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2021-06-16 14:26:07.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"ca019ee4e193c4c58e2b61673b34cc841b0e0750","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-10 09:06:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"936d707847ec21f147f484b876ebb3d710895493","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-10 09:07:39.000000000","message":"Patch Set 1: Workflow-1\n\nWon\u0027t work yet, recheck after Sunday 2021-06-13","accounts_in_message":[],"_revision_number":1},{"id":"bd6b66a844cd0ca09739091431ea8f3460798172","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 10:44:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf568ac687614490b83a32281203a365 : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9b0bfc4c8d834d358b9f15b12a289716 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b33097020cf24e5d80d39be4a2ae28d3 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02ff80f8eeb84f3bba73a98a9abac523 : SUCCESS in 3m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92d68e25df0a47d3b1d1f78689603ede : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce12b920d614fa9b4760d74616a0caf : SUCCESS in 6m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/87bc61aa283c472ca3f0440ca1791564 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/59ddd0fc47544c27b6d2a6b383ca7383 : SUCCESS in 9m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a703cae52edb4f4191820962caa32c62 : SUCCESS in 19m 19s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/c68f7887b9394c1c849afd581bf3db0a : SUCCESS in 1h 07m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ef401a1ebe7b4645806a7186871a59ab : SUCCESS in 53m 11s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/4e0bddc48bdc484fb7beef0bc708e73d : SUCCESS in 15m 23s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6e703e75bdb442cb9c0ad2def1e6470f : SUCCESS in 16m 44s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/fa418e2dce0e4a77a1c4d897a0f2edc3 : SUCCESS in 1h 30m 37s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/60c67f110e5e467499fc383b5fd94a34 : SUCCESS in 19m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a5a2a05a97834ee5b584986307380554 : FAILURE in 13m 46s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/49187413b2784defa32c639633a3d40a : SUCCESS in 31m 26s","accounts_in_message":[],"_revision_number":1},{"id":"efa925d65b5024fae66d433c10f3ced1b7381358","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-14 07:09:39.000000000","message":"Patch Set 1: -Workflow\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f33bd4d634cd9e957da8f25b3a1d4508a3c1f07b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 08:44:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32a676560e7c48eb9b29c21289481626 : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2ad0f01fe5e0495e9e506f86a303907b : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/851f7bfcfa7d4146b20e8e100581316f : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebacfbaee8644710955da785d9366a17 : SUCCESS in 12m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e1aba8ad3554b32b30ae9355e0db367 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f0c8e033e14eb8803bdbf2f80420e2 : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/71c637521a61414aaa307c5663a6164a : FAILURE in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d41882e9186d46218fe573a341015d74 : FAILURE in 3m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9c30948b7e944671b896e2821b2bae06 : SUCCESS in 16m 52s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/9b388f1897ba4f14aa938d9ff0e1556c : SUCCESS in 53m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/aa5f744d66d4411cbd9e054f21792708 : SUCCESS in 57m 40s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/79a28564b7234a729be214ed97ca2a2e : SUCCESS in 14m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0c68c826346e493c86474d6b2384c12c : SUCCESS in 15m 29s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/058405cd969e495cb4cb6adbc57de18d : SUCCESS in 1h 31m 41s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/5358b8450c9c4b998abc3882e8296883 : SUCCESS in 21m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/13293e4d755749b6a765ccd209a82930 : SUCCESS in 1h 14m 06s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/91621097f0214f4b867822fd473523c0 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a1dc8bacc281cf8f8eb345019d1050e12681a7d5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-14 09:28:38.000000000","message":"Patch Set 1:\n\nrecheck\n\ntimeouts from mirrors","accounts_in_message":[],"_revision_number":1},{"id":"f8272b5efdcacacddd6ee9224395823570dcdc4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 11:11:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db26bf7869a44913b9ba31e0dd07c043 : SUCCESS in 5m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aad34aae74bb4fef9734206f1e66d28f : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba26bf0f35a24bd8bf17bcd2a9b6f7c2 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c40844ca89645e9ab83f410f8a38f6e : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6d85a8e5d68452aab7c2ad872cd267d : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51cc28fe8b64728a7c976be52d25ec3 : SUCCESS in 6m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5682087a48340808e956ca1b50a3c7c : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e00d175a73754bb4bb266b9043d5026a : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/729e47f2ee414663a6cf4184ff384672 : SUCCESS in 19m 57s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/56c766f5afb94f998f42a95ac76fe517 : SUCCESS in 1h 01m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f4e5eeb78eed46ef9623f022c0c83131 : SUCCESS in 1h 00m 09s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/9e50aaed09c84c7b9bd851ed64e6b68d : SUCCESS in 15m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9c6c385b85b9475d954b9c954136c76b : SUCCESS in 18m 43s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/dc7eeebca7a244fbb7b692dde62ff3c1 : SUCCESS in 1h 42m 12s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/aa77b860101045fdab663d0711171b2e : SUCCESS in 21m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f6ab26f40ef640f18e360a12925fbac5 : SUCCESS in 28m 42s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/7a2d7c8fd1a141f5994547a15cdeccd1 : FAILURE in 36m 33s","accounts_in_message":[],"_revision_number":1},{"id":"073c9e990eedd0038c3e9f33ec7e5405cc030f10","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-14 12:25:13.000000000","message":"Patch Set 1:\n\nrecheck\n\n404 from packages.grafana.com","accounts_in_message":[],"_revision_number":1},{"id":"8e45e4344b56ed82d821e128dca84aab171cabc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 13:58:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/661df1d9d82e44ecb212af3bbbb4e4c3 : SUCCESS in 4m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b0aca4bdf7bd43ddad08eaea2437bfe5 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8916afd5faf1407681cfab7aaabfe536 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26619c777ea04ad0adbbfc8efb4c09a7 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d76dc880f67405c861f711e53de0d10 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3568e7e4018946d583e657df9d121718 : SUCCESS in 7m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f6f515493db44cab8c4547c0b364b98 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a34716d80354b148a6965e191bdb127 : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c1f62aa10b54f05b0af54fd51bb89d5 : SUCCESS in 17m 20s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/c61a3f30a2aa488fbe2ec08bfb5def9c : SUCCESS in 57m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0a06c3da67784c94aab5739c60bd57df : SUCCESS in 54m 43s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/14b45619c3284bb29c1fc82010f4611b : SUCCESS in 14m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8cd09742534f454ebd3cca6f37af35b9 : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c37aab5a9c8e4db58a6dc02bb99bbcb6 : SUCCESS in 1h 29m 25s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/edc2ca2e74e346449b793db7895270e5 : SUCCESS in 19m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fda37484b5024ffa99c71fb0832a70d5 : SUCCESS in 24m 12s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/aa8a581833d3475fbdb7f4a3b53d4f8a : SUCCESS in 34m 40s","accounts_in_message":[],"_revision_number":1},{"id":"10bdd4f311908f9ecfb6a6566fca0ac2b5c5d8d4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-14 14:09:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"36618c7f44c1cb1b476e9f3e99fea39ac852e8b7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-16 14:08:01.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"9a833529a8d6e44d237f1a2a2f9c040e56c1db2f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-06-16 14:15:00.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"222ebfcc1dfd074fa5ee519eb82ff96322ba6ca1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-16 14:22:26.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d8712f950dfcd5379cba33c171c16dcfc1ac830d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-06-16 14:26:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e63aafbb21ec053cd38d11e6beca449be3a7a3db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 14:26:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"880121410d8f50b2381b674ffa13a29e75d445ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 19:28:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1f4fe453c78249c697fd53f6ceab2279 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56455569eaf24bd58eb2409d646d28b6 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1d302242eb34d34986d1c4e74ee651d : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c9e50713dd64592aeebd642c493c5a9 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf6e7b03ad9f44a09c0f7bf87fbac41e : SUCCESS in 6m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89dbb0e8f59d4baa8c8111d52fe80b08 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72e1ff1bf9b54e83807be957fb6dd04e : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b7dcaae466445479a0486f5e8c83d8b : SUCCESS in 18m 44s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/46970b78a1c44837b4901ab2b297abbc : SUCCESS in 58m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8ccd27db6ee647e88f6f23ca43eb5692 : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/10fef6c3f4b3465db28f38c429b43fd7 : SUCCESS in 15m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1f8463c8925845bc97be7a9097f204b2 : SUCCESS in 18m 25s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8757abc444f849da8b166e55b204cb41 : SUCCESS in 1h 48m 25s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/618767486bf14909a51428688f1ace30 : SUCCESS in 25m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7fa37acf8be046078ba3981dc035865b : SUCCESS in 36m 51s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/83f9ef0f979f4ed592f1f72d38d0814b : SUCCESS in 45m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5903b2c2ac1f9064b1e6e8c3270b7a198a502360","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 19:28:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e019d2e6372b409990e28d266eb3f8553714d9a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 19:29:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515391ef4bff4d32b111a55030651fc2 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad86b8803da435e8b47349a5a6bb3d62606479c7","revisions":{"ad86b8803da435e8b47349a5a6bb3d62606479c7":{"kind":"REWORK","_number":1,"created":"2021-06-10 09:06:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/45/795745/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/795745/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/795745/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/795745/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/795745/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/795745/1"}}},"commit":{"parents":[{"commit":"cfad3bed429b32e4dc941e573b00c22288686706","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cfad3bed429b32e4dc941e573b00c22288686706"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-10 09:06:36.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-10 09:06:36.000000000","tz":120},"subject":"Revert \"CI: switch to quay.io for container images\"","message":"Revert \"CI: switch to quay.io for container images\"\n\nKolla container images for stable/train are not being updated on\nquay.io. We need a new bifrost-deploy image with the systemd package\nfrom CentOS 8.4 in order to fix the following error when deploying on a\nCentOS 8.4 host:\n\n    Could not find the requested service mariadb: host\n\nSwitch back to images from Docker Hub, which are still updated weekly.\nAn image with CentOS 8.4 packages should be published on 2021-06-13.\n\nThis reverts commit b26f4c1cdf9a70db267b73535e06b882a07917d0.\n\nChange-Id: Ib91fda6420a8255ba5d3cbb7de98613c492eb9b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad86b8803da435e8b47349a5a6bb3d62606479c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad86b8803da435e8b47349a5a6bb3d62606479c7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
