)]}'
{"id":"openstack%2Fkayobe~795783","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9","subject":"docs: improve automated seed development info","status":"MERGED","created":"2021-06-10 14:00:29.000000000","updated":"2021-06-17 11:27:41.000000000","submitted":"2021-06-17 11:26:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795783","meta_rev_id":"0cbaf8c9a7e39c93ceed0137728023bc144d9daf","_number":795783,"virtual_id_number":795783,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 11:26:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-17 07:34:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-11 19:54:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-17 07:34:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 14:00:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-10 17:36:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-17 07:34:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"360c7aeeba5ecef4b620091e931d86836e983d5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-10 14:00:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee42b3abadc5504842c579baf9645bfcf3ed817b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 17:36:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a7ec9583cd0422cb6d8910f95519f60 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ece3c602f65e4a2398dc1672f006c11b : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/710a2486a43c48c89f540a0fb44b65ae : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8455135214a34064ab46050c9bbd327a : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c614cecd9ec4b3aae21668ea514052f : SUCCESS in 8m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75ea1ac11bc84e03a1f17e2a2c68ab66 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a758137d729b457b8f9f764f5caddd55 : SUCCESS in 9m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/74665d2499fb469f98900330de52bbb5 : SUCCESS in 20m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c0965bc56161480d874480e2486c55ff : SUCCESS in 1h 10m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/fae7f94d2afa46589a132eeb8f78f4fa : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ae2bebda529944e2a388507eadca161f : SUCCESS in 16m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8eaff2d561324eeb85ee911a6ae82bfb : SUCCESS in 1h 54m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ab0d21e7c46c410d9ce3025176c914e7 : SUCCESS in 26m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/48ec8cb50bbe4bbcad8649bc3430cd9c : SUCCESS in 28m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b8b9ad0cc9f04b9d83b2ab7d8a68fa10 : SUCCESS in 10m 24s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"3417342591e739c518213df6a851372023d76353","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-11 19:54:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"35d540f08f88732cb45af07b04a34afb7112cb5a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-17 07:34:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c1a5df546ec4c8574682c53ec716b844a028060b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:35:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ff6d06888f70cdec34210f73ce158a13a31ef78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 11:26:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4e1c7a14494dbbba760c65d5057b0f : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a961af3057eb4624ae217e5f9d5ad240 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f79f7b76eb44486595b160c0fe4d8fbf : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd009981f8d4f388a75189dfac368c2 : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8c2b1e7d6ac4e6492638ae0eeb0bde3 : SUCCESS in 11m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e8e67780e1a148d6b5829ccdae271262 : SUCCESS in 16m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aba863c87cf243ab99ab692617596513 : SUCCESS in 28m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1a30ce1e17054ac4abb0736bdc9d35cd : SUCCESS in 1h 04m 27s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/cef21819dae84110a409ce3a2ebdf851 : SUCCESS in 1h 15m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a036f0850a2745ca82dbcbecd45df291 : SUCCESS in 17m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e2932614ca8245aaa1320b5fa58934f4 : SUCCESS in 2h 10m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/12106ba7cbf5426e81a064648e1a3f3f : SUCCESS in 22m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/21ee167787544177adea76213fcbc14a : SUCCESS in 26m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/233b2909870641a0a4ae7f6f113c3c47 : SUCCESS in 11m 17s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"19fdad08ee9dbbed056889aed9a484ee73bc8fad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 11:26:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0cbaf8c9a7e39c93ceed0137728023bc144d9daf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 11:27:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a27edf49d39142c7a2bef287babc1dde : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0746885bdaad11a86e27af97ae229ebd53795b14","revisions":{"0746885bdaad11a86e27af97ae229ebd53795b14":{"kind":"REWORK","_number":1,"created":"2021-06-10 14:00:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/795783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/795783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/795783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/795783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/795783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/795783/1"}}},"commit":{"parents":[{"commit":"8a8905ebce3c0bd9d4a3c54db6acc9c8b914f6cf","subject":"[docs] Improve contributor\u0027s automated setup document","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8a8905ebce3c0bd9d4a3c54db6acc9c8b914f6cf"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-25 16:14:06.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-10 13:59:22.000000000","tz":120},"subject":"docs: improve automated seed development info","message":"docs: improve automated seed development info\n\n* add \u0027bridge_type: linuxbridge\u0027 to tenks configuration to avoid\n  dependency on OVS\n* extend seed development environment testing to include overcloud\n  provisioning and deployment\n* remove seed hypervisor and seed VM environments. These are very\n  stale, and largely replaced by a-universe-from-nothing. Add a link to\n  that workshop on the same page\n\nChange-Id: I9928e5912e6770bdcc1d5d0884d2f101c16ee6a9\n(cherry picked from commit d2e3f6f1aec2c1ed301144ed26872394799346fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0746885bdaad11a86e27af97ae229ebd53795b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0746885bdaad11a86e27af97ae229ebd53795b14"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
