)]}'
{"id":"openstack%2Fkayobe~795831","triplet_id":"openstack%2Fkayobe~master~I67f0e48ee03b150f6fe52b1a3e7e727b79207231","project":"openstack/kayobe","branch":"master","topic":"image-regexes","hashtags":[],"change_id":"I67f0e48ee03b150f6fe52b1a3e7e727b79207231","subject":"Prevent matching unrelated images","status":"MERGED","created":"2021-06-10 15:17:21.000000000","updated":"2021-06-24 11:24:43.000000000","submitted":"2021-06-24 11:23:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795831-image-regexes","meta_rev_id":"435cc128d949c36a1d71a75b619fb0c7477ed1fd","_number":795831,"virtual_id_number":795831,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-24 11:23:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-06-23 13:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-06-23 09:30:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-06-23 13:40:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 17:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-11 19:59:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-06-23 09:30:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-23 13:40:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2656fb0aa7cb269061ad486541fe7a29cf31e76c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-10 15:17:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"528a7161db6d37edfa9a7014897e339b7835d05a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 17:18:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29c325ec5c3c4746af10071958cb8d01 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc66f5838d74932a3f7c807b3b50445 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7115da8b10d34ab6a1bb978e21c2b512 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f06282e34fa45cdae4baeff5c3b1101 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26e58508ffe94f3f8191728cda1712af : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7bfcfaeb55e4312928dcfa9c83dee56 : SUCCESS in 10m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/26bb424b84c74d56897e4c121cc63643 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/75252f0917444035897261a0efede750 : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fbc44da8b0614b79b893d3c0e288bfa8 : SUCCESS in 19m 25s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/edb67e901232488b9e6a73b1564e8b35 : SUCCESS in 58m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c67e3d90f294883910641d7ee1bcf35 : SUCCESS in 43m 46s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/65ca4bb9902d49d1b29a52d7f0f924e4 : SUCCESS in 1h 01m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/843dc22883954e039c5b9b72c8be0925 : SUCCESS in 23m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24171f5c3fa5456fbebd0842e56e241b : SUCCESS in 12m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/88707f43921a446e8941e8576bafe20e : SUCCESS in 1h 55m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fce2f78115e14c68a6c1d7d9e451541f : SUCCESS in 29m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45e3988da8034ca499a99a143dffd870 : SUCCESS in 34m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/55919921e45247729c24cb8530b22d13 : SUCCESS in 25m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d1f403fb5dfa43859375fcd96e186edc : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3598d59a8f7f4684b0005a959951dba1 : SUCCESS in 10m 17s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"aed949b13bd25a8bfa45ca178c4bcebe83a85752","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-11 19:59:44.000000000","message":"Patch Set 1:\n\nWe could add this for every regex","accounts_in_message":[],"_revision_number":1},{"id":"4bc9b5a1ee279c960344c2d2f4365f78246f1116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-17 14:27:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"79e50732cab226b21533e904713b76f87c8a29fd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-17 14:31:08.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1:\n\u003e \n\u003e We could add this for every regex\n\nI was a bit afraid to break something, but I have now changed all regexes.","accounts_in_message":[],"_revision_number":2},{"id":"d2722b2836a630da773a6a41a2a720ddc999d19c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 16:15:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cb94434bfde4f3b85d99436007da6cc : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c115136674da4f9798c475b066951f64 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b438120e87ce46af9719794a3c5f598a : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c84fae63916f4c36aa31c0d2d4fb96e8 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bf3bc7e32c54f928a36fd807bed8234 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3377aa1fe7bc4a1d86cf615c6724d0fe : SUCCESS in 8m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/875bfc0dce464c9994e7b89280dc8244 : SUCCESS in 7m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/41f3b2f791a7431a9c9bcc4bbfe147fc : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2bf72bb2bec54dfa80872a011c2ee3c0 : SUCCESS in 21m 40s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5ced3e46bec64aba982c6e2a8442cc4b : SUCCESS in 1h 05m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c7c13797dd84e1fa764f80c2182ead9 : SUCCESS in 48m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d8c52c41982647adba33e5df60479bc7 : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/fa60fae6195642e99e090634a2b79763 : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab5fbe20926e42c9a5c3449a2944e092 : SUCCESS in 15m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a13f2d52acc4426b81917b254ab58371 : SUCCESS in 1h 44m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a5b85416f63347ebb6d48892dc9cdf58 : SUCCESS in 32m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e2567e11b6d4d8c8284fa9ed04dccc3 : SUCCESS in 27m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1a2f53a6c5bf4c0bbb6f6d7066c0dc73 : SUCCESS in 29m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0c1702158ab9474c8f17cd83a838ace7 : SUCCESS in 12m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4a2043c666a4873af86fc350b628b87 : SUCCESS in 13m 23s\n\nWarning:\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"8605bdd5f7cd0931dcc0d81f82c0991db7fb0d1f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-23 09:30:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e139f1195b2ef8a52c0b8a2de70c891e6f120597","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-23 13:40:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bcad3bdc6e0761f0b3c566840d39cd386d2e5446","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 13:40:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e5458cfadf5f0e348a33aa96ebfb6819b40aa908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 16:30:38.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/604bb00f20604a6bba01ffdd9c287061 : SUCCESS in 9m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b23621a38f74871a49d5a5b4d87c0c7 : SUCCESS in 8m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33a7ea990ec54c5d80174d71e03f9af3 : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8134b77d9711469290480f70085ca60c : POST_FAILURE in 12m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/509190dc90914ce5ae4f59ad2bda4720 : SUCCESS in 10m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f46508befa04c009510eed6b25d5902 : SUCCESS in 12m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dfd5c21a1eb647829f44ae7c8c2c7f71 : SUCCESS in 19m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6ae2de90992245c69fff2fe088971813 : SUCCESS in 1h 03m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5643eb2a00240ccbc8d8a508ecee124 : SUCCESS in 56m 52s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d123a523341c48f5b4b8c1710303c30a : SUCCESS in 1h 06m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/cf9568d91ae04177963c07daf45a0c38 : SUCCESS in 19m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32b9eca0c3114e088b76b459e5a11ba1 : SUCCESS in 16m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/31f5151c7af4443596a17079e2847862 : SUCCESS in 1h 55m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b4e1824b0065435a82c24b495eafcb39 : SUCCESS in 31m 31s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59f43c47a0ed48e19f98ee098db794c2 : SUCCESS in 31m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/01294c62f4a44474aab74865b1e75952 : SUCCESS in 27m 28s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e106200974904f4fba3b013394cba3ca : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90af9ec87d5c4396845ad79afb2940b4 : SUCCESS in 12m 36s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"732e6d0b7f1df650255c46f4df1b2c115cfed02e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-23 16:32:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e663327966fed1ba1c9e0feb163468c1c594b9e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 18:29:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e41cbfdec8804170afa87bbe0a1803e9 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ea1a95f751c4f60b604256a7bf8166f : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/68a845dd467340c59065a97a71e75a6d : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f330573346ab436c872df237c75da2c5 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88b06340505148e5b8f5592acaa31cde : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f3d8a36c71446e0bbf0970e563a20c5 : SUCCESS in 9m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/921468a301024e44a6452b2c71a4a74e : SUCCESS in 7m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17bb88fd366344339f92cd37eff3ea65 : SUCCESS in 12m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b22186cf899c4276a8a0130aabc50a38 : FAILURE in 10m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/81278c02aeb54fe8b8e017bb98b278c1 : SUCCESS in 1h 00m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14f8dbead0b14bff80445067c308470c : SUCCESS in 45m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cf5b96988e194350971246a902d94097 : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/108b593204ab4f6880fed843dea32259 : SUCCESS in 18m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4323dda46aa8488fbd7d3231911533fb : SUCCESS in 17m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6d2889267bea4f42bff9d951b40ca7c0 : SUCCESS in 1h 53m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ecace804dd274308aa577356b9197f0f : SUCCESS in 36m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58b89f6ea44c427b91541e81c8639373 : SUCCESS in 28m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2dcb0b6e5c3f4011945ea060e8cd5c81 : SUCCESS in 27m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ec7af6d40fd04f759f1e4e8cb72e879f : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af7f7664142d4ad88a3bf8a98be7f607 : SUCCESS in 11m 42s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"c26e5ef0ec5bef296692d78e43677572878af54e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-24 07:33:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fd65c327ac43389af8d1744b28b92599da97403f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 09:26:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c614b77d49b849ef93a471fdb819baa6 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26273c16593347be855a43c16fefce82 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee442aeeb73d466fb952383d5ebaac36 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46cf27be0b674f54be1811e535fb0ef9 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/952acd2084a04a35a5d4245a3e699516 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/618844bd91a8463d8e8bb6f8fd070623 : SUCCESS in 10m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/135f342832ff4eb4bbc393478223e469 : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e858ee54ac794298ad416a759595fa49 : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/52292450c79f472cbd95c6fda1411f96 : SUCCESS in 22m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7cbef9bb22a845b89a9ad39272c302c0 : SUCCESS in 1h 07m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/355bedaea1994f1ea2b404409d591385 : SUCCESS in 48m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1679deaa3d7e4e7dbe2e5d484414697e : SUCCESS in 1h 09m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9119afc275964f27882dca835817f2f3 : SUCCESS in 18m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b49adf93e1ab4f9e99f91b8312d63e28 : SUCCESS in 14m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6e5ac74edb554901b30b44996c3df529 : SUCCESS in 1h 46m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9c6c4e8cc1e449ccbd8507166f525cc0 : SUCCESS in 35m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61dbc9207b224d2283c39b554f709e1c : SUCCESS in 33m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3090ab38aa884fffa94235b049d981b6 : SUCCESS in 27m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/06a99b3e6ed44f3f999d2859522c6a3e : SUCCESS in 14m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4adce6c032854f18ad4e6c4c0ac69a71 : SUCCESS in 12m 57s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":2},{"id":"19465f2296590ba37bb32b14becc8980d50273e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 09:27:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"deee69b8bebd32ee02c5e00b04359605013b924b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 11:23:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b885c7dfb4e4eccb8e8779fb4ca867f : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28de7ecee80f4908a6f7eb9666dc18be : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88beddb87c524000895a3e13b35472d6 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e547dfb2ddb64a9690030c531893b222 : SUCCESS in 9m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85edd79be348453aa184c4919c06c971 : SUCCESS in 6m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/256db7b51dc4492599a0ebadc01a303e : SUCCESS in 12m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dd1554e550a84f35a8d94fa489ab7feb : SUCCESS in 18m 17s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/146e433ac26b4f7fabdab87f5ccfe172 : SUCCESS in 58m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7c1520bce7748a09603e2a717a32ad8 : SUCCESS in 44m 07s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b2724b7dd16747a194ed8498b26f5c4c : SUCCESS in 1h 17m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d1af036c6aa84246afded1f8d76ab79c : SUCCESS in 18m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e58dc3318f0c4b32a4b0545897680c2f : SUCCESS in 15m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/050f1fc63f1e416c93123189ac36e65c : SUCCESS in 1h 53m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/54d3e89406414c5e9eb8eff047e0e605 : SUCCESS in 38m 36s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a9911de93274456b897d7d80c4a1446 : SUCCESS in 26m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/88cf0e854b0246c8bff290df761d84f7 : SUCCESS in 28m 09s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/033d12ea0f8b4196a43ffd8911dc313b : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25919bfff24f443db3dcc2444abaf1f3 : SUCCESS in 12m 05s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":2},{"id":"b21bcb2b477f01aa749815c95cf22cebe752380c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 11:23:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"435cc128d949c36a1d71a75b619fb0c7477ed1fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 11:24:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06cee06c423940bdac6c8622d14a0a0d : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f575f4b7fd11b4155b7f4f78981f71fd66f4d2c5","revisions":{"03badb61265999c59dc3c2f215424a1a93272e68":{"kind":"REWORK","_number":1,"created":"2021-06-10 15:17:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/795831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/795831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/795831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/795831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/795831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/795831/1"}}},"commit":{"parents":[{"commit":"79a3c47d0f201429afde58e900058d89becd0372","subject":"Merge \"docs: add variable for current branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79a3c47d0f201429afde58e900058d89becd0372"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-10 15:15:49.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-10 15:15:49.000000000","tz":120},"subject":"Prevent matching unrelated images","message":"Prevent matching unrelated images\n\nBecause regexes defined by Kayobe are not strict enough, Kolla can build\ncontainer images unrelated to the deployment. For example, the default\nservice configuration will build:\n\n- centos-binary-barbican-keystone-listener (matches keystone)\n- centos-binary-prometheus-memcached-exporter (matches memcached)\n\nThis commit makes related regexes stricter to avoid unecessary builds.\n\nChange-Id: I67f0e48ee03b150f6fe52b1a3e7e727b79207231\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03badb61265999c59dc3c2f215424a1a93272e68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03badb61265999c59dc3c2f215424a1a93272e68"}]},"branch":"refs/heads/master"},"f575f4b7fd11b4155b7f4f78981f71fd66f4d2c5":{"kind":"REWORK","_number":2,"created":"2021-06-17 14:27:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/31/795831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/795831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/795831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/795831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/795831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/795831/2"}}},"commit":{"parents":[{"commit":"79a3c47d0f201429afde58e900058d89becd0372","subject":"Merge \"docs: add variable for current branch\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79a3c47d0f201429afde58e900058d89becd0372"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-17 14:24:54.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-17 14:24:54.000000000","tz":120},"subject":"Prevent matching unrelated images","message":"Prevent matching unrelated images\n\nBecause regexes defined by Kayobe are not strict enough, Kolla can build\ncontainer images unrelated to the deployment. For example, the default\nservice configuration will build:\n\n- centos-binary-barbican-keystone-listener (matches keystone)\n- centos-binary-prometheus-memcached-exporter (matches memcached)\n\nThis commit makes regexes stricter to avoid unecessary builds. Note that\nsome images are still built unecessarily, such as nova-compute-ironic\nwhen ironic is disabled or neutron-metadata-agent-ovn when ovn is\ndisabled.\n\nChange-Id: I67f0e48ee03b150f6fe52b1a3e7e727b79207231\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f575f4b7fd11b4155b7f4f78981f71fd66f4d2c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f575f4b7fd11b4155b7f4f78981f71fd66f4d2c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
