)]}'
{"id":"openstack%2Fkayobe~797072","triplet_id":"openstack%2Fkayobe~master~Ib756630d50cb9cf89342f32bf1d4e07df31b1a62","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ib756630d50cb9cf89342f32bf1d4e07df31b1a62","subject":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled","status":"MERGED","created":"2021-06-18 12:11:41.000000000","updated":"2021-09-20 14:09:34.000000000","submitted":"2021-09-20 14:08:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"797072","meta_rev_id":"1774628f4f32bd377eafd13560bad2284253861d","_number":797072,"virtual_id_number":797072,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-20 14:08:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"all":[{"value":1,"date":"2021-07-27 08:38:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":2,"date":"2021-07-27 08:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2021-07-30 07:36:38.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-09-17 12:56:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-17 12:56:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-18 15:46:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-24 10:52:49.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2021-06-25 10:20:25.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-28 08:05:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-27 08:38:09.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"}],"messages":[{"id":"1cfa25cb48c88eeeef46e9e389408bccce1853b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-18 12:11:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e69c0414096a0b80bb6ae6966edebfa7de389a98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-18 12:13:53.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6bfc289ea2e9d9d774782bde22504dd3703c2f3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-18 13:38:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"408396bf981fcab028932034c68cfb204e6ad9dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 15:46:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/686638e8674446ac9ca70521eb2f0d9d : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bc59d4f095849709c38fd66efecbdfd : SUCCESS in 7m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab083c94288645d0bc59f743b448daf7 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6149f7fe819342b4ad8e187b763a44c9 : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ea94c802a874bbbb16ed75545dd2cab : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3354dc7762f340a7928886269f3ae460 : SUCCESS in 8m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3670aa5d92774a6db95ed336003e3150 : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a2812f2a7939435ebd4b48016c2ba300 : FAILURE in 9m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e3e339cd6d794de8a0c0d6ab0949aa93 : SUCCESS in 18m 44s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2dfcbc36da3b4347a8da5ff7093ef187 : SUCCESS in 1h 02m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85459d3dd2df4758a004cbf2c45e331c : SUCCESS in 48m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/baf160ca426f495eb22bf4845b0da1c9 : SUCCESS in 1h 15m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3d8f14a6a6474026bf9ed6c5b7e17a57 : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/301fbfecb21a4941995066353624c837 : SUCCESS in 13m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7fcaaff8f3344088a1cfc0895ad9b5c8 : SUCCESS in 2h 01m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c89595e708c14a149e5a47a43ecbd812 : SUCCESS in 35m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4251bcc58df345668895dfc66a0f8a4b : SUCCESS in 33m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bb5eedb3e20c42aabf61f29799895ee6 : FAILURE in 16m 45s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6b3e520217134f8dac03f23c5820194d : SUCCESS in 14m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00f4c66659b948f3b32e6ee5c5c449db : SUCCESS in 10m 08s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"b76c159c22a1fcf6d1a82fe35d8391867a5eaa50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-24 10:19:31.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"4e24395940f622a13fcad46ca18af3cbb313e344","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2021-06-24 10:52:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9b5a2b2fce04ec4372d21a50ced8833075523652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 12:06:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c028ed9b51248b689e06115b883b3f4 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ef50d40a8664f51bf2c4fddde3b345e : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7514725430b942e1bae1345ae8efbf0c : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0c90431752b4695ad5045586c5fd51d : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d100cc40f17469c94acf83e662170b5 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e40521a4700e4940a3aed5814a0ef4c0 : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f0bd3ad508ef4e448f8e5f78dc221a78 : SUCCESS in 5m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a7421f923e442a5a2763d09a5829c00 : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d39909a4a5024642b96074a376b35606 : SUCCESS in 22m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8f6b4258ced7410c978e065c0225e124 : SUCCESS in 1h 09m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7283fd9ab1484ad3b7b0ef40f06cc946 : SUCCESS in 47m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3249ed964a074b26b5c95cba630a409f : SUCCESS in 1h 09m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c17a960fc54c4f6faa0a4f7bb3eae1d8 : SUCCESS in 19m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81366536a9344ba9b52a098f4cc17b57 : SUCCESS in 15m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3bb49ad6abef4deea9c83093b7cec7a0 : SUCCESS in 1h 43m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/31f531f469064a76a2152d7628705597 : SUCCESS in 32m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cebd6636d9504003861dcab76d373ff7 : SUCCESS in 27m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8bbc6e64300540f492835e6d0cae47ed : SUCCESS in 25m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9a15358c633e4adfb47310e1dedfe2ef : SUCCESS in 14m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a96e7fb1e9945d0b409389c135cd3e8 : SUCCESS in 12m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh","accounts_in_message":[],"_revision_number":4},{"id":"36208fbd8a35be79e25abf05e4c4f04efe198a38","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-02 12:54:19.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e78a11dcda9eb72e2cd128a63fda7c902203e979","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-07-26 09:46:21.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nWe might want to backport so would be good to have a reno.","accounts_in_message":[],"_revision_number":4},{"id":"3c64ca1f4ee283bd9fa711a2f1b0c3e859cdd8cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-26 10:02:14.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"84df949e3c8820e041d2e6764185b9ef83f4b09b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 12:10:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/365f0a7eaad04a29b50041ab7840e94b : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/284b85431a7e4f56831661fb6ec5facc : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c8818c37c9b4902a8991980334b3df8 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/428a491c15bc4ffdae38c9c99a2abf4a : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dde4062c26eb4e38b40ac067b30006f3 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c938111f53d42f0926b340d1fbcec11 : SUCCESS in 9m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/129b1b44eaba4aa08376969391bb6195 : SUCCESS in 6m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/63e32a21443f40979256a29fa1930364 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/534c56c8f5ee4f6e96ea12d5b4798963 : SUCCESS in 21m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f1ca98419d8b4f648f4ce9e82847ca45 : SUCCESS in 1h 03m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c25a74d991ef4db5b074e8ee1a889d71 : SUCCESS in 47m 08s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/31b25e1904a64cf59a10e835779ed744 : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7d780e7cc530432ab6f547bc28b7f1b0 : SUCCESS in 18m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0e45fdfc3694fdbb013ec0d29759884 : SUCCESS in 15m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e9d4ffc5db604ce5877ce84a8962ba71 : SUCCESS in 1h 56m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0febcc49bc6a4d1686a844f152d150ec : SUCCESS in 38m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d19732a2ff4d468f81e62ccfdbedf878 : SUCCESS in 32m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/abe4179565304299b9546214f346e864 : SUCCESS in 25m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/00470a7ce3c04ce681035927fa27ea1c : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aba71269c35147469a48d8e0fa625267 : SUCCESS in 9m 51s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":5},{"id":"06d88524957db0a2489950b29a2ad0b66142d651","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-27 06:36:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"96deb0403b9d68ee987555236bc5cb8555d46fd5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 07:47:39.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dc598f37d10e4c557feea36d9a01948368fab5e5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-27 07:55:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c54f13831b6c965562bc7ff1b1bf1b48e5524166","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-27 07:58:17.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4348f2f865c23cd54cdee223837347c7341c757e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 08:02:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0aa1ae6424d5f8d1eb199a550c1e46db2b372b5b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 08:02:40.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7007da7474b1fabab6a03350608e71747e482027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 08:21:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e00195daf9b54fb3b1610c5bb3dedfda : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b82258172c754819bc76dda59415d540 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37e56d8586aa43eb86d6005a677fcaf0 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3f4b7dc57c746438153804d31c09121 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79c03ff4f8184fa3b2605983dd69a58a : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3765462bcc9d4cdb86d9bd5c87e871d9 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9edca90b0b974815a3d9712e9db687c5 : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dbb8ad37d09b479aa7a1de69a03873b8 : SUCCESS in 7m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/25ab29b3492846d4871e40cf2884723e : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84e4ebb7e3c544098d9e2a1ac5970356 : SUCCESS in 22m 55s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a2b1e0dd9fbe46baaf21ff9345f7256b : SUCCESS in 1h 07m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3fa4975baa34938ab3d42add20b6c8a : SUCCESS in 45m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/264bf7323f6249bea1c3fe1090995aea : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f225cd548e93432682e4721a70c55273 : SUCCESS in 17m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8647b8b98194dc0b8a9f33622cbac0b : SUCCESS in 15m 37s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/61e24ce4480541fea30440637b96b0db : SUCCESS in 1h 40m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/51d563aa6cff461e835566052027a651 : SUCCESS in 38m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3c1ed4b29a7479d89fa3224ebf09cc7 : SUCCESS in 27m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/48f305ea13e2403fa3f7ffe7386c41bd : SUCCESS in 27m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5b2e2000935e4ec9bfb253fa1f41c01a : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2768820026d4a02b7d79df47ba68896 : SUCCESS in 11m 39s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"9e419727933b83f580648ad2116df64d06a9ac20","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2021-07-27 08:38:09.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"70c426244c8a384576fc6814fdaf0e6aed5fcb3a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2021-07-30 07:28:13.000000000","message":"Patch Set 6: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4b3052d0172d2e3cc16cb7ee13942665b56410f3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2021-07-30 07:36:38.000000000","message":"Patch Set 6: -Code-Review","accounts_in_message":[],"_revision_number":6},{"id":"f9b0f3f0c2b1185c18d77a05050c37ab0a196da1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-03 15:14:09.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4e876c4545b875f4c7b9d5460fabd7d20914bcb6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-04 04:41:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5b08e90265d9bb21ac6d7ca09c077d3960ec664d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-04 08:14:44.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b799be267436de0759450b9e714b0800a93b3fc4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-17 12:56:53.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"c5ef74351459f167cd10122439052c0a2c3c2b8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 13:05:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c4d8e5fc1ece664e61cee14982afb20164bac0c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 14:58:25.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57245aa903a4aadb3ac29f953242b64 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92d24031e8a44a5a9c9f3c480c9513fe : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2962cb29475f43258ce2f7d07f28e446 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090a245dbc76474299ed8defa33941bb : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3e6ffbe13934471b4b136085d3faa10 : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/367c5f66d53d49749f087a95fe4d4d73 : SUCCESS in 6m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9fe12d331b7c4c77831420299dd903ce : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c449044cd60b411f8fd20a2483488f86 : FAILURE in 7m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4640a167ab2a4aa69410f717c9db55c1 : SUCCESS in 1h 07m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eac017d998874ff38d3db716e5a2599e : SUCCESS in 44m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6844004b66624a6a94987f044d99d002 : SUCCESS in 1h 02m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a14ebf56ea3441b2959c610cb5e87add : SUCCESS in 18m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/179668e43e524ccf9c9f04bef9c4d034 : SUCCESS in 13m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fbd0ae0bb9be433f88bbed1b97f488a3 : SUCCESS in 1h 48m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1c00661263b34b23b3e1308db0db8102 : SUCCESS in 33m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e04f3aee17e34678a9ba2a0a8c8a8701 : SUCCESS in 29m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0ca331ca4d484ce3be99d0308f6c94d3 : SUCCESS in 27m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c91cc2159add45108892923cc43eb9d9 : SUCCESS in 13m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/523cc71c988b471dbe1f217abe763c55 : SUCCESS in 11m 48s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"1d796a23691905a4b3ed66e0af16c5c5fd041040","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-20 07:06:38.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"b6ff8b59f9e2642be67cb491ae3ff9ebc6b90c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 08:58:18.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/faf76a5f2de54ae7b79f3f8050bca3f4 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e494327dd9a449cfbe10589381809302 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3ca4f4a288d4e8d9358497a04edb51f : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/134c5bf0d8504390b0101b008a8e2b01 : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e57f0feda6564c44b682536e3ac38cd6 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054d9a3c687f41128df9d8afee7f07f5 : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0828dd76b746403ca72daaaf120df954 : SUCCESS in 5m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c8faa874e014967ace185604426892c : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2ab6020a3a9d40fc8728b6bf0df2c0cc : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7252dbee36c64d7a8c0726fc16c72649 : SUCCESS in 32m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/99e693a264644279bec223ec35ad96e9 : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46ba0a3dcf5f4e7d83f556ac48dfcf11 : SUCCESS in 42m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/db52814220764c0a80ecade755b48ffe : SUCCESS in 1h 04m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/669597f9cba34d7a82cff854da8c9b83 : SUCCESS in 20m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b745dcd0b9504bb783f27eee789df742 : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1bfcc2f65e434e54a9c3cde35ed753a5 : SUCCESS in 1h 40m 03s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6d79ae7002ce4b0abc1c97f372efdaa4 : FAILURE in 17m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f7baf306b294a238ad76cc5b2431ee6 : SUCCESS in 28m 58s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/63bdb89e6952474fae946646f48c887b : SUCCESS in 26m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8ecf3d6d3b3c4bb19c8933d0ecc49191 : SUCCESS in 13m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef12c7a629b7442fa1a1d5c479568ba1 : SUCCESS in 10m 50s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"8574e9df8e21a5091435277221cadf828b89c7d0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-20 10:20:27.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"252cdc9b8905c91bc8bb982445f54cf7948dd114","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 12:14:36.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9df04c7fcc7f4bab9e009de84bdbc532 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6de330d7c4845abb608daa4526c57a4 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/589f2cc553924382b8c8368ac9e80f5a : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c4df293df874f558e7e262295b93ce7 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a5071a0a0ee49d99937af19823a946b : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ed06e1ab84a49c09ceb303d96c6ef05 : SUCCESS in 8m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18ff126ec6704036aaea5ebb99776aed : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a60d431a8ac54cd98d463f4319a90cc3 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5353fde0a2b64545abbfeeaa3f25dfeb : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7f6ae86d8e17406780958ed7e0202aa4 : SUCCESS in 33m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6360cd79016c4980a103450fb8aa8570 : SUCCESS in 57m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90f42c156c9c4a3294c947e009e9297e : SUCCESS in 47m 14s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/429a7e24feda4e4f97e3b514e524ecca : SUCCESS in 1h 09m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ee73f746bc27424cb4795120ff4b204f : SUCCESS in 21m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ec68386a61243328b7fc6a09d5b28cd : SUCCESS in 14m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fc85d170bb4b47b7b7ddeeed19bf04d0 : SUCCESS in 1h 52m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/645d22baba71429ab36d5300512c3b8d : SUCCESS in 34m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14ae50031101486da33e3f43527d9eb7 : SUCCESS in 29m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/76d5edfd434f45008538f6afa1800462 : SUCCESS in 27m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/17832b96174b4bfa9cb704df827f45f5 : SUCCESS in 15m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0293b70e701f4f559a94d2e1269291f3 : SUCCESS in 12m 25s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"4d52e435949fe6d56fc88992ca572512b3c3170b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 12:15:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d040a5fce9fa85fd67d59c7730c409c348c7809c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:08:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d974ca8fa8f1453aa075d77278ba2710 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7739dd26bea4519befa2e7d5cc511cc : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0c8295ebf034d0c96e727916973c88d : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a9ceabfe0c4a2d8b86d868475a1326 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32d6f59aa04544839be3d83d196fbf99 : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d4d550de9df246bb915926d26c915b20 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/32aab549e3114d9ab22eda69a38a0ce1 : SUCCESS in 10m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f03d6c5470bc424e8799de0029e28c85 : SUCCESS in 29m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2e54d3f614a64f4e8445fea335090b88 : SUCCESS in 59m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/517ed7300a794a808d1ad6b549c24afb : SUCCESS in 46m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/86f0730bf63d4c6c8439e7435298f4d5 : SUCCESS in 1h 03m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b593b70ca4864443a1ae1de65d8e2d2d : SUCCESS in 19m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04f8ff7e2167436c902eb061478d9250 : SUCCESS in 23m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/992d0380080a4080ad703278c3e45b05 : SUCCESS in 1h 45m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/562e441be08e4f53a70f694bca2e222a : SUCCESS in 40m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aae2a426f0a4236a414fa7f3e3d4721 : SUCCESS in 34m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bb8822c7cade48c79a4e6ba2e87a6e6b : SUCCESS in 28m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/80944c4cead043af90deda49da02d5de : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b185ad1c3b443a189269cba6aee8cbe : SUCCESS in 12m 38s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"85eec4bd8efdefb923ed40a458963fe60a63eb15","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:08:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"1774628f4f32bd377eafd13560bad2284253861d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 14:09:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a717bbb0a7a3435faf1b322a5a64fbf9 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19eba69f40e2412fa50de7f231fb177b : SUCCESS in 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"fb36029b59ff4abe00abd506fbd6bd02a2e2fa2c","revisions":{"1411a104100ad8e502e40aac6165d23152c22995":{"kind":"REWORK","_number":1,"created":"2021-06-18 12:11:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/72/797072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/797072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/797072/1"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-18 12:10:27.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-18 12:11:38.000000000","tz":120},"subject":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled","message":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled\n\nkolla_enable_ovn currently just builds OVN images and deploys OVN without\nconfiguring Neutron to use OVN\n\nChange-Id: Ib756630d50cb9cf89342f32bf1d4e07df31b1a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1411a104100ad8e502e40aac6165d23152c22995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1411a104100ad8e502e40aac6165d23152c22995"}]},"branch":"refs/heads/master"},"49bd729d708ee13950d51ccea6b7d24295e16826":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-18 12:13:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/72/797072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/797072/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/797072/2"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-18 12:10:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-18 12:13:48.000000000","tz":0},"subject":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled","message":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled\n\nkolla_enable_ovn currently just builds OVN images and deploys OVN\nwithout configuring Neutron to use OVN\n\nChange-Id: Ib756630d50cb9cf89342f32bf1d4e07df31b1a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49bd729d708ee13950d51ccea6b7d24295e16826"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49bd729d708ee13950d51ccea6b7d24295e16826"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9abb0af2b32d666743538f0405825deab78d405b":{"kind":"REWORK","_number":3,"created":"2021-06-18 13:38:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/72/797072/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/797072/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/797072/3"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-18 12:10:27.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-18 13:37:52.000000000","tz":120},"subject":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled","message":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled\n\nkolla_enable_ovn currently just builds OVN images and deploys OVN without\nconfiguring Neutron to use OVN\n\nChange-Id: Ib756630d50cb9cf89342f32bf1d4e07df31b1a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9abb0af2b32d666743538f0405825deab78d405b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9abb0af2b32d666743538f0405825deab78d405b"}]},"branch":"refs/heads/master"},"095a400238c1ef67478618667a5e86dc5c2fda18":{"kind":"REWORK","_number":4,"created":"2021-06-24 10:19:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/72/797072/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/797072/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/797072/4"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-18 12:10:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-24 10:19:26.000000000","tz":0},"subject":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled","message":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled\n\nkolla_enable_ovn currently just builds OVN images and deploys OVN without\nconfiguring Neutron to use OVN\n\nChange-Id: Ib756630d50cb9cf89342f32bf1d4e07df31b1a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/095a400238c1ef67478618667a5e86dc5c2fda18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/095a400238c1ef67478618667a5e86dc5c2fda18"}]},"branch":"refs/heads/master"},"43d83d15f254aa051a74df640b934b8cee9c59ae":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-07-26 10:02:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/72/797072/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/797072/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/797072/5"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-18 12:10:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-26 10:02:07.000000000","tz":0},"subject":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled","message":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled\n\nkolla_enable_ovn currently just builds OVN images and\ndeploys OVN without configuring Neutron to use OVN\n\nChange-Id: Ib756630d50cb9cf89342f32bf1d4e07df31b1a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43d83d15f254aa051a74df640b934b8cee9c59ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/43d83d15f254aa051a74df640b934b8cee9c59ae"}]},"branch":"refs/heads/master","description":"Edit commit message"},"fb36029b59ff4abe00abd506fbd6bd02a2e2fa2c":{"kind":"REWORK","_number":6,"created":"2021-07-27 06:36:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/72/797072/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/797072/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/797072/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/797072/6"}}},"commit":{"parents":[{"commit":"2445ec740e3445f813ba6b23c086bf113c5d96cf","subject":"Merge \"Fix url to ansible site in kolla-ansible.rst\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2445ec740e3445f813ba6b23c086bf113c5d96cf"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-06-18 12:10:27.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-27 06:36:18.000000000","tz":120},"subject":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled","message":"kolla: Set neutron_plugin_agent to OVN when it\u0027s enabled\n\nkolla_enable_ovn currently just builds OVN images and\ndeploys OVN without configuring Neutron to use OVN.\n\nStory: 2009080\nTask: 42892\n\nChange-Id: Ib756630d50cb9cf89342f32bf1d4e07df31b1a62\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb36029b59ff4abe00abd506fbd6bd02a2e2fa2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fb36029b59ff4abe00abd506fbd6bd02a2e2fa2c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
